)]}'
{"id":"openstack%2Frequirements~954575","triplet_id":"openstack%2Frequirements~master~I15780557b57c06ac1c474d717f3166a3504626ed","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-07-10 20:36:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15780557b57c06ac1c474d717f3166a3504626ed","subject":"update constraint for aodhclient to new release 3.9.0","status":"MERGED","created":"2025-07-10 09:39:13.000000000","updated":"2025-07-10 20:37:40.000000000","submitted":"2025-07-10 20:36:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954575-new-release","meta_rev_id":"1e8f554d6b369f0ed730060a5787a8a9736689d3","_number":954575,"virtual_id_number":954575,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 20:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-10 18:40:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-10 18:40:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 11:24:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 14:17:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-07-10 18:40:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"f0b4112b752d15e3442e7b22be233fbb6ea55373","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-10 09:39:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b5988d1a31bcbf9ae5f850c7f2535272640b0b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 11:24:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5044c6a1aab4415e8b5f1eab4c73348c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810ab5ba74874428b05569aaa20dfd23 : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ab80e195d34ac9a80a9e72379db555 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/128abcd9623e4c90a329d27df4d8a186 : SUCCESS in 2m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a525cfe64dc449df8ea0fe53bd87497a : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e813caf88064ec3b4b1cd28bdd4a5b2 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0764dbeb241c4acdbe29eb34f1413d34 : SUCCESS in 1m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3f608971a0f450d892d48ecace7d6ea : SUCCESS in 1m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4ec276729d0a44f3bc1cd020a6d1e366 : FAILURE in 2m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5b5ed871a15b4a2f99194173bec4271a : SUCCESS in 3m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6aa513352ec648279540bf0aeab8a151 : SUCCESS in 6m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/800cbc1efbba4814a376ec67c117bbb7 : SUCCESS in 7m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/88b7a4d3eb204563ac8119ad409e6da9 : SUCCESS in 9m 30s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ad47b6e710014b198d3107f5f4fe9527 : SUCCESS in 2m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e56c0c9346654a249c5246926f85d6cb : SUCCESS in 3m 00s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3984c6074b04499cbb2ac459e6b24ce7 : SUCCESS in 2m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/600d8803794841fea8e839d2b21b8da1 : SUCCESS in 7m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8682fa576ee54567b5224b0cf88106a7 : SUCCESS in 2m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6311dd0ae7a54a36b192d31a2e2b56e6 : SUCCESS in 2m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/69c81a49afc24a158d628a777fb7d08a : SUCCESS in 4m 21s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6bad1f9e4af2437f9324950d8f891713 : SUCCESS in 5m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34b0b3a3ab564d2990bcf9ea3e59b7dd : SUCCESS in 3m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a3140c5fc0b545fbb55f77625ffdb778 : SUCCESS in 5m 40s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8625d46665ec4ddbafaa3677be56af0c : SUCCESS in 7m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d5055433b3224f199987a52deff61e4f : SUCCESS in 1m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/12c6a29242694453b7e028893f16561d : SUCCESS in 8m 02s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/35337ed914c441b89600b550c4c3da69 : SUCCESS in 2m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1e0b1b842655458c8c93bf9d78472888 : SUCCESS in 4m 16s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/50ea0e1d50f346cb976f627daeddf41b : SUCCESS in 10m 04s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f670304f3a3b4a30acd51002b9c79d4f : SUCCESS in 21m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cf7a59d25324435fb3009ceb8dc6d786 : SUCCESS in 7m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5eb0c55179064d32bfa24be3e7046426 : SUCCESS in 8m 22s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7ec06d06c8474531ac21aba377abea08 : SUCCESS in 2m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2245e02aa6a5416388fd192ca56d96a7 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d2dfe1505c9415d8fd10e2f46b7f6f0 : SUCCESS in 29m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1c284917d13e439889e4b04dd16bf949 : SUCCESS in 7m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/55d594c00e4f4030ba7ad48043033033 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30fb173ac3b9476da147a668dfb3e68a : SUCCESS in 7m 30s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3e15c7cb403d41b8a542f5dbd6b9545a : SUCCESS in 16m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6a1194aca2d4445d986a4ec08dafcd3d : SUCCESS in 3m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/adb6e5fedf524cf6a087012748e2af7b : SUCCESS in 55m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0f4aec0b114423d9a7b91183db6d66b : SUCCESS in 1h 43m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b6bc13e0f29e40206c8f4a8a0ba2697dfa680bf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-10 12:13:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"da6d1a22486eb5817f87e80b22b6ea14b8fa12d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:14:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb9285ed467f46c1879045c30613df2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017bbf91edf3471b9cb5a189f374b2cf : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff94470afa9a4a9fab55e3ebf9aadadb : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59acadadbd4b4d5fac8bfc472f441cbf : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57a0cee66f954c7aa01065475f0994d5 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eacab894f9c44131b4688f57eb57db66 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3515089baa2e4615861489a9dc1b43a2 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4cbcb2d13fc0486499a85d4d41acaed4 : SUCCESS in 3m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6af56c0c9fba4a16b3e605616953dff7 : SUCCESS in 8m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0906ddac6e1a49a58e8d208206b9a853 : SUCCESS in 3m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3460f3e5463046159dbbdbf5d09578c0 : SUCCESS in 19m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0e9d15faf8c64a91bbffb8f621b8d945 : SUCCESS in 20m 40s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6d1373549e6c49bc98ef8fc43ef716d4 : SUCCESS in 28m 04s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/39a737bf95534175b09ec73df262aecf : SUCCESS in 5m 02s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f08bfc41e4984be693c2c66125047178 : SUCCESS in 4m 24s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d2ff16344f794c0ba52e21fdc62c5211 : SUCCESS in 6m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1ca475e0721c489cab72eaf164fcafe9 : SUCCESS in 11m 58s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/864f0df04d1f4aaea4b2922472043391 : SUCCESS in 4m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c2039266cbef46dcb6b4b09b4268cd00 : SUCCESS in 2m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5391695f077140858b810bd671a5048d : SUCCESS in 9m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/65b0ae13a6824b7687f5ab2c5162eaaf : SUCCESS in 15m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1693995101d348e1856298b5b88f5fac : SUCCESS in 9m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/988bfa4cfe5745eaa4ba4c88da25418f : SUCCESS in 8m 11s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f6f6d3ff127c4852a757f70a9d5b2e7b : SUCCESS in 12m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3f2e8b29ee9246f78777f265d3f4bd24 : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a0472710ad2241c4839b8ce5993e4cdb : SUCCESS in 9m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/78e8182814634420b6e7654e996b9bd1 : SUCCESS in 4m 36s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/13f210a437bd494e9258359e135b4d0d : SUCCESS in 5m 34s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3ea530809da84e7dbf873b99d20ca164 : SUCCESS in 10m 08s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2701edadd6e44042b14ee628353a4c02 : SUCCESS in 17m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8d4a09cb110c41629d299b54caf96e40 : SUCCESS in 9m 06s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/556f3546af9147d3a51ca2b44e320ae6 : SUCCESS in 10m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/036e00be2cb2467f88def1b1a121ef9c : SUCCESS in 2m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0414d66e6c034b6daffcddc0decb6fbc : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e18aaed7a564f0baa39bb1059569e17 : SUCCESS in 27m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6a00c063d06f4a26afd6b68e3aea26e2 : SUCCESS in 7m 14s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fde1c7f679a14720b819d528cf8cfc41 : FAILURE in 3m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d06faa0d1e1a459eb21d9c0bd944b4d3 : SUCCESS in 8m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/09803456bc5b4c6a87f5f09f024b5fb6 : SUCCESS in 6m 44s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/349dd1bbe52b4588a6fc46964806e691 : SUCCESS in 3m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ebb38e552b74e9cb38de8d5d2cf8496 : SUCCESS in 1h 04m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1f4f35817de46fc819ad8906b5cbe9f : SUCCESS in 1h 34m 18s","accounts_in_message":[],"_revision_number":2},{"id":"c76248d09fb05a69a58b2065158055ce7e6dc96c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-10 14:17:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9cbbd2a1bc60996ce38a2eb9e74009aba98c7c0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 16:05:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f56c2c203f4b18acfc672138ba82a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d429ea7d9bb14461b2d09d0c3d5289b4 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecf2fb6adf354bdf831ee1c8c2207825 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57a0d75fcfb14a6a8fdcc833a544a411 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71d976ac52bd4fa2adbb19d0dfdb1a73 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1823158b51d4a7bbe12b0745f100107 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70f3d8af742e43d782328ab50eed75ac : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f591be08b04c466ab604fc4d29e08d0a : SUCCESS in 3m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d6dc924f7f704fe090e3e873ccb6881e : SUCCESS in 15m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7c618cd0699f45c0be18759f506a22b4 : SUCCESS in 9m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2a71ef2203c142c2a6bcdea3779b0e17 : SUCCESS in 8m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/632e27596f24479bb6d49a64a620bc9a : SUCCESS in 20m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/db871da49659469dbdf047e60e2137dc : SUCCESS in 9m 35s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8504724d7bc8478b9b1f84ccccd139e9 : SUCCESS in 5m 00s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8db6ebfcbadf423a83745a123d14ac38 : SUCCESS in 5m 48s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5b1979a264d44b32b259c54cf2bbe58f : SUCCESS in 2m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7e034c353ddd4728a22476d98230211a : SUCCESS in 11m 38s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6a7a323dcd5647faa23bee7d5840e1cf : SUCCESS in 3m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9909673ade584ee49d68b037c9cd79ea : SUCCESS in 6m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d6b8b2421504491cb1507141e4bf9da1 : SUCCESS in 7m 30s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/827629be1cd64bb4892a40c2ab1dba1a : SUCCESS in 15m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ecc1553dcf34f599769525c8aec46cd : SUCCESS in 8m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8438fc46b28940ab85d00c7f3487ecd2 : SUCCESS in 10m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/75e5404b80e24e7c955ab15e4a2c7d36 : SUCCESS in 11m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7df1e8cb98c24786a9a8e65a23c3ab6e : SUCCESS in 3m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1bc3fab78e0c4a3481c8de57abd6d827 : SUCCESS in 9m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/a9b8db2d15e1406d8fa6ab46bcd15978 : SUCCESS in 2m 37s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4d12581220bd4c209f01197193c71bdd : SUCCESS in 5m 06s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e61c2c0db6a54b6eaf7fb60abafb1564 : SUCCESS in 10m 52s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5be7aa4b38ab4835b2213b663bf4dc1e : SUCCESS in 18m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/244cb91783f04d858f8320a2c4183688 : SUCCESS in 6m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5b2a66e8de4e4a9b833ff0c18ad7f086 : SUCCESS in 11m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/78e94aa1ae684741b6ef0f52d51588ce : SUCCESS in 6m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3fdacc1d529341a9a468e4f46c415a33 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/816d55afb4d24393bcfb023b48f87fc1 : SUCCESS in 23m 55s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0ca7758c30db47c99280ff224a8c3555 : SUCCESS in 5m 36s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/390530cc18b749f4a3526006e1dc4e4c : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3c29423371e4e848c8a89d49af110de : SUCCESS in 5m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6307f8a1b98948d6b47a7fe9ac60f902 : SUCCESS in 14m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/eed6c5e8b70f441cbe50a1d0c71ef60f : SUCCESS in 5m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44d0f99895e64aa2a0f70142a5402ce1 : SUCCESS in 1h 01m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/689cc27a1a3b42bbabda8cd28489703d : SUCCESS in 1h 24m 19s","accounts_in_message":[],"_revision_number":2},{"id":"eaf25a35cd62652a3153a47fda3d589b789532ef","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-10 18:40:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8f5c99cb21c8b1e77b4cbacea4e465fd2a2d71c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 18:41:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0f14675df6fcb013b6e16e1e79b3823e27b2d82c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:36:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b54aa703410459d90b67f619faa880e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d555c54d634a0dae890326dc84b555 : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e48de1f597f4632b4f19e095381a424 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35031e4e0c254c7ab74cf260d4aa99ad : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08ecd5141b22487b9541a751adef1213 : SUCCESS in 2m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca3bd74a1bfa4910932252fcd0bcf3d1 : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/055fa3a1aa3e4ec49eae1ad637d9a17d : SUCCESS in 4m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6aa047402afe4effa573e722d0eb903b : SUCCESS in 21m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/840f5cb3ba0642a58a55e0ff062c42f8 : SUCCESS in 11m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a03262108381437ab60fbaadb4dcf1a2 : SUCCESS in 20m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4265f45bf462476ab2ade119dc2f5bbd : SUCCESS in 12m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fce1592ac82d4e13b22d7002954650ba : SUCCESS in 10m 44s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/60c890a8428e4413a3defd785a855c79 : SUCCESS in 4m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/65322dcfe55344c2844ff28fc647f079 : SUCCESS in 7m 52s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e993bcdb8a1a4a4192b6708b88194645 : SUCCESS in 4m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cd8f639390b1422fb5d08b2c74970c18 : SUCCESS in 14m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40b73faabaaa45208eb2b309ad97c305 : SUCCESS in 7m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a4d46222db9a4208804011494233af48 : SUCCESS in 10m 04s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/84e1bb2c8a4047dcb8709b75a2b9aaa2 : SUCCESS in 12m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/807026bba78f41a4a8aa57806b0c1325 : SUCCESS in 1m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/95e32e6770674e1c815c3279f57285a1 : SUCCESS in 11m 19s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e7a6aa6bcb5c457db22f80355f8e96c6 : SUCCESS in 21m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4e0732b0fe084336bb72b8a8a73b9c48 : SUCCESS in 6m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/34adaf506c8d409cbf1b7c2dc2357bec : SUCCESS in 13m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/cb4c21fb84764dad8b1ccd0ce155bbdb : SUCCESS in 4m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fd74c5148d334b6284a0d866b8a5342a : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3b94655ab1f4495a3ba9c37fa03310a : SUCCESS in 31m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/398a4953415b4eb8aae5f936e4a2a963 : SUCCESS in 3m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dc0594e62bbc4ca6b4104647c442950b : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e4c0ae25573405eaf20fb9386634d44 : SUCCESS in 3m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/066081b421aa4bec83704cf4045efe89 : SUCCESS in 6m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5c3b2d2812e34fe8964c3e28d0874af0 : SUCCESS in 3m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ff92b8ae8cd4dc9a4de69c86701277d : SUCCESS in 32m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bdbaf4e342b410a983f895c4fe24b3d : SUCCESS in 1h 41m 23s","accounts_in_message":[],"_revision_number":2},{"id":"fc23ceac1837d4f6fab2ad0139ec2d9e6e67ffe6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:36:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1e8f554d6b369f0ed730060a5787a8a9736689d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:37:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f80a5b0bc07247cebe823affaa2a91d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ce35e6f5e124358b7a5400df30b61a5 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b0a5049434666ab905ea41dd9069baed0077dfd7","revisions":{"6f0f2cc057b3407111678ef06794f08697b89ef8":{"kind":"REWORK","_number":1,"created":"2025-07-10 09:39:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/954575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/954575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/954575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/954575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/954575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/954575/1"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:29:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:29:11.000000000","tz":0},"subject":"update constraint for aodhclient to new release 3.9.0","message":"update constraint for aodhclient to new release 3.9.0\n\nmeta: version: 3.9.0\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I29d8389a77fe0cbda9fa943fd71cb36a94de2e02\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I15780557b57c06ac1c474d717f3166a3504626ed\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f0f2cc057b3407111678ef06794f08697b89ef8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f0f2cc057b3407111678ef06794f08697b89ef8"}]},"branch":"refs/heads/master"},"b0a5049434666ab905ea41dd9069baed0077dfd7":{"kind":"REWORK","_number":2,"created":"2025-07-10 12:13:44.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/75/954575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/954575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/954575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/954575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/954575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/954575/2"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:29:11.000000000","tz":0},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-10 12:13:26.000000000","tz":120},"subject":"update constraint for aodhclient to new release 3.9.0","message":"update constraint for aodhclient to new release 3.9.0\n\nPin version 3.8.0 for Python 3.9\n\nmeta: version: 3.9.0\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I29d8389a77fe0cbda9fa943fd71cb36a94de2e02\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I15780557b57c06ac1c474d717f3166a3504626ed\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0a5049434666ab905ea41dd9069baed0077dfd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0a5049434666ab905ea41dd9069baed0077dfd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
