)]}'
{"id":"openstack%2Frequirements~954577","triplet_id":"openstack%2Frequirements~master~I693b37e9c327a95f4eb832b95f9ae29f14e18dfd","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-07-11 00:32:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I693b37e9c327a95f4eb832b95f9ae29f14e18dfd","subject":"update constraint for ceilometermiddleware to new release 3.8.0","status":"MERGED","created":"2025-07-10 09:39:45.000000000","updated":"2025-07-11 00:33:12.000000000","submitted":"2025-07-11 00:32:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954577-new-release","meta_rev_id":"354cab44d32898c1309a915e6d04001db22ccc3a","_number":954577,"virtual_id_number":954577,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-11 00:32:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-10 18:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-10 18:48:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 11:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 18:48:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b5d53982677c23ead2c16398efbb907a1430e120","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-10 09:39:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77cadeea3ba56ec3d67ce96e8d66462ecf7a3bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 11:23:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/edb6d7a1a63e460f8e70c4cbbb27a099\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da889d3fa971476c9a383a4f5d825714 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4167fcd57b9943ad9462230204a2668d : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/010b6697176a4534a609790acae9f7fe : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1375a30cc82f4998b36b685bea905508 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d53af7ea32134f9b9866493ba7d86c93 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a9bc2424acb4c4bae8cf4f5290ef4ad : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/734b77056c2f4592b2456a5645a07365 : SUCCESS in 2m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/805d250bfe184550a8d11a43ecedf0b2 : FAILURE in 6m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/24fdb3df5bca40b5989435cb8a2b3ccd : SUCCESS in 10m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/86730b263d574201ae4c2d7c8c923036 : SUCCESS in 12m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7d0ddbfea2b14ac4b9903f0536f03d42 : SUCCESS in 18m 04s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b265c2c393f9473bbc9801f1086eb013 : SUCCESS in 19m 17s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/059b374086b247f2a9fbe80098ca3775 : SUCCESS in 5m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/892ac372600a4960850f7b7c918793ec : SUCCESS in 5m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6ed40b588ee64f70b9c2c049da6bffe4 : SUCCESS in 6m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/77d0cb6e19ff4ebe9a4ff6b0bb890a24 : SUCCESS in 10m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/99b570a48d8d4fedbe6c38940c9242eb : SUCCESS in 4m 23s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b4dd7dd954844405a112b0e9f93904c1 : SUCCESS in 5m 05s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8fadf3f4c32f4e0dbf657e9bdb11e530 : SUCCESS in 6m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1f3a6a6cfbb24c00bef347a72b54456d : SUCCESS in 15m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ddd982694dfb4dc090c91195562a09e7 : SUCCESS in 6m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4352f1ad622b46cf9670203497a294a1 : SUCCESS in 8m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9a9d30f86a314b3ea70c492ea0ff6aaf : SUCCESS in 13m 15s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a42002eeaccf468b9f26c6b536040645 : SUCCESS in 2m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/29c8d923d3ad435592e65bba68bc4f46 : SUCCESS in 12m 31s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ec34a45c72bb4b7b9e41c8157b63da2d : SUCCESS in 4m 09s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/163300f7b8f145f894ba24a1f06f0c47 : SUCCESS in 4m 17s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/bf42b49d88c54f2fa7faa060d8a62d26 : SUCCESS in 10m 24s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/254cdf47199246a0b25d9ed305d5e170 : SUCCESS in 18m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b536e006d22047b79d27decc05fec351 : SUCCESS in 6m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d595d757eeac416591ab832d14e07338 : SUCCESS in 13m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3f618084a5a142c3afa6e1577a64858f : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b05239d430441b499bc05e43c898869 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eae94753808d4f06b5c6420cfa1af227 : SUCCESS in 23m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f632e6824e884630acadd9a0a1d3f888 : SUCCESS in 5m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8ef622e996f045b7b8c554b44cfabcbb : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b02c5c86d6940cf876a343fe29361bf : SUCCESS in 6m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bfc5975085dd47639079ee3d6fb40053 : SUCCESS in 17m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f643f5ed6d6d4dd99646bd87b4deb6ec : SUCCESS in 4m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f734cb3eea4440c807edd7a2eab30aa : SUCCESS in 55m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19adccc0282247c28238ab0d4c419a0d : SUCCESS in 1h 40m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ee6a5bb7b8ff6f0e16a752807c6ecba15a431fab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-10 12:14:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"56683d220ea19879f6c604e03d8420926d5c6a0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:22:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dda2dddfb5d045fcacd2c3b51eec8437\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32e609560a3f44e9ba9d3defaf7d8056 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4428336c11af4b7998f3e2756a4e6d9d : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c977da182e84311b2a920a2ad131bd4 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb06f796ee0c43989e4efff45c26543b : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/044b08a135ed4789bcb20e1595d55f2b : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5c909599a214320ad5defbeaad69047 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ddd5dc5dc4754bf5b7682576441280a6 : SUCCESS in 3m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2ad6e0b5b0ed43d1a6f51e6916054026 : SUCCESS in 15m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/418180122fa1460e801aeb46ae05bf15 : SUCCESS in 8m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fc2a75beaae64996ab06c7c7fa6b35e5 : SUCCESS in 16m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b0c93944d428405381bb05295266aaea : SUCCESS in 7m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7b2b3828bd3d4ad889603a472400c7e9 : SUCCESS in 21m 14s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d8d3a6c4ffb648829947d914f989ad06 : SUCCESS in 2m 54s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2c7ce4e3491140a492d458b0b93dff53 : SUCCESS in 6m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/416eb3f70ade43289860105907fb94e4 : SUCCESS in 3m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8991b15c36e9410bb594ef6c868c142f : SUCCESS in 11m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/867a1035f335470b90b1316cc6ac1dde : SUCCESS in 6m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e93daba77bc047a7ba6122236f72a89b : SUCCESS in 5m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5c5e61239132400e90d4754f847f32cf : SUCCESS in 7m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bec61a0cd1f5460ba9f6eb9026c4f229 : SUCCESS in 14m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/029bab2023f64b0ab352ee0001f811dd : SUCCESS in 4m 12s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/60372ff6e65343b58682e37ce67da2f5 : SUCCESS in 8m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4e0d24aae9634cde8d2696e631b164fe : SUCCESS in 13m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/45d855e1ccb0464f818e6928b7de4994 : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8ee031b71f95489e9363a93f8fe9121f : SUCCESS in 9m 07s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f58ad5d10fd7409997d0350066d3ff62 : SUCCESS in 4m 30s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/754364e00b22429183b0d3df758bfc26 : SUCCESS in 5m 11s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/975a074ad5194d5a8ede397937009ca4 : SUCCESS in 11m 19s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d780a17fd4f448dd9a8e78977ccd6d66 : SUCCESS in 20m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/120e1237b094486abda4338c0f573d56 : SUCCESS in 8m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1130558094fd449f9ffb9dd144491825 : SUCCESS in 12m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1ad403ed727b46d2957eea26f7702771 : SUCCESS in 2m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e18a6a37c5f94088b06b8ec1c978afe2 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4fc0476818848e68ca30129d202a56f : SUCCESS in 27m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/38ac47b0ec9a4dc89514f1ba9959f0cb : SUCCESS in 3m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/81e21677b50b47f896bac77d0de1bb17 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02f7d5288f524e1bb365055dd86829cf : SUCCESS in 6m 26s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d77302fd3ca24ccc9ec5ae49a46be659 : SUCCESS in 16m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d80ad6632bc942ffb3fc58de225a1fde : SUCCESS in 4m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ecd2c165d134ee48f275234c02442bc : SUCCESS in 59m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86609f396f7a4b14a01fecfb2666be46 : TIMED_OUT in 2h 38m 07s","accounts_in_message":[],"_revision_number":2},{"id":"fdb545e6cef5ec9f67cc9bf71e78a68111209696","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-10 18:48:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c95a12b86712f071a6096f8acc3b143b74963017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:40:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11b542c13eb2428a823dc6ed77176075\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e735fb90aa984a23964ceba4b9991cfd : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f659512524ad40419f93dab8e3c89f69 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45a8dd2ec8574df2b0854b21e86887e9 : SUCCESS in 2m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0de069e70e0843519b019d88dbbf59c1 : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84182fffaa1046ab9b2bddf85d1c89fe : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95b97b0c0e8c4e1a9d4e4a735b440282 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/24d1bfbd7430492b8cdf3508d8d8a8d8 : SUCCESS in 1m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c77e80c4f0584e42a7497a6b71308f0d : SUCCESS in 16m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9518a87c78544d72b5c93e23d2e05a31 : SUCCESS in 9m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c67649ab03da4eddad5375dea56be0d9 : SUCCESS in 7m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fe51993c72a4487a8942863872efb3fe : SUCCESS in 21m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c1e7faa70375451e864c4cd1b1cc1de3 : SUCCESS in 17m 18s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/814f10f80d984d4889f7473aad4d7b66 : SUCCESS in 6m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0fb7860b6e6747f5a856590c43d17814 : SUCCESS in 6m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9092a8cd10034fbbb0944818daa64711 : SUCCESS in 4m 11s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9e1b58dea9994621bcba8608bdc21614 : SUCCESS in 11m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/098542ee9b864b048c3590bd22fdbbbe : SUCCESS in 4m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2f1ab1c1468344fa95a963f29d633cd6 : SUCCESS in 4m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/72a4e838a7684d1aaf7f81e7c74a7d41 : SUCCESS in 7m 21s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/47f6ae864d3044e7acd1f4ff0840e7fa : SUCCESS in 16m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/82ce80348e5b42d6b38f6a3715b16a9b : SUCCESS in 6m 31s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8f40450308184079b81f128d4d0ae5e0 : SUCCESS in 7m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/49a0317ad63c48a4b02a1d8575b20dd2 : SUCCESS in 13m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/192abb00541141f190667e2b1c23fdb4 : SUCCESS in 1m 53s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d78755ed58b7456db157fa1a77a0ce76 : SUCCESS in 13m 50s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8f233dae30014dce9ef895f76ef0a9dd : SUCCESS in 3m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/259d0b02a585457e82c57046444657f1 : SUCCESS in 4m 11s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d59224e98c904f039f3a328558992653 : SUCCESS in 10m 43s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a9dbe9c7ed9b408d9f61b6014cfc80dd : SUCCESS in 25m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/59cc103b66e040e98eb6a5f80133548a : SUCCESS in 7m 18s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1c85d0d372b245f8932052e654ce4690 : SUCCESS in 11m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6bf43e1bfc534a5db6ea4f32d47a79d0 : SUCCESS in 4m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9cb91fc205ec4b8d8fe60e3876983593 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87fcc36803334f1881a4a82817a696d5 : SUCCESS in 23m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/52541e97639745909e0238d337ca3c4b : SUCCESS in 5m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/13c1d33124b546f98385701f7f01bd65 : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dce26cf76cff4e078e080a0dae2a58be : SUCCESS in 5m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/27499ec3fd1a4d46a719324f8942d2b8 : SUCCESS in 6m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b695f891e7db4a86bd56fe7d0a9a3526 : SUCCESS in 5m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbbd950a8a154426a298de9d3c717098 : SUCCESS in 54m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ade60b4ae27e4db2bf8d2de15190de28 : SUCCESS in 1h 39m 20s","accounts_in_message":[],"_revision_number":2},{"id":"360e2c8c7d11e63498410b797ce792743008e01f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:41:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8ddf9d3b9d7f2278e202315c3f51772725d524ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 00:32:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c07bc6c7b064440292341bd5aaf8d871\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2771f143c21b478ebe4bd7556ffb19c2 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c2e8c460404b4991469abfb4fb9f0e : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f1affc666b94d81bfd57d0677bbbe7d : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ea1eeba2c034268a1a42734a80d9988 : SUCCESS in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eeff16a9487b4e3a8de47d6f33280862 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36310bf57e6a403f88ff48ac41941b34 : SUCCESS in 3m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f7a33126e33d43ef8bf9dc62a03112d0 : SUCCESS in 17m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/589b41650f9f456ba8b1eac43a68190d : SUCCESS in 9m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2b9c0d85411b4a3d84e0ebf87d8d4229 : SUCCESS in 17m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/995ad015ed454add8a4b8eb287a9b7b6 : SUCCESS in 20m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6a30b415be1b4d95b01b3b551e5f0f6a : SUCCESS in 12m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bb118ec6fd2141b49b81ba66f2db2b74 : SUCCESS in 4m 48s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a80c27f63cd646f1a7d911146207705b : SUCCESS in 5m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/698c51bda08d4fa9adc3b2aeaf90c1eb : SUCCESS in 7m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6ccd92e948f84f0ca903f73f1b5cebe1 : SUCCESS in 15m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/be76de9241884e25b794d4f5de9f473a : SUCCESS in 6m 31s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bee2ea3e64d648a59732ac2e0c66c41a : SUCCESS in 7m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b63c024eb2854713aebbe7ecd1dcb2f7 : SUCCESS in 12m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5aa95771b0c24e1982c569120906908f : SUCCESS in 2m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d3deff4525fa4d46b95a24372c03a084 : SUCCESS in 9m 48s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/22c06d081c554bde8c3587edcb5ecb44 : SUCCESS in 20m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d96242fac914ce5ae262a3f631fdeb2 : SUCCESS in 7m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4a5da116b9874d819fea7af9920f076c : SUCCESS in 14m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d8acd0ca80934054a5144045a80e9893 : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a26878c1844749549f3590c368d86c5f : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f7792cdfd064d3a831830271f6266b5 : SUCCESS in 29m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e0db2189cfa1472d8d4f7cb8d88a7b47 : SUCCESS in 6m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/62cd755c1a284a08877e530cfd619484 : SUCCESS in 4m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3cf95b84f50e470d8904d1fb8624463d : SUCCESS in 5m 52s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3ee69ea3e70d4475a12306881137ccbd : SUCCESS in 22m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ee4f26f46b6b4f5aa6ee35cc1127898f : SUCCESS in 6m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a475786fa004c44aac7e8143519821b : SUCCESS in 50m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af7afa3dd52745f59dadd26d0f4c0e30 : SUCCESS in 1h 36m 52s","accounts_in_message":[],"_revision_number":2},{"id":"bdb2c8feb0ba9fe75e7c5bb974a41f0081c21382","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 00:32:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"354cab44d32898c1309a915e6d04001db22ccc3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 00:33:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7821d63a184447cba6e8333a5a4193e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7de1ce5863ac45fcb88abe8388839b8f : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9af63fa09d48ab2176a3dbfe4e17176428e3355e","revisions":{"c601a6d177893dae9b18345176d50eb3e71e1f48":{"kind":"REWORK","_number":1,"created":"2025-07-10 09:39:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/954577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/954577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/954577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/954577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/954577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/954577/1"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:29:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:29:43.000000000","tz":0},"subject":"update constraint for ceilometermiddleware to new release 3.8.0","message":"update constraint for ceilometermiddleware to new release 3.8.0\n\nmeta: version: 3.8.0\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iae91084234b962cbecb910d5459933a481b26dd3\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I693b37e9c327a95f4eb832b95f9ae29f14e18dfd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c601a6d177893dae9b18345176d50eb3e71e1f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c601a6d177893dae9b18345176d50eb3e71e1f48"}]},"branch":"refs/heads/master"},"9af63fa09d48ab2176a3dbfe4e17176428e3355e":{"kind":"REWORK","_number":2,"created":"2025-07-10 12:14:50.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/77/954577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/954577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/954577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/954577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/954577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/954577/2"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:29:43.000000000","tz":0},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-10 12:14:33.000000000","tz":120},"subject":"update constraint for ceilometermiddleware to new release 3.8.0","message":"update constraint for ceilometermiddleware to new release 3.8.0\n\nPin version 3.7.0 for Python 3.9\n\nmeta: version: 3.8.0\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iae91084234b962cbecb910d5459933a481b26dd3\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I693b37e9c327a95f4eb832b95f9ae29f14e18dfd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9af63fa09d48ab2176a3dbfe4e17176428e3355e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9af63fa09d48ab2176a3dbfe4e17176428e3355e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
