)]}'
{"id":"openstack%2Frequirements~954579","triplet_id":"openstack%2Frequirements~master~I8b7bde7957b7eaeb73d19186b88b197ba4f224e0","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-07-10 20:48:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b7bde7957b7eaeb73d19186b88b197ba4f224e0","subject":"update constraint for mistral-lib to new release 3.4.0","status":"MERGED","created":"2025-07-10 09:41:47.000000000","updated":"2025-07-10 20:49:15.000000000","submitted":"2025-07-10 20:48:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954579-new-release","meta_rev_id":"23f9732d8f326ffc7a168f343b28cccbdccecc4a","_number":954579,"virtual_id_number":954579,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 20:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-10 18:41:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-10 18:41:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 11:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 18:41:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"8c714fd5b36ea17f4ce3c1a459404f4b6f945e3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-10 09:41:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df01e4410defa86e5479114dd2d8970a74baa1db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 11:35:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68046ab8154b4a23b8ff069dff609c93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55680fdefda42cca672e4f6f34d452e : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2447ff4a6d47fea4a93b011a847cac : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/432db1f39f5e49cc8eabdcca41ee05e1 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b170bfbe429b4dc884e2ab0af8eeb9ec : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e09238ffd404949aead48f0f20531bd : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39bbc87fdd264f639ca82d0fa74ef27c : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/453afac88d9b481aa7f9699f328d08db : SUCCESS in 2m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0129fb5b08de4310a7365e91dc7c8b51 : FAILURE in 4m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8cf1ff7012c34f8580f85ccda83b0e35 : SUCCESS in 4m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/976b947bf4ca4c80b44ca2734cb56d05 : SUCCESS in 16m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a858cae61fb14fc3a53bce90029717cc : SUCCESS in 9m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9e682f2caae04ca9838bed72641abfbf : SUCCESS in 22m 32s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9f352b4e7214477fb20bc7fe7fb2a6b9 : SUCCESS in 4m 50s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a6c68333fe0f4af9843a2d0463f24ef2 : SUCCESS in 6m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b65177ab256e4ee7af7e7465a684a6e8 : SUCCESS in 2m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7c8a310220ec485f81ac2f77883a6531 : SUCCESS in 11m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/58027a02edd8465892c6495e4a8ea638 : SUCCESS in 5m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ebdb8f7fe6d64e81adbbbe29184d3e9b : SUCCESS in 5m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/597e0243fce74df6be05505af73f730c : SUCCESS in 7m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ba9e5925723544a8abbc704b924e137c : SUCCESS in 14m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad21e668a8d7416fac72bc6161a173cc : SUCCESS in 5m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/be92a5c9835749609511584eb8a48690 : SUCCESS in 7m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5e4a11f78fbf47d7b45278c844bc9f4a : SUCCESS in 12m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dbefc51b7dd643a6888674dd7d45ff09 : SUCCESS in 3m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/27aac85200f8418ba25d3cd0497323d4 : SUCCESS in 11m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/29f7bd0a5baa45dd8607dc0624bc9c3e : SUCCESS in 6m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0f5f448e05d949e4a4b6deca0b20e8be : SUCCESS in 4m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8e25bcadfcd142cfaddf6ed655c08d9a : SUCCESS in 11m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/706906b65cd94a568f377fa2249b26d3 : SUCCESS in 16m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/85269e44b0634ee993707fbf2c307dfb : SUCCESS in 7m 06s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7a46d164d9204814927a48ed3d1f8f71 : SUCCESS in 12m 16s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/dd931ef3e5d3476c8e713adeae570f1c : SUCCESS in 6m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0cd435e2343f455d976abf1af83bdb8c : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c6de815097b4cbc92a050482565bd1b : SUCCESS in 38m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e8e8623114904c3c8f4ea75ca5d98346 : SUCCESS in 5m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b4aaca0069ef45f0913dd0ec64c88375 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e0f767047054486da21a3ae7f8ef1db8 : SUCCESS in 8m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/713ecc7eea62462ea863e622c65a3863 : SUCCESS in 15m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2052639fc1464fcb9b91fe79cea5e25e : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eafdd6a3a16b4791b8273916cbb5ccd4 : SUCCESS in 47m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d2bd1fe6c43406d943b4b475a5015a5 : SUCCESS in 1h 50m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5e8c51607c048cb3201f8677a2a2edbd75b558b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-07-10 12:08:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5ebbcf56af5c183e0c53f859957a9d5b83953d76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:35:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8448dc192d0948968c630b912b5dd66d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d3313885f1742e7a2a9b83f6620f2de : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d211888bf444c96a1342a119f643d8f : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0fb801fb22b4f489ab4febae5bcf9af : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a0893e55ba544fdda475b4ad56377258 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5caefa5626684d4aa03e676ee1eeb0a4 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3480a7f17273409c8d4a94e463cfd71c : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b83f4425522437cb3dfef69592f91ca : SUCCESS in 3m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4cc2e111da4144869da626d611ee5e95 : SUCCESS in 15m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/507edcbf307a49d68bd22e4cf7f5f1e2 : SUCCESS in 9m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bccc13995ad344d0aa2a20ac993fc589 : SUCCESS in 21m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ab354d771e2341ae84a7f212a23a572d : SUCCESS in 23m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9a9d06e7c07e428398ef0dc0e6ccdff6 : SUCCESS in 26m 07s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7f18c73273894d4684aee61e72117e9c : SUCCESS in 5m 16s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e7f023c522d54363b7dce5ec78e7c5b5 : SUCCESS in 5m 56s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ae64edd04fd54042a6c99444444dbef3 : SUCCESS in 4m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b980276bd07349369cc466c30bedde01 : SUCCESS in 11m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/33535efc446f4659ace1df42584bfdc5 : SUCCESS in 5m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7e4c042fb08c4ad28aadc08935a48840 : SUCCESS in 5m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/26b764d64fa64d19860612fa6f895dd6 : SUCCESS in 8m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/742c476a9d044d13a57d0aedb72f77ea : SUCCESS in 14m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0c0d7bba0814dd28731b946a3791783 : SUCCESS in 9m 50s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8fd89f2d95854f65a1fd14172b3ddf84 : SUCCESS in 7m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/950fca1a815e45b1b1b544ea8ce706aa : SUCCESS in 13m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dc49485713684a8587ce6ac7a0db48ec : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0cc46b849f744f878d12870d11528b7f : SUCCESS in 9m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3a6719d292f243c3bcc7064c2e653ba3 : SUCCESS in 4m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cd9f1776ee6a4501813ebb9c41c4016c : SUCCESS in 4m 44s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9ad6fa598dca42e1aafbc74523176e06 : SUCCESS in 11m 29s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d377cfdd734f48cd8ad8481731a13e7a : SUCCESS in 20m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/929748fe40304005a5127542ed4eb984 : SUCCESS in 6m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ec2abbcbe6fb46ada5e3abd2ca48d6a0 : SUCCESS in 16m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/41f0232969cf49dd8725b7f249f30314 : SUCCESS in 5m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a41690ede4504f80bc32e5d2d3108d2b : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50880eff23b74a34b974f1c96988f44e : SUCCESS in 24m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/31be50a1daa4442ca1420802a6f81027 : SUCCESS in 5m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3a67af75d24d4fff82da20ff73d49355 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8490163c708543339ea16de11a0ebe4d : SUCCESS in 6m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7c6a5601e00c4471bd846894143879b3 : SUCCESS in 14m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9f409b3764a24af5b7e526605c1399cd : SUCCESS in 5m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddd7360edf134966a74a0422107aafef : SUCCESS in 1h 05m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/122d88f0adc645158070db79210abb77 : SUCCESS in 1h 41m 33s","accounts_in_message":[],"_revision_number":2},{"id":"f515cbc0e81b1f44720fd8b572f55fab943a4155","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-10 18:41:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99c8dc29a782dd9478cc2eeb91945514215dd658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 18:41:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f2d41c53223109254133653c642bd6091382b4fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:48:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8722a6656cde4bf49528dcfaa58e1b8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba1a3382f0304c4d800782d176b1c11b : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b93e80c8fa64af7a7b36c150168c4ce : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3584ac3bb2b478e9143965fdda36e3f : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13833c8a74d24193a8ae481528526620 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3093ff37961546a8a374c6a195412ae3 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4cfbb263b354e3da9e28c2f1b00d3ac : SUCCESS in 3m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/13998eebb833489f89fe10f457fd883d : SUCCESS in 15m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f089903a2ba446b5a86aee57c12cb364 : SUCCESS in 7m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9412da42eb264fa2a84632dfaac84522 : SUCCESS in 16m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e75aa434848e421a9a6cf34d2f3eea75 : SUCCESS in 20m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fa8dbcbbc470458b833032fab16b127c : SUCCESS in 9m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9c0f30b06e49423d8e51cfc28347b9c0 : SUCCESS in 4m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/203f4d3f892d4661b9783c9ea9558972 : SUCCESS in 6m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/dd10033794ac4805bdb535dd07fe339a : SUCCESS in 8m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5a01b695f2864fba85f7f3a8030bf2a3 : SUCCESS in 6m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4cae637e4a294b2bb07cd9f0049b6aab : SUCCESS in 10m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fe2c5664083f44c0bb2e4f1ec79150af : SUCCESS in 9m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f61baaa5eda6406db96d461409e01e5f : SUCCESS in 11m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/648ecbe6835f464ba6238243f4b34e68 : SUCCESS in 2m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1240b83dd37e4e4cbef8b37bcabc91c8 : SUCCESS in 10m 21s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ee4287957f1844c1a56644484642498e : SUCCESS in 21m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5384f7c34a334c41b03c112a08e1bae9 : SUCCESS in 7m 56s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fffa281bb3a34fffa7b5a7ece959d96b : SUCCESS in 13m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6b113a115f504964aa2dde0e182415d8 : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/083c2b0008af4b42a9357781c59ef97d : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e53a32ee9af4464593d766257c50a077 : SUCCESS in 24m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f00f41f6516840618e11ff34a7256c15 : SUCCESS in 5m 21s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/80f3406751f742ca8c8e4da6db34637a : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c27daa1702be4493bbd453a56493b794 : SUCCESS in 7m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/040cbf3d17494771adbe17eaa4c412dd : SUCCESS in 17m 52s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e997eebbe1914849a252e1a3f4909611 : SUCCESS in 5m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47f0b0d5211a4315866dbefc57512bb0 : SUCCESS in 1h 06m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25c782cd6ae844008968aa97b18b8522 : SUCCESS in 1h 46m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0ee1f2fde5afd568ac09c64658c4496dead8356a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:48:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"23f9732d8f326ffc7a168f343b28cccbdccecc4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 20:49:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81aaf4faacb2415984e05ed994970695\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dcb32a3e5b947afa7b4788a5e754cc5 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"130dec49ffa88043bd888ad40add57dfc6faacba","revisions":{"7c4dcd39cf994476209553ceb5fb120594ab75a3":{"kind":"REWORK","_number":1,"created":"2025-07-10 09:41:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/954579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/954579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/954579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/954579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/954579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/954579/1"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:31:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:31:45.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 3.4.0","message":"update constraint for mistral-lib to new release 3.4.0\n\nmeta: version: 3.4.0\nmeta: team: mistral\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I56118948fec93a9990304b5d93b0f19e01a2f9e3\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8b7bde7957b7eaeb73d19186b88b197ba4f224e0\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c4dcd39cf994476209553ceb5fb120594ab75a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7c4dcd39cf994476209553ceb5fb120594ab75a3"}]},"branch":"refs/heads/master"},"130dec49ffa88043bd888ad40add57dfc6faacba":{"kind":"REWORK","_number":2,"created":"2025-07-10 12:08:59.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/79/954579/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/954579/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/954579/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/954579/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/954579/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/954579/2"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:31:45.000000000","tz":0},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-07-10 12:08:19.000000000","tz":120},"subject":"update constraint for mistral-lib to new release 3.4.0","message":"update constraint for mistral-lib to new release 3.4.0\n\nPin version 3.3.1 for Python 3.9\n\nmeta: version: 3.4.0\nmeta: team: mistral\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I56118948fec93a9990304b5d93b0f19e01a2f9e3\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8b7bde7957b7eaeb73d19186b88b197ba4f224e0\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130dec49ffa88043bd888ad40add57dfc6faacba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/130dec49ffa88043bd888ad40add57dfc6faacba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
