)]}'
{"id":"openstack%2Frequirements~954583","triplet_id":"openstack%2Frequirements~stable%2F2024.1~I561c0008b08ea7a786bbefbf230aa372a3bebe08","project":"openstack/requirements","branch":"stable/2024.1","topic":"new-release","hashtags":[],"change_id":"I561c0008b08ea7a786bbefbf230aa372a3bebe08","subject":"update constraint for os-brick to new release 6.7.3","status":"MERGED","created":"2025-07-10 09:49:02.000000000","updated":"2025-07-11 05:09:33.000000000","submitted":"2025-07-11 05:09:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954583-new-release","meta_rev_id":"403b87cca2a944280c01e7f926ad749272bd6bb9","_number":954583,"virtual_id_number":954583,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-11 05:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-07-10 11:55:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-07-11 03:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-07-11 03:33:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 11:23:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 11:55:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-07-11 03:33:57.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"621912b1c36f64c5a3bb4c6bbb2dc87ff2cddc50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-10 09:49:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5584909d3abf62ed3b78236abd35ec69c6f12461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 11:23:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12a1c8d836dc4693ae90c2e9c45d5877\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f669bc1dcc4736b7c42cb9a1d93e2f : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec83ae635c5240ea820496ebd16f72f7 : SUCCESS in 2m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/851abb80973740a3a830d48f5fbf40bc : SUCCESS in 2m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b514cbe31404e808c0d32e84edffb42 : SUCCESS in 2m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4706e50376b44239b42987dcaefb0858 : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6745d8ad700d49b6b3e4dc01d55a4752 : SUCCESS in 6m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c1f50259a9864308bd0e82d45a0273c2 : SUCCESS in 13m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8d1739497b5044f99023a01896bea8e0 : SUCCESS in 3m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fe42f34f4a79428996aa60b20a9deec9 : SUCCESS in 17m 32s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/fd3158141c8644c9a595b798adbcb68f : SUCCESS in 2m 56s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/dd11c036a3d348f494c3b8172bac924f : SUCCESS in 7m 37s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f7beeab3088949869744f54e49526230 : SUCCESS in 3m 42s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/f3ab0611ae2f4bd3b7839f949f3cc08c : SUCCESS in 6m 55s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/09f3bb2356224e9183741247c09e423e : SUCCESS in 4m 18s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d7e3e3654ff84ff8982fc340e31f4603 : SUCCESS in 5m 05s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ad46303e950c49c1b149654967d97949 : SUCCESS in 4m 11s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/1d955748206c4491b8b360497b46358f : SUCCESS in 11m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/608340f84cdf4098a495afe2d38b6092 : SUCCESS in 5m 09s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d93bdfa629d94452b5819633a2c6d53b : SUCCESS in 6m 45s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/8d873d4a8b2f4685a0d4202969b2b764 : SUCCESS in 6m 49s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/6bf7b2ff72bf4e4aab0ece92ff7e71f3 : SUCCESS in 2m 14s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ed1e82af0e0840a186f32cb62a66d5dc : SUCCESS in 7m 17s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/fe695eeac3154d3db2affdb9b55afe4e : SUCCESS in 4m 31s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/bb87a080f60f4198a7a2a98bc06760cf : SUCCESS in 3m 52s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/70fcac6bd3194088adca06c3cab21842 : SUCCESS in 9m 11s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/98d02d05d22e47ebacdac07708e95fc7 : SUCCESS in 22m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2d44a25165cc402fa3773f6e5bc847ce : SUCCESS in 3m 21s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/72f79bc828e1493fb826c6d2fe6f65e9 : SUCCESS in 11m 50s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/047ca3c375c54e2bb284f994b3d3c9e2 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5883fc3849b44c919d6a63a17110c8fa : SUCCESS in 2m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d01f073a751648498854a7a8075895f5 : SUCCESS in 20m 39s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/9f8ef7d2e1b14c639fd332063bc203eb : SUCCESS in 3m 52s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/192418ab359847e194b57d1596bb0c06 : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32dfcc8a34cb467f81da73a9cd423fa0 : SUCCESS in 5m 06s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/be2c98b811a14eda80eba121af23d257 : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2535c2a7753c4945bc9b0eeceac58819 : SUCCESS in 50m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01211692e18c48d4a09c72bf140c96f7 : SUCCESS in 1h 28m 47s","accounts_in_message":[],"_revision_number":1},{"id":"434be904bbff970ab312741dd0d1f74d7bd3fe84","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-07-10 11:55:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0aed98be9f0cb0066429ce4748d520a42485fe3f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-11 03:33:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d28b96187eb2c4c5a5fd309b0a3d2a017e90a242","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 03:34:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9142469851a889a304e1a1a5a106c945ca0b9e22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 05:09:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad5917dff5e44e85a139c467b6cfbd7f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a625903cda984127acf16f293bf6bc83 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4106b3e3d917409c98abc18f594331ae : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61f26f386cdb4e4697858dea009fe2c7 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a1ca251876843c9891ac0f02951d4af : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a06bfeddf634579809ac1ce86843787 : SUCCESS in 2m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/34151f687a5f4f938e95e03bc82ee5ca : SUCCESS in 12m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/03e9f9ccc54a4c0dbc914620ac4e140b : SUCCESS in 9m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8a27837eea1c48998291582745a6015e : SUCCESS in 8m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bf32ce3b0d10406cb093481530262dce : SUCCESS in 16m 52s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/933bf75c00634363bf8f15ba7c38a6b9 : SUCCESS in 11m 31s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/972b441170ec46c38597105f0627a9cf : SUCCESS in 4m 24s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/2be84954e8d742b4acf494e224c92294 : SUCCESS in 4m 48s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/589e6914c90d4adc81ef90f81eecfea6 : SUCCESS in 7m 43s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/f190180d27a845d9a768ab33936c170e : SUCCESS in 13m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c4baa20fde294d16855d79db43bc7f06 : SUCCESS in 8m 59s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/5cecd4bee84a416e847610255af2c2bb : SUCCESS in 9m 28s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b6b87235d1c14cccb45e1eb2586ffe15 : SUCCESS in 11m 57s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b9c09d2bc0bc4c3fb89072717a88eb8d : SUCCESS in 4m 13s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1d2c2a1f8d1f45b2b9a3885569410b37 : SUCCESS in 8m 09s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/daef0b8b3dae40c7b316e81381a75be0 : SUCCESS in 22m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f4ad2846a68e40aa8d8cb774ceed3d6d : SUCCESS in 5m 55s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4618667df74b4dd1ba65b565f0f26410 : SUCCESS in 13m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/0a0c845fd02a446fa62bf4dbd16294e8 : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3ba6037bdfbd4b25aac9c29c2cb43c79 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a761d21eaa71456999c4384429b889f4 : SUCCESS in 22m 49s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/359d8fe7d8ab483893536ba75719e67f : SUCCESS in 6m 12s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/8189b90579c5428ab4e40bbd78ccaa82 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c13ff6702654dea877cf5ba48815909 : SUCCESS in 8m 38s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3faef33f342c4226a7d7fe235be8eb7a : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c401fa7e0024146b6790eaa20ed8c91 : SUCCESS in 56m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b21e94b055374f0fbf4eaa2269647535 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":1},{"id":"403b87cca2a944280c01e7f926ad749272bd6bb9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 05:09:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d07a0364d4957683b9d17876afc5254cc4a9592","revisions":{"5d07a0364d4957683b9d17876afc5254cc4a9592":{"kind":"REWORK","_number":1,"created":"2025-07-10 09:49:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/954583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/954583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/954583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/954583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/954583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/954583/1"}}},"commit":{"parents":[{"commit":"a3832313b21dec85f1a8bb5f9330f4d50c60b621","subject":"update constraint for sushy to new release 5.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3832313b21dec85f1a8bb5f9330f4d50c60b621"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:39:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:39:00.000000000","tz":0},"subject":"update constraint for os-brick to new release 6.7.3","message":"update constraint for os-brick to new release 6.7.3\n\nmeta: version: 6.7.3\nmeta: team: cinder\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: stable/2024.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia897f1d3ca16d82fca543eca14928557ce2d8381\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I561c0008b08ea7a786bbefbf230aa372a3bebe08\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d07a0364d4957683b9d17876afc5254cc4a9592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d07a0364d4957683b9d17876afc5254cc4a9592"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
