)]}'
{"id":"openstack%2Frequirements~954585","triplet_id":"openstack%2Frequirements~master~Ic40221d92bb1b2f9b42e9817ca1eeaf0090d8983","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic40221d92bb1b2f9b42e9817ca1eeaf0090d8983","subject":"update constraint for oslo.middleware to new release 6.6.0","status":"MERGED","created":"2025-07-10 09:58:51.000000000","updated":"2025-07-10 15:43:37.000000000","submitted":"2025-07-10 15:42:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954585-new-release","meta_rev_id":"502cea06d4a3e20c2df6d5e1cf153f5f990253ed","_number":954585,"virtual_id_number":954585,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 15:42:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-10 14:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-10 14:01:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 11:56:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 14:01:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"2130076e704674a718fb2a2ff1de8fc226f7547f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-10 09:58:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3b986d1c3d5a3046ce0e34ae1caf2969f993a98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 11:56:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26a0f740913e42128673b242541a657f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6e7213ce6f4d17890792906bd929be : SUCCESS in 2m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/675822eefd3748d49d1319d8edf0b82f : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7687a8af856434181532e573f9e2b79 : SUCCESS in 1m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e374ce7cc3a24167be4f24171a7633d8 : SUCCESS in 2m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b908da7f847548e8ac8dd093c3a98cff : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68f813baeb554f61a30e18d20de0c716 : SUCCESS in 1m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/785414b1efe04ec58ea7b282f1bbdf12 : SUCCESS in 2m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fc9138bdf7804fde8bcff73f00de3eb6 : SUCCESS in 17m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/daacbb4d32cd4d3781cd239f935cf8ea : SUCCESS in 6m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e8654fe5c1234cb1a966bfbaaccb34f7 : SUCCESS in 8m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/67ea5b42245c4d44bec8c80d28d6f54b : SUCCESS in 18m 49s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0cd602d230ca49c98d7e7fd28c7e80bd : SUCCESS in 27m 08s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/89c4bba3f69b48739c64304993d79420 : SUCCESS in 6m 26s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7963606d2cc0461fb6d949bf5ba06d63 : SUCCESS in 3m 29s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/df311eb21fe9401490537567f4833d30 : SUCCESS in 4m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/64699baa0e1b4d688cc72679911e9712 : SUCCESS in 8m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/508fe59ed18f46fe908d8f6de7f6d1d8 : SUCCESS in 3m 44s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3e670c1055814891a1e0c9f6f9d9d4a4 : SUCCESS in 5m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0e69cecf571644bda627d5b543e08390 : SUCCESS in 8m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2d3db9e3592d45508b6f66fdaa82be24 : SUCCESS in 14m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e926428b072a43e89d2d621906187f10 : SUCCESS in 4m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8a4859f35e464368a2430a66b9ddd0df : SUCCESS in 7m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/197e4aa1c567406aa5dbe46c8ce26ec3 : SUCCESS in 9m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8e77df1c2b0a428cbc0d4457fd409b93 : SUCCESS in 1m 51s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c70c32af363c4f38bf221eabe5bf4cdf : SUCCESS in 10m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6fa8d8d70793418183bdd82d36071794 : SUCCESS in 4m 01s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/55e27677818849a9b64096bb64bbc7dc : SUCCESS in 3m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7f92740030294f22bb90f9455453ad41 : SUCCESS in 11m 42s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/dd000a51e22144438d70837625aa0f6c : SUCCESS in 21m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c80edb1db4234ab1820353226c943264 : SUCCESS in 6m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9c626aefc430464c8f81546e7103aa8a : SUCCESS in 12m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a62a9eb4d1314ee8a8ebc0e7dff27d50 : SUCCESS in 4m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c03f7fa19f364b14b64ff81ca8f86628 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f914d93007ca48ffad90b6f85825cbfe : SUCCESS in 25m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6379790a0fc04366b7b5d91143105fcd : SUCCESS in 5m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c875a85cc5b5477b9bc971d2762dd2fe : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f9b7585d9624888a9c668ce25986325 : SUCCESS in 5m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/273a552991144e3f93911b6ba2bec64a : SUCCESS in 16m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/192fcba9e8824ca98d3d135be63f355f : SUCCESS in 5m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ba120692e964195b4fc77dbce4edb01 : SUCCESS in 1h 00m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b11358fce6a47a396ff8e532675a84e : SUCCESS in 1h 51m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0af2e5120ce9bf11ff7b7460589384b981eeb082","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-10 14:01:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f77ecf7324c3b655b5804b8556b345848857d578","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:02:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89e81839c6181a622928ca753962cc073901a719","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:42:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9b9b14dc74245af9e2fc951b141ec85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c66ce248a5c84a3e83b16566f748082f : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21cd613bd261472e84d2c347976984c8 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24398bc80cfc4f4db724f511d4ba4dcd : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a6858d641104062b0c697ee63b7e7f4 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a50ad332e3c40c7872c43d70aeec120 : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/20f078087b2b459d931b1f054b2cfe54 : SUCCESS in 3m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c8cb34d342514402a2a7822c39fc5b0e : SUCCESS in 16m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c551978359b34b6ca43e52bf31933e09 : SUCCESS in 4m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d0230a9d4c384926b1dee6c98cfd6a01 : SUCCESS in 23m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/05827286e21547949e0c20c3efe6e4e1 : SUCCESS in 11m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6d6a1aae5e514ea59afe84525f3668b7 : SUCCESS in 12m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bada3860e71d46dd84d0050b0cbb3f28 : SUCCESS in 5m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a2d64767e763469e8489c8893cb3852d : SUCCESS in 4m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/706e9368da574514b16db91358251947 : SUCCESS in 4m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/775e7694f41d41bf9526c3b75b82e27d : SUCCESS in 15m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f4f406690a9e4eaa9b72240404bdfaf4 : SUCCESS in 10m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/be430b0300b74c2dbb12b0cf158bdb02 : SUCCESS in 9m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6c40652e35f64a36b375592e5b452259 : SUCCESS in 11m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/66edbd30f47847349c6cd8e8724b08d3 : SUCCESS in 2m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/08b1ec30e6b246beb93502e942693ebd : SUCCESS in 8m 40s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7ec24030f9b64f3d8a1fe7188ae582d3 : SUCCESS in 18m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3f1a44952ff24b8a85d6d874e7bb98ec : SUCCESS in 10m 13s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a25d12af791542d594d1853cc4c89e77 : SUCCESS in 13m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/47a4cfc04e01463abc1064caa8017667 : SUCCESS in 2m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f6b2808c20824e7784e4e5f1bca155c5 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/60e2992d8df34ed582ac23f78840b683 : SUCCESS in 25m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/54ae627cf4534186a2ffd206dccaf1f3 : SUCCESS in 7m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c0e8125ff65c4c6c8c549ecb9ed5e17b : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82a38a372a1a44b4a55adf16870c163e : SUCCESS in 6m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2a02bd2e51ce4d65ab8a600a80748ec8 : SUCCESS in 17m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/86e11be951014084b563fc45cd713b17 : SUCCESS in 5m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ce38909180a40438caa6253a76cb74b : SUCCESS in 1h 04m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f89190138a5b444982d1fbd9a22c9ae6 : SUCCESS in 1h 39m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ad793f1ae9fcf20a4bf0f101224c4721d221c707","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:42:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"502cea06d4a3e20c2df6d5e1cf153f5f990253ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:43:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c61717a80594a78bcd63c4996ee2044\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5163f6e1144bf384cf94c73040a0ac : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0195b0ed92fc07f5961318f6a44efb848bb21383","revisions":{"0195b0ed92fc07f5961318f6a44efb848bb21383":{"kind":"REWORK","_number":1,"created":"2025-07-10 09:58:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/954585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/954585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/954585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/954585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/954585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/954585/1"}}},"commit":{"parents":[{"commit":"1515ee60df1b366e5cd401f2e8a326807204c732","subject":"Merge \"update constraint for python-ironicclient to new release 5.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1515ee60df1b366e5cd401f2e8a326807204c732"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:48:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-10 09:48:50.000000000","tz":0},"subject":"update constraint for oslo.middleware to new release 6.6.0","message":"update constraint for oslo.middleware to new release 6.6.0\n\nmeta: version: 6.6.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia24a6e4a3ea3df961be07903a7c43490bd2ddab0\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic40221d92bb1b2f9b42e9817ca1eeaf0090d8983\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0195b0ed92fc07f5961318f6a44efb848bb21383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0195b0ed92fc07f5961318f6a44efb848bb21383"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
