)]}'
{"id":"openstack%2Frequirements~954744","triplet_id":"openstack%2Frequirements~master~I723a4b7a4ce64f236eebd477f262041a6e505ea8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I723a4b7a4ce64f236eebd477f262041a6e505ea8","subject":"update constraint for oslo.messaging to new release 17.1.0","status":"MERGED","created":"2025-07-11 14:36:52.000000000","updated":"2025-07-17 12:00:26.000000000","submitted":"2025-07-17 11:59:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954744-new-release","meta_rev_id":"c1950f1d0d92f4dd321fb7227bb2377c12bcf324","_number":954744,"virtual_id_number":954744,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-17 11:59:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-17 10:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-17 10:44:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 16:04:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-17 10:44:41.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"eb64d7e7dee5207e42bf047fb5b2d1157c87b99d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-11 14:36:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fbd545b3f3f387c406832856f7d9045371fe167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 16:04:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69ff7247ba6f4c37bc9639ed4107882f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a3a4225fda47489a6a4a0845300da7 : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07b874858f9a4c81b378e2dc2775faa1 : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56007c675fa744108a07cc694192d468 : SUCCESS in 2m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85a8fcb4c8e144c09436def07b9e51ad : SUCCESS in 2m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9696a0774c134d4894b0e2423862a2b1 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e5a01a95aa864112a95fd56e112262ed : SUCCESS in 2m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5fdaf92cf24c45e194afa47dbfae21bc : SUCCESS in 2m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2164bcae6ca44672ae277955d8625d32 : SUCCESS in 6m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/55fd6651f47f42269606e2cb2540120f : SUCCESS in 4m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1b263c7f1428499fa9286f444bb89be2 : SUCCESS in 7m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/00b059ad54ad4131b9a180e639e4a5a9 : SUCCESS in 7m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/457be1a5e90e44408da7fcf7a61bc409 : SUCCESS in 9m 33s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/56a494a2d6684d52aa5d7736e1a1f06b : SUCCESS in 2m 54s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/02fef3a9c64747b9a316421b9b81b6d1 : SUCCESS in 3m 56s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/88c6daa29c3f42e3a5cdbc6f9ffb4eea : SUCCESS in 2m 27s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7c1f0bdd07d74a76a74854d17f658241 : SUCCESS in 6m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/dd6baa942ddc4b9497a811803a1cb2f4 : SUCCESS in 3m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ed4348eecd8145dfb3da9298c74d7a11 : SUCCESS in 3m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/204f23a42c924aea87efac504b108248 : SUCCESS in 8m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/209105656b6d4a889b6f8ac4936a0404 : SUCCESS in 7m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/acc3b925a7bd4f0da697d128fa5d003c : SUCCESS in 9m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/258a8d27d2a044e29b6f164649f3efdf : SUCCESS in 6m 25s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/585609904ec34770b6dcacf8e27d06bc : SUCCESS in 9m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f2fd0c1151af4e70b5c4fc970c6e7673 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/08cf00faab484e9fa93e7e719c78da31 : SUCCESS in 7m 34s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/29d8fc57cfaa48a181a20eed1c8d09d7 : SUCCESS in 3m 19s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d3fe87531d5a4e26861133a422270973 : SUCCESS in 6m 21s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/60a62931db404b2c9795e63e9d4c924b : SUCCESS in 10m 41s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a969c5cc3c6a48d5ac38b45fea3174c1 : SUCCESS in 19m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/78db20318c7843fa861d2781b433fc03 : SUCCESS in 10m 06s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8ce0e952c124452e9d683cc2d87ebd5f : SUCCESS in 14m 15s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5a286652d5ea489396da5881a0aa8f37 : SUCCESS in 7m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b137426c5c9d412b982ffeb4f4c0783c : SUCCESS in 2m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/299c094112ae41c19cef786f49abd014 : SUCCESS in 26m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/62bb9383cfc6492bad6d796c3f8f2ba1 : SUCCESS in 4m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e3a0722318c847aaad1f7bc9741864be : SUCCESS in 3m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a4141554dae461683fd476991fca0b7 : SUCCESS in 7m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/75a7daa9e5f847f091146e33b07fadbe : SUCCESS in 15m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2948dad2393a48c8a37c96fc882e4f77 : SUCCESS in 5m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9068011ae93c43a99a113e89eee2a56a : SUCCESS in 1h 01m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f809f5760aa341d992ac4cbc790a1d44 : SUCCESS in 1h 25m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f8e048ede4075dc274e5d20e5d56a5d54d04f7be","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-17 10:44:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f12b48c844d17c6dab80e70c5fc3f62792c4ffe0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 10:45:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aefe40367be36af75755e8dc271ef408c559dea7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 11:59:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72caed565ca94121934909b7573b6ac6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c02fb9fa2043e3a5651b012415350e : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd4bcba125143d8ba6897cde7b5415a : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c059d77452744cb9b50b45195f38e274 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8cc9f82f4b5941c4bc2c3baf385a198d : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f680add439d45e1bb60f246656fdff3 : SUCCESS in 2m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4b7ffc64c0748139856edc654646d47 : SUCCESS in 3m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8dffa87c9d124ee4880096773c99eba3 : SUCCESS in 14m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/744330589fc74e8e9c9aa04d0476be92 : SUCCESS in 10m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d48d8c4e45554082b46a6e1981ea8835 : SUCCESS in 15m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5708bd34443740b6a2ee707ccb6887c8 : SUCCESS in 23m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3b476812a3f8424a8925fa1f437de399 : SUCCESS in 10m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5f6f1fa5b4ae4b37912fec7249b1eb3f : SUCCESS in 5m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/55e7481800b84bbea67e0d128042a296 : SUCCESS in 5m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2496469dd37e4d78a9a7f89c00bf1020 : SUCCESS in 8m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/686b0bb339524b7b9eef31ce47b99501 : SUCCESS in 17m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3979533ac7844baebd181299fa26a643 : SUCCESS in 7m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1c4f23be2a884743a3170e6aefcf2280 : SUCCESS in 7m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0b615415f2db46a0a2109575918ab7d2 : SUCCESS in 11m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b3cd9522752c4e5884835e459e8400e3 : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/62887263eea040c297a2af4443a8e865 : SUCCESS in 17m 05s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ef652da190d142c6b0342b0f11955815 : SUCCESS in 17m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/38db851b7f3c470f89b66610ab63c1c3 : SUCCESS in 7m 29s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c359ebfebaa14869b11440df7097b002 : SUCCESS in 11m 29s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/28c740e7b858429b951bb19d13ba4545 : SUCCESS in 3m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d72bb9f5ae3d48e2baff4fe1d0153136 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5bd1e783891c463b93c497c4b12650da : SUCCESS in 23m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b4e1915a1c2d4f44bdcaf4ad0683f892 : SUCCESS in 3m 21s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/47b53135b3ff4d6fb1f4442b19d3f3be : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57ff13493ed74331bfeee6a4e679c607 : SUCCESS in 5m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a7af742ec1c74e08a4be75c1eac691d2 : SUCCESS in 17m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/81ed9b5796c74eeb9dfe7aeaf9a33912 : SUCCESS in 5m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffcc0b9dd3464302af5a725128a49c51 : SUCCESS in 47m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc0cbd42ec194b73a1c9a7906a8aa2e6 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b58ebbc92dc827dd3c5c8b36f2afe32aef41fcff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 11:59:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c1950f1d0d92f4dd321fb7227bb2377c12bcf324","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 12:00:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d0238310a1e4a34a3814b244b0fea87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b1a2f57f694f4bb5ec0ddde907a504 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8fb27f8e3953421e302aca723f1d7d3f41e99e5f","revisions":{"8fb27f8e3953421e302aca723f1d7d3f41e99e5f":{"kind":"REWORK","_number":1,"created":"2025-07-11 14:36:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/954744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/954744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/954744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/954744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/954744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/954744/1"}}},"commit":{"parents":[{"commit":"e4b22f2172e01a0f3e8c3e2bdc008dd17d72eeca","subject":"Merge \"update constraint for os-client-config to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4b22f2172e01a0f3e8c3e2bdc008dd17d72eeca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-11 14:26:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-11 14:26:50.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 17.1.0","message":"update constraint for oslo.messaging to new release 17.1.0\n\nmeta: version: 17.1.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Ie0b81d8ee69cf283fda90bcd4581fc960cbd24c8\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: I723a4b7a4ce64f236eebd477f262041a6e505ea8\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8fb27f8e3953421e302aca723f1d7d3f41e99e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8fb27f8e3953421e302aca723f1d7d3f41e99e5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
