)]}'
{"id":"openstack%2Frequirements~954772","triplet_id":"openstack%2Frequirements~stable%2F2024.1~Ic14bfa262eb0a63d9cd2eac719c73b7681169975","project":"openstack/requirements","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-07-11 21:11:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic14bfa262eb0a63d9cd2eac719c73b7681169975","subject":"Drop build-wheel-cache-ubuntu-bionic-arm64","status":"MERGED","created":"2025-07-11 18:04:50.000000000","updated":"2025-07-11 21:11:56.000000000","submitted":"2025-07-11 21:11:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954772","meta_rev_id":"1b7992d93cfc05d1368b57cf075890d0af8ff906","_number":954772,"virtual_id_number":954772,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-11 21:11:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-07-11 18:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-07-11 18:21:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 18:21:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-07-11 19:59:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"410a2228ea12ac7cca8e8cfb914e4450aac2a5f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-11 18:04:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0167601503a3928397a7d41fb14ce538974e97b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-07-11 18:21:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"10d5ed4c745bef68f1e72012b3dce1c6367167d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 19:59:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6b3ec3a017f4ab1a1547e05d7e984cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9331fc1cdd4f7b8087680063924345 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc3e232df2bc4ab3b8fd1972dc52299f : SUCCESS in 2m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a7265f987714d3d84ec57f38f614fb8 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1bbe17417ab40548dc29819e782d7de : SUCCESS in 1m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ab20ba3f5f9445c848748dde4299b93 : SUCCESS in 56m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/817c51bef9b74520a34716829b022702 : SUCCESS in 1h 49m 07s","accounts_in_message":[],"_revision_number":1},{"id":"62ec5abed8e79d32f7b96463deb689d0c672867d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 20:00:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73c15ce0df3b691ab82f760040069eb3914e285d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 21:11:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85b0add2ea2e401fbfdffec482551f99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ba0841b18a54aeaa21878be410671ed : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84700b44943049898ead4264d2a2932b : SUCCESS in 1m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/326f882681bd4a11b63e48dacf19e676 : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b29b4d6fbd3d4b1db6703bb9afd86d9a : SUCCESS in 2m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c966f8ed04944263a868cac056f5fce1 : SUCCESS in 38m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb64dc0579334139b23b0e301c0f0960 : SUCCESS in 1h 10m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1b7992d93cfc05d1368b57cf075890d0af8ff906","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 21:11:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f31015dc1d3385687a466b764a2a94416783cf7d","revisions":{"f31015dc1d3385687a466b764a2a94416783cf7d":{"kind":"REWORK","_number":1,"created":"2025-07-11 18:04:50.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/72/954772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/954772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/954772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/954772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/954772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/954772/1"}}},"commit":{"parents":[{"commit":"052c27e7fc77e99091325fb81c44db32b80fe3b9","subject":"Merge \"update constraint for python-ironicclient to new release 5.5.1\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/052c27e7fc77e99091325fb81c44db32b80fe3b9"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-07-11 17:58:48.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-07-11 18:04:38.000000000","tz":-420},"subject":"Drop build-wheel-cache-ubuntu-bionic-arm64","message":"Drop build-wheel-cache-ubuntu-bionic-arm64\n\nThe OpenDev team is dropping bionic arm64 image builds as part of the\nnodepool in zuul migration. It seems that much of the use of this\nplatform on arm64 was for nonvoting purposes and wasn\u0027t really used for\nanything important. As part of this cleanup we need to remove the wheel\ncache building for the platform which necessitates this cleanup.\n\nChange-Id: Ic14bfa262eb0a63d9cd2eac719c73b7681169975\nSigned-off-by: Clark Boylan \u003cclark.boylan@gmail.com\u003e\n(cherry picked from commit 1f2be26ab5a4d601127817b90c794e61cc40bd12)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f31015dc1d3385687a466b764a2a94416783cf7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f31015dc1d3385687a466b764a2a94416783cf7d"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
