)]}'
{"id":"openstack%2Frequirements~954774","triplet_id":"openstack%2Frequirements~unmaintained%2Fzed~Ic14bfa262eb0a63d9cd2eac719c73b7681169975","project":"openstack/requirements","branch":"unmaintained/zed","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-07-18 17:22:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic14bfa262eb0a63d9cd2eac719c73b7681169975","subject":"Drop build-wheel-cache-ubuntu-bionic-arm64","status":"MERGED","created":"2025-07-11 18:13:16.000000000","updated":"2025-07-18 17:22:11.000000000","submitted":"2025-07-18 17:22:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954774","meta_rev_id":"25ac7b4c104b7475d137b6ede6271078f66bf5e8","_number":954774,"virtual_id_number":954774,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-18 17:22:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-07-11 18:21:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-07-17 06:18:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-07-17 06:18:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 18:21:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-07-11 19:43:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-17 06:18:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"92df77234b1b484fcc59c88241011eef479c6516","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-11 18:13:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83c5779a0321318d37139f42550468569f01cf59","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-07-11 18:21:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6db8a90f1272c8eda00a22ce52ff7343ea7207c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 19:43:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de2ceb00c12942ceab774cfe0dcc7f06\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71d666d8782646ea84bf8c7d55ea9677 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a491f3d9bcf84bc39f86188427d49da1 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/583b09dfff84428c8ba22b933e314035 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b93c3ef3b6e64edfac15209eed668cdb : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/978ec3fb632747279405e90939ac298a : SUCCESS in 4m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/861fc66c276c43088dcbe045a261f9b4 : POST_FAILURE in 7m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0aebee8174e34753b1a432fa9a6fefa7 : SUCCESS in 1h 25m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f0fed6978ec3a86103b2a6b7c4d707edb3bc1a21","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-17 06:18:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b57a73ebdff73812cf7413429551b52d9affa45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 07:35:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccd4e7466f524f2cb8653467d668a024\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a8a7105f406488784fbb731a302019f : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16301766a2f143c58c285c4d9a55b4a1 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fde7d4f2cfa74e13b0e713ee58b37107 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4cd75ed6d8f34d029fd74094a48ad149 : SUCCESS in 2m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1076f4d40da442eba3b849f3a8e652be : SUCCESS in 2m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5389546d4a24a34a93b7a5e16c5b3a1 : SUCCESS in 34m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9f92e2a2ee8469281afde1506a8ead9 : SUCCESS in 1h 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"685b4a30818d628d1f2763b4e33f79848bc6f367","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 07:36:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"278fa4683f1c21486f6a465b2ec110d12cee5b63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 08:38:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db1b74ca4f4940718de4c696c329b8bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e62e3000c64c3a9ca4bc65704a9ec1 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9912bfaba7a2405290c202a7d2d7ee6d : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbf1ee8f0e3a4b69bc9cd960d37cffef : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f1bb093f2e3d48bfb83ba3c6c542477e : SUCCESS in 3m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76f8c40d09be44e9a45555f2a9317946 : FAILURE in 23m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b5f1165fe144acbac72d3c5b3bb73be : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"bd0bfec95fb0d2ac4a29a81905b1cc158b142c5c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-18 14:10:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb37686b145051031b670b63b3a7c840a51ab5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 16:01:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e55f9ec0fa494320b72c76effe55b146\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04bf7ebfcc84450babfca47dd9135504 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f1b67a78e0044169a5a2e9fd5cfc90f : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43abec36f5ef4d83982622be45d9809b : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6770e75eef7548a59cbf16db76b68f3f : SUCCESS in 2m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bea0b40e66c343f3bfca70d376c79c67 : SUCCESS in 2m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2df23866aac64811b97a7fcbab4e8061 : SUCCESS in 1h 06m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d0b775af9f84964a5f384d5cb6669d9 : SUCCESS in 1h 44m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8e937e5a16b12cafaa1473249f5c04edb617171e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 16:02:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d0ade6359618fed89b4a230339eb8e5caa859ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 17:22:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bd95df89402446ebbe64ad3502ab2ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7cc2ed991a240d1807af062e90f1c51 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e04e835c410c4126b96ef8fa214135b8 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f77fc3dfc1184096907aab35666e686f : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6733f188d1864162b9881ebab8acf6de : SUCCESS in 5m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0948dd53483d4be692ad2501780b2578 : SUCCESS in 23m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb4deb2ac1e6494e9e5d3908df02e38c : SUCCESS in 1h 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"25ac7b4c104b7475d137b6ede6271078f66bf5e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 17:22:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff29e00bc479f08cd603192157318604cc6c20ce","revisions":{"ff29e00bc479f08cd603192157318604cc6c20ce":{"kind":"REWORK","_number":1,"created":"2025-07-11 18:13:16.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/74/954774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/954774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/954774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/954774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/954774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/954774/1"}}},"commit":{"parents":[{"commit":"1a3b0b7508e206c1863f5a700c1e359728010a61","subject":"Update packaging constraint to match 2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a3b0b7508e206c1863f5a700c1e359728010a61"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-07-11 17:58:48.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-07-11 18:12:40.000000000","tz":-420},"subject":"Drop build-wheel-cache-ubuntu-bionic-arm64","message":"Drop build-wheel-cache-ubuntu-bionic-arm64\n\nThe OpenDev team is dropping bionic arm64 image builds as part of the\nnodepool in zuul migration. It seems that much of the use of this\nplatform on arm64 was for nonvoting purposes and wasn\u0027t really used for\nanything important. As part of this cleanup we need to remove the wheel\ncache building for the platform which necessitates this cleanup.\n\nChange-Id: Ic14bfa262eb0a63d9cd2eac719c73b7681169975\nSigned-off-by: Clark Boylan \u003cclark.boylan@gmail.com\u003e\n(cherry picked from commit 1f2be26ab5a4d601127817b90c794e61cc40bd12)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff29e00bc479f08cd603192157318604cc6c20ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff29e00bc479f08cd603192157318604cc6c20ce"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
