)]}'
{"id":"openstack%2Frequirements~954850","triplet_id":"openstack%2Frequirements~master~I8aed2c2cac1823b8d640bfaec3651ac5c830924b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-07-16 08:33:48.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I8aed2c2cac1823b8d640bfaec3651ac5c830924b","subject":"Updated from generate-constraints","status":"MERGED","created":"2025-07-12 08:07:36.000000000","updated":"2025-07-18 18:49:27.000000000","submitted":"2025-07-18 18:48:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954850-openstack/requirements/constraints","meta_rev_id":"43be4c4d537832d57335b04679201a1f4f61ebbc","_number":954850,"virtual_id_number":954850,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-18 18:48:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-18 16:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-07-18 16:49:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-12 10:13:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-18 16:49:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"dac65a0c5efd49a2cbd23da3e7bf1223fe4976f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-12 08:07:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a83af83d131abd3e15bfb62b777326f26f0ce0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 10:13:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb93a2e7f8a94c65b70833f3011b5bf7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8a19c7f0b3e4f9c81de208e0e382fe8 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87dcaba0bff4490ca0d99ad2ba785ea1 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd7fd64e304143c0800cefe993adc278 : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/610b6329ea3245d3b703ef1d7b425244 : SUCCESS in 2m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ac3c562bdc34bad8568c04ce2c9176f : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c585a0d14da42688af201a512796d29 : SUCCESS in 1m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/38cb898c82634f42904ad08fb5288754 : SUCCESS in 2m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/528c94ac541d4857ae46a78cffa34218 : FAILURE in 3m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2c275c05160c43b4b5fda7a6bc000982 : FAILURE in 10m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/96b33dd6be5f4c0ead20537837dbf758 : FAILURE in 12m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/756b3278bb834d2aaf5288091949dcc1 : FAILURE in 18m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3fe45d32a4e14decb0d4d00644c92b31 : FAILURE in 28m 20s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c3590d929a1446fbb2eeaeb8769666e5 : SUCCESS in 6m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4dd66ca8ce07491c961b0e577b8af339 : SUCCESS in 7m 49s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/302eb23b0c294a1bad713e97e8fe6d09 : SUCCESS in 6m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/68be3c5ef6bb4ff3a83331d4a99874fa : SUCCESS in 6m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a80e1f67f04445459b5417084b3b90c4 : FAILURE in 4m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6a5196ba783d4a78a3a279a92b3c0ab0 : SUCCESS in 4m 27s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b8a439247ad647b0858f6b332b1cab01 : SUCCESS in 3m 30s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d7d9b6d968f24f428ae3818fcfacd7e0 : SUCCESS in 15m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3fd9ce3dd8e1474da1f78acc7785c29f : SUCCESS in 9m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6fd79fff635249da9ac69957b3de5e55 : FAILURE in 2m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/686918966ec142b48060f207bbd0b22c : SUCCESS in 14m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a4aac2550f2848afacb7d7826f4abe68 : SUCCESS in 3m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c420de0762ab4a8f8b958b91dcb18e93 : SUCCESS in 5m 32s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c0156c80190d4f3d968126b32deec0e2 : SUCCESS in 2m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ee4da35727274d38b5cdeeee3548d13d : SUCCESS in 6m 54s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/745d6e413c1147b0b594763100e1d752 : SUCCESS in 9m 05s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b1d1dac7a5564d319cc9653cd472c9c4 : SUCCESS in 19m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a4cbd7345f26469a9170065c3554ac0e : SUCCESS in 7m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5454ae9fe8204c0c9bf16e2f69c100e8 : SUCCESS in 13m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4efc2232ee104af1a97ae18daac807c5 : SUCCESS in 7m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e979fc817fc447bc99c0420a88131bba : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dcd8309ea7e343cea273338d7b35afff : SUCCESS in 26m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f2fa0d8653864af997da35d984fa487c : FAILURE in 4m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/78c8a722d9ab4ee3ac0ed83b0eaa0bcb : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/783cc890ca9249398b563a717078a576 : SUCCESS in 8m 19s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9c2d94f3db2143de91ac9a77ca935b1b : SUCCESS in 15m 03s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c8643c3d47b74da08347084c1a259feb : SUCCESS in 5m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b535dc21b3d466c9943d85a70c01c30 : SUCCESS in 53m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21f4e299b04243f4928ab4c7ad034ae4 : SUCCESS in 2h 01m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b53d93b2caadd71c3dfde88282891bf2afabb6b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-16 07:01:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0d83e4ddf58963d5bbd294a69f14dc58a094d667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 07:51:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3b9a082aa984614b8eb85cdd9ac7c8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59abe441a754b308e272dce3a643f0e : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff298c9eaeb4cc8a7d2e2a02deb5b7e : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/448d8b3acb3d4db382f286433d52b6f0 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0891e94bff2b4e52ba13bbdcf1d5e3f5 : SUCCESS in 5m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48e67bcf4b334826bfb85f3d6bd54327 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/376f876b239d4fefba8551c40512adb7 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4532f5b25c541328d77b424887a016a : SUCCESS in 3m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/27e295e78c0c413fa2b6d493080358e1 : FAILURE in 6m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ed1b47a1baee4d58a2df7726aad6842c : SUCCESS in 15m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/17673804c44947baaa3a47bbeb9bc726 : SUCCESS in 16m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f01b2d76e28e4669bc3ae1b96cd2ddd8 : SUCCESS in 20m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ab32880209e444e083379b0b540513ac : SUCCESS in 24m 12s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7b8f44fcce9248be9b695d22dbb2792c : SUCCESS in 4m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3ac9a72ecd474a0e8b41f4e3205622dc : SUCCESS in 6m 27s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fa2285238b7d4ad692ec4438cf3381b8 : SUCCESS in 4m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7292bafc68034f0586ce6857c9696a28 : SUCCESS in 9m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/594c8fcfdb0d4f6c9b6dba380dc20ce9 : FAILURE in 5m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0b136d0da1014e979bfd602f2da221bb : SUCCESS in 3m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/11894a2ca1aa4a888bc7cba26b68fb22 : SUCCESS in 8m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ac7f8186387a48ba97e187bbcc9820ef : SUCCESS in 15m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9564ce4599e64c16b8e5f126d70364f2 : SUCCESS in 6m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/92aed1d13cc044f197730068cea9a8c1 : FAILURE in 4m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/93dc95cd45564718be7a05707d452be3 : SUCCESS in 10m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fb6f3c6b2b7a4d4da69328b916c510bc : SUCCESS in 1m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f410588348cc4496b95c758aa9166e11 : SUCCESS in 11m 08s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3904091f8c6641dc81ebd82fedd668b4 : SUCCESS in 4m 24s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ad4647eee66a4238aed29024e758dec7 : SUCCESS in 4m 48s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d3f3491d7c9247cebf8b40deb6f1ec21 : SUCCESS in 12m 11s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3e15e4c76216470497412413c8a67e27 : SUCCESS in 34m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fd887f1620204912ba4c422c7c8c003d : SUCCESS in 7m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/577352124834468698459f40ba631ae0 : SUCCESS in 8m 22s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2f93f123fbf447d9b5f0f9525bb15fe5 : SUCCESS in 3m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/63ef00c862454e05a15b63e9eed9fd66 : SUCCESS in 2m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f0881bf0963439c91bf2bd177283148 : SUCCESS in 39m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cd45689dcc6544e5af2eb046b08e876b : FAILURE in 6m 28s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/97f234ee741d44b9b80f5ede7b2b65bc : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e57fead10c2341628eeac9dbde5bf3ec : SUCCESS in 4m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/40fd99c2eb2a4efea2c5da31948b9610 : SUCCESS in 6m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a153bdf499f74f85ada884f4a237f724 : SUCCESS in 5m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a076c3445b0b42d9b1992fb62613825e : SUCCESS in 37m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/114910ad2aa6414c871abd7a7dc81219 : FAILURE in 48m 52s","accounts_in_message":[],"_revision_number":2},{"id":"958db713571b7f31ee519b055f46f753e95aab4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-16 08:36:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3585fc4e516a27f4085a272f23f1683c228515aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 10:22:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6228e9a5b634809a06a84801832702b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08e44bce36294875a105e3839a6b4d29 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a307ab44c04742c089d0babc7316d5b0 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55ad871c79364c44869e9c266426ec69 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fa0a03c3cd44058828fdfba0435f431 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b9d3d0a07a44864a8f72de3e9f9fd9e : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ca77b447bf54512911424044b5917a2 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/008f7c8823ff4595b32406c5326e9ae6 : SUCCESS in 3m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0315935c7a7c4230b6897c6b5a7085e7 : SUCCESS in 15m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3075de29f00f43c4b0d8d40347864af0 : SUCCESS in 10m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3934bb22154a4d16b2f197243839d6a7 : SUCCESS in 17m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c8625a78c91c444f9fb37111c995111c : SUCCESS in 20m 12s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ebbef6681c0c443c9f2ad550b3f6ec0b : SUCCESS in 23m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c48f0522f0d144be891df19e9dd78668 : SUCCESS in 5m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/dae96c0a69f5494ab35c6631fb7a2c89 : SUCCESS in 5m 08s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f02f3f8ee8834789a41c87ec837857f4 : SUCCESS in 4m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e29e69fc212948faa206396836008e0e : SUCCESS in 11m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8ed0b4536e854fa1aeb26af6093516f5 : SUCCESS in 4m 44s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4d27b6288d7943a8ab1f04f3ed402b1b : SUCCESS in 5m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7bcbcf4b98cc4a209c477c4a3f7b633d : SUCCESS in 7m 00s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d3526c927a9742d380ca89a131c91045 : SUCCESS in 13m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/21d0bc8642944b5a830a3ec604025d6a : SUCCESS in 6m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8b2bc8c9810b4051aa095f675cce8ee6 : SUCCESS in 8m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/be73e660358f45fa99929a1217f72d5f : SUCCESS in 10m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/694e6c73df1240e0871a7c7c47240e04 : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9f9714650c9742dbad9c03d151cae0f2 : SUCCESS in 10m 25s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fe8e9c33f35746778fe58e53c1569d5a : SUCCESS in 4m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6b784de2318f449c9ebcc4a59af37fd8 : SUCCESS in 5m 14s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a03afec4541344ccbcfcdce7682005b0 : SUCCESS in 13m 12s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f0637a9713de4c798bd8589964c12571 : SUCCESS in 17m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/35b063cfa7d94877ba5b13c8dec90bcf : SUCCESS in 8m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1240dce5b3f44425a35d87c185aa1bfd : SUCCESS in 12m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/39380c83a8244095946111205df3f4be : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d3ac56e04fc4a55b97d78ab657abc0c : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db5d0e6e3ad54c24bf24e9285bd930f6 : SUCCESS in 22m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4fb9e9baf9364fdc9f9c954aa9299f57 : SUCCESS in 7m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7c66bd7785124ebf9daddfab3b1747c4 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0adf548477d8427fbfaf62cd3121cd90 : SUCCESS in 5m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4b499d577fa94287b5dd69dab528dff1 : SUCCESS in 16m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5dc7c52af60f47e79eb2c4315f7e6db5 : SUCCESS in 6m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13d64e1f81fa4399bcfb0dc36d99e121 : SUCCESS in 1h 03m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e2d61b3f13a4db588072185eddd9f28 : SUCCESS in 1h 37m 21s","accounts_in_message":[],"_revision_number":3},{"id":"d3ff4d595811b3da48aee1a19e3cda2292268ebb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-18 16:49:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5fbd7deadccfd687810055f6dcd38b6c87ae444f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 16:49:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"157ae4efc1587a1cf67fa35caeb214770b5f8c3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 18:48:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79e42c64b82b44f7851d51023ebf659e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73703becd1174d0fa82cc90bd81fde1a : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5e37d1379f4a41afbb8391f9d060cf : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a2bc20df648428ea4f8cb3c921d4bf3 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d807e89506043fbba21b22bc7a12d7e : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5237f35982a4c26ad45d3d825638fe1 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2cf70a0b25534ac6b290cd45be09a880 : SUCCESS in 3m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3b45170431454612aadb1603b0093334 : SUCCESS in 17m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d98705e296d0495988aecf020caca0d4 : SUCCESS in 9m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4ec79f0d4a2a48e4a0f99ba5d5eb8c99 : SUCCESS in 16m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4ce5ef9e46a24e43bf4196c55294e6bb : SUCCESS in 20m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3d1bb923663547a1b71db7a4e69e533b : SUCCESS in 12m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/cf7fe62d72bb4f3bbe661e1c658dc264 : SUCCESS in 5m 00s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a861335f723e4c8ea7daee75aa264874 : SUCCESS in 5m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a58e1e2e67ff44f4b38eccc84603a7e5 : SUCCESS in 8m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0400f728d5914da89a11a3cc3a8b74b7 : SUCCESS in 15m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65c188798ab245d382515957f3521d95 : SUCCESS in 6m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c4e1f6d6304447f7aa28e02e77f2cb71 : SUCCESS in 9m 03s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ec18a954af97437daddbaf6b00f74475 : SUCCESS in 13m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/798e5f4932334530aece8c9c8d2b7c99 : SUCCESS in 1m 57s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/eacdcbbed3714123961ecff558171511 : SUCCESS in 12m 16s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1ffbe6cf12214cc59ab19a6ef97082a3 : SUCCESS in 21m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a0741a8678594784a52696c2a1f0eb0e : SUCCESS in 9m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e1211cf71b4f417aaa1238bacf8b26d4 : SUCCESS in 13m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4d2b06ca2bb84761adda6e1917958998 : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee23349670ec42cda80b51c1bf702cc8 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce373a95d83742f691d7d0d64020e986 : SUCCESS in 29m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e6ab617cae694dd289df3a42bc22406f : SUCCESS in 6m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a928755d113d4b489d1c524ee29bd14e : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c84907b98f4b44389c374407783ed4c4 : SUCCESS in 3m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d38ebf8ec7ae4f5e888ee72d3a249f2f : SUCCESS in 17m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cf8cdcad5b8c48b596c6a13b02cb57f3 : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8575ca5f7c1143da900a465cab4ce65b : SUCCESS in 37m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fb26ad1e5c24a48ab86f99fd81dba8c : SUCCESS in 1h 29m 31s","accounts_in_message":[],"_revision_number":3},{"id":"a450184c8294cae6d68acff3e832e50ed567dc41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 18:48:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"43be4c4d537832d57335b04679201a1f4f61ebbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 18:49:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81833c1ede3d4f92b538ca1af0c935f5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c5769ab2c544508e406eea562a5fdb : SUCCESS in 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"81137c2fd51940a798183aa5ae101b14329ea5ac","revisions":{"fd949a82415ccd22840145107bd7155230ab13d4":{"kind":"REWORK","_number":1,"created":"2025-07-12 08:07:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/954850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/954850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/954850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/954850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/954850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/954850/1"}}},"commit":{"parents":[{"commit":"1f2be26ab5a4d601127817b90c794e61cc40bd12","subject":"Drop build-wheel-cache-ubuntu-bionic-arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f2be26ab5a4d601127817b90c794e61cc40bd12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-12 08:07:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-12 08:07:34.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I8aed2c2cac1823b8d640bfaec3651ac5c830924b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd949a82415ccd22840145107bd7155230ab13d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd949a82415ccd22840145107bd7155230ab13d4"}]},"branch":"refs/heads/master"},"0d80d69a96b891c58d1291ecc70eb3a23bb84e27":{"kind":"REWORK","_number":2,"created":"2025-07-16 07:01:07.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/50/954850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/954850/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/954850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/954850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/954850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/954850/2"}}},"commit":{"parents":[{"commit":"666756b65fb12a4b3f6666f4fb26cbd5ceccc7fc","subject":"Bump py-pure-client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/666756b65fb12a4b3f6666f4fb26cbd5ceccc7fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-12 08:07:34.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-07-16 07:00:21.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I8aed2c2cac1823b8d640bfaec3651ac5c830924b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d80d69a96b891c58d1291ecc70eb3a23bb84e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d80d69a96b891c58d1291ecc70eb3a23bb84e27"}]},"branch":"refs/heads/master"},"81137c2fd51940a798183aa5ae101b14329ea5ac":{"kind":"REWORK","_number":3,"created":"2025-07-16 08:36:31.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/50/954850/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/954850/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/954850/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/954850/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/954850/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/954850/3"}}},"commit":{"parents":[{"commit":"666756b65fb12a4b3f6666f4fb26cbd5ceccc7fc","subject":"Bump py-pure-client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/666756b65fb12a4b3f6666f4fb26cbd5ceccc7fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-12 08:07:34.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-07-16 08:35:54.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyroute2\n- pyasn1_modules\n- gabbi\n- infoblox-client\n- pysnmp-lextudio\n- pyasn1\n\nChange-Id: I8aed2c2cac1823b8d640bfaec3651ac5c830924b\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81137c2fd51940a798183aa5ae101b14329ea5ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81137c2fd51940a798183aa5ae101b14329ea5ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
