)]}'
{"id":"openstack%2Frequirements~956261","triplet_id":"openstack%2Frequirements~master~I70240f2b5455356c357e87e3ac98713abc917e1d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I70240f2b5455356c357e87e3ac98713abc917e1d","subject":"update constraint for os-apply-config to new release 14.0.1","status":"MERGED","created":"2025-07-31 15:48:30.000000000","updated":"2025-08-03 16:38:27.000000000","submitted":"2025-08-03 16:37:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956261-new-release","meta_rev_id":"ad8cec3c2018bfa3827f6c917db98380fa0bb202","_number":956261,"virtual_id_number":956261,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-03 16:37:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-08-03 14:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-08-03 14:56:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-31 17:31:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-03 14:56:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d40bc48b1a059053dc878a11543dadf81317dcde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-31 15:48:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cacf7f39089be594c3c14be585ab407b581fb42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 17:31:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e2f3baf40334803835245c89ffd26e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88e4e4ee87b347158915c7be7cef49b9 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/371b88bc1a3c4e68a5ec480c47880864 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29367ec5e3824b06aac812bf1caec05b : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aec9ff8479784c69b52bc1d5e3a55f1e : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/543aebda6b054a1e90aa22554980a7db : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df426500750b406ab4290f83c8fdb2e4 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f620fdf6f7054692a1e51b7dd41faa60 : SUCCESS in 3m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8ebfa7c5695742d386a205b5f5992309 : SUCCESS in 16m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ffa4f4760edd4657b61069bdc348fa93 : SUCCESS in 8m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/396d3292940e4fe5bef24c0da4ef51b3 : SUCCESS in 15m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/abda2571323144b187fe390ea5a213a0 : SUCCESS in 16m 01s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6d46069a1382466dbddd6a98ceaab556 : SUCCESS in 30m 49s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fba82407e9da4de588ed4058ef8a7c92 : SUCCESS in 5m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e170a07363bd4b88a6d2cf33defcc64c : SUCCESS in 4m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/164e523feac146c4999c1b7598d3c452 : SUCCESS in 3m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/82c116a915b04b63acd489500707bf7a : SUCCESS in 10m 10s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2fe0dff4c9ae4581b61630fec363d563 : SUCCESS in 5m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9a0776d295cc459a8c799967338bd2ab : SUCCESS in 7m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b9eacf655ed84d579537909ce5fec353 : SUCCESS in 9m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8854db1304b94c598046f4fe316e7b93 : SUCCESS in 15m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0af21a689de44d1ad02085d2cd73590 : SUCCESS in 7m 00s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cfb257e0b5894bff99cd02f3de4afcb9 : SUCCESS in 9m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ad21bfbf7d594f4e920b4d678576fafe : SUCCESS in 12m 16s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2937886654824ff4b34b608a076c3fc9 : SUCCESS in 1m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e543e5905b9e4f5f82fadccae4fa0282 : SUCCESS in 10m 04s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d0d2c5cb2bac4fbd9fbfc889486c4c2c : SUCCESS in 4m 07s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8ca7bd728d884e168b325f8b6ddb58e1 : SUCCESS in 3m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f62f525afc0746a3a38c3a273dfa5c86 : SUCCESS in 10m 15s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/01e2bd0cd6ff471fbfcecdce31f1b4f7 : SUCCESS in 19m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a3aedd88ca944303821b4d2f895a26b1 : SUCCESS in 7m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/907ae0fb3b9e4768aa05802a21f30809 : SUCCESS in 12m 15s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c40d18ad685147ec936864787e6e56a5 : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9efcf75a91514f4aa265270af6480c8a : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bcd64719f5304a5198b87246a628f392 : SUCCESS in 27m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/26416387ebeb415db4374aa2558a6efa : SUCCESS in 3m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/44310b837ee64fe4affb6310a34efd9b : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56c68ebc6dfe41b3b35a9596480bca7a : SUCCESS in 5m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d9ac4405b90c467ba252d661edffb030 : SUCCESS in 16m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1cacd3f89471400b8c8d4ea31e7442e1 : SUCCESS in 5m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4edd9a90c1474127b323cf2f0593c2d6 : SUCCESS in 54m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db035ea9563444e3b8fdcdc5398598cb : SUCCESS in 1h 37m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6c8e57149d09b725f128b4740305d2a8dd1f6a1d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-03 14:56:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1eef74542ef3d4f8112ad7ee94d219e70ffaf02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 14:57:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"606177d0d2ae07a3e2804032774ec4afb811604f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 16:37:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d9d43f77d34b78b9f90fe7a6b1b8e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/532da298226e48269b35eb8642b12881 : SUCCESS in 1m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c9996d3489a40fea6e568eaaa5dc222 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54290bb543304b6f860d8dca70b0cced : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be274acbf17444b480b33e1fbaf17127 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dfa0aad34a2a44acbd163e4d11eebcac : SUCCESS in 1m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0108ca2ad23d4251a1ab53cc36e1cc0f : SUCCESS in 1m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c3955c0917fd4bec8f003e23eac5c680 : SUCCESS in 6m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bee6863082954dc18aa5505ac4fe0252 : SUCCESS in 11m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4547325c6bc9408a813f0c5cb52b2816 : SUCCESS in 22m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/49383c507f3845fe92669042778ff293 : SUCCESS in 7m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b9a822fce14347279c677a5e3918544a : SUCCESS in 12m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c0cc4decc810432087baa3b6071d413f : SUCCESS in 4m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7ef42bd094ed4eaeaa709bca52284d61 : SUCCESS in 2m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ce8402faf4ab4fa697ccde1ea339f7c3 : SUCCESS in 3m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/619d022ca5be4383b50e8a3cb5621b7d : SUCCESS in 15m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d152b391e2f44cd0a13cba30fbc023b9 : SUCCESS in 6m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4275384c43a44c07880082320bde7d23 : SUCCESS in 5m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7317e22b4cad4d548f39f42ed4009beb : SUCCESS in 11m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f0134ab0436643da878f0cec75b1214f : SUCCESS in 1m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8fd0e6fa0f24458c8bfa60d2efbf87b3 : SUCCESS in 5m 43s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e4f59bbb99d642e08ec45f176fce9353 : SUCCESS in 25m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/91d640ceef5c4ded90c472a35d989eff : SUCCESS in 3m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/05d687c24a61400fae88d98c30a4abff : SUCCESS in 7m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/19a8f114b4344d0ca4a2400aeba0b5b9 : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/797274522d1e440dbcb6b5bdafcd86e9 : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9bb54733f85e40deb8a4fdf0f18621c0 : SUCCESS in 24m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3d37fae51b2b4c85b4e47735c9f09977 : SUCCESS in 6m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0e5b90deba074f37b616357c43de3e64 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae36b4bdab9a49628b017dedd95f6279 : SUCCESS in 5m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/900ac93bd3914c88b35b4224780568cd : SUCCESS in 17m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/952b56ccf1e64d44a67b979a026186d4 : SUCCESS in 5m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e81b19804e0a4c1eabf8cd00b139700c : SUCCESS in 56m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0e8fbafab0d449fb029268f909bdba0 : SUCCESS in 1h 37m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f9d3a6d8045353bfe3974f410891023ad7bacf66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 16:37:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad8cec3c2018bfa3827f6c917db98380fa0bb202","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 16:38:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efe951db890342a7825d524f8d6797b1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44424a4789b4a57bd3329001e99964b : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15b196069c7671b04bbd5ef0c0aeae1320111a9e","revisions":{"15b196069c7671b04bbd5ef0c0aeae1320111a9e":{"kind":"REWORK","_number":1,"created":"2025-07-31 15:48:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/956261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/956261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/956261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/956261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/956261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/956261/1"}}},"commit":{"parents":[{"commit":"2b2b12da983a9f3a0ed6559f5d3482e9f85b9a98","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b2b12da983a9f3a0ed6559f5d3482e9f85b9a98"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-31 15:38:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-31 15:38:27.000000000","tz":0},"subject":"update constraint for os-apply-config to new release 14.0.1","message":"update constraint for os-apply-config to new release 14.0.1\n\nmeta: version: 14.0.1\nmeta: team: heat\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I6438c3399b6d216feff7c8f21c1d58b82314ecac\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I70240f2b5455356c357e87e3ac98713abc917e1d\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15b196069c7671b04bbd5ef0c0aeae1320111a9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15b196069c7671b04bbd5ef0c0aeae1320111a9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
