)]}'
{"id":"openstack%2Frequirements~956265","triplet_id":"openstack%2Frequirements~master~Ib10b08dfdb2fd93661740207ed3716b3689409cf","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib10b08dfdb2fd93661740207ed3716b3689409cf","subject":"update constraint for os-collect-config to new release 14.0.1","status":"MERGED","created":"2025-07-31 16:06:03.000000000","updated":"2025-08-03 16:44:39.000000000","submitted":"2025-08-03 16:43:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956265-new-release","meta_rev_id":"ad68dcb3b7ce6b8ed5c29b9d4504c7cdce875da9","_number":956265,"virtual_id_number":956265,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-03 16:43:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-08-03 14:57:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-08-03 14:57:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-31 18:20:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-03 14:57:04.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"329c13261afbadcddb02fdf4ff27dc60be66609b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-07-31 16:06:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fdbc4aed08d25a482cdbb2f6a72367e1c3904c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 18:20:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60e43e633afa459f93bf73a5fe4ba8b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/313d94e35cc845849ae2fb8892abbe72 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdcc1c9c2b104ce39195a130c8236f9e : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4bbc7834ede3468ea9dc32bc70ae00ee : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8daf8d937f74d7e86001041c99c503e : SUCCESS in 2m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5c1b6065339450698acebbcfeff245f : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4493402948dd4f12adf6cf5c2bcdf67b : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c08c9eddd4243e9a17f1cf9ccd65f1f : SUCCESS in 1m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4c9857749f7a45d2ad32dc9c47eb919c : SUCCESS in 16m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6e88010bd5a24daca1d56533cfa7c14a : SUCCESS in 4m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/45fbb2ba408443ca8db401a8a1065f50 : SUCCESS in 15m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7df637de2c50473986c77f5d73fff806 : SUCCESS in 7m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4e212a7c8f024f95bd8467233e982d0a : SUCCESS in 27m 26s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7b146d807cb04ed794464b2eb8471ac7 : SUCCESS in 2m 43s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d4cd5680f725421dbf824936fa387f99 : SUCCESS in 4m 38s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c57b924b3e7a46dcb2b406f5824ee3ae : SUCCESS in 6m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9b6afd40900349e7b74dba3e1b320e4a : SUCCESS in 9m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c61ff8f32cd94e6491bc805f3d5bdd12 : SUCCESS in 3m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a54746cf51b949b5b975728c5b4a4057 : SUCCESS in 7m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/89fb4dcf3cf84a468b2c802cb905e516 : SUCCESS in 6m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2a8a57e8969e4dac9aa8a3553bf1e9cf : SUCCESS in 13m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d15fd51c9c44c5a81b6aa52708c1639 : SUCCESS in 7m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/531a5b22a5624f40a5e2967ef7e3600f : SUCCESS in 8m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/395dd60c352c4c63adf970568a27ca2f : SUCCESS in 12m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/abdf3161b8c34c4680d4e43cee80a71b : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/80fc06093f5245efb9dc918147fd7e31 : SUCCESS in 10m 24s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7f1799a65c2a4edeac3c748cc3071a00 : SUCCESS in 4m 39s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f0bb47cfa480495a837e8ecbb6cff730 : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e8408916dd6946489289a682b2e9aa28 : SUCCESS in 11m 53s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/db92d2830bf74c5f94f49959b8050096 : SUCCESS in 21m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/79846670dfd34f7ea8ad1d3048cbce85 : SUCCESS in 9m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1555cbb1f54e4ebfb3f59d360c991ab6 : SUCCESS in 12m 25s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/10da19cab8584e0c953be1955a39b883 : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b1e9cad40df745f3b32790f46b77902a : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e4090eb38044185af333d4194de7430 : SUCCESS in 25m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5553ca593af340809d73fe293a09616d : SUCCESS in 3m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e84d261a26d94933b190417c506e7121 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/085d51c9e72c4576b1c9a93f9ba5c23e : SUCCESS in 6m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ed70bd8d932d428dbd35ccd30337592a : SUCCESS in 12m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/99d5e4490a014a819f0a7ebcdf2d27fe : SUCCESS in 5m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a9a63fcdd6b4366a0ffaabcd633e8ea : SUCCESS in 56m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcf78043cf7d4692a86b4fd761d19ba5 : SUCCESS in 2h 06m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b475362bbdfea242c2f062b3e6e66dd780f1e21b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-03 14:57:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71511aa68e6e87e058a8f679f3e99593875921f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 14:58:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d34a85b61b4a4b17e07bc676ce5d9fa6b94569cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 16:43:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06d9651549914a209ff8cae7fa71f047\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6dc318d10e047798f6e907317c0dec8 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31dff9c97f54413b9374eddd06b33bd2 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68c7c1acc68a47a98592df454a5d24b8 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d88e682cbbbc486ea4fb14d6cc0b9c36 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/caf4735e46344d1680adb7e3280a6ff8 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f6d0a57d0044cba8d63c3d232c5e6cd : SUCCESS in 3m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1cc2925b79494c8a815084240fdea999 : SUCCESS in 14m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ff463d9efd444080853fb74ca9f8eabc : SUCCESS in 12m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/117aaa3d3cff4a21a8b7b3c67e655365 : SUCCESS in 19m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7630ce6593954897a913c838484483a6 : SUCCESS in 21m 34s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/86e62ad18410429fb970abf8e83afba4 : SUCCESS in 9m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/edce749311b04a07b706f8c4d6b8d644 : SUCCESS in 7m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d11ada3e0ba04afeb204e2bc4b028a2c : SUCCESS in 6m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/218c9f10b8ae4ec59eecd4129196831d : SUCCESS in 9m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f10a1cfbf5394cd88caa00375f992bbb : SUCCESS in 14m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0253227763404e89844dfa0cead18f82 : SUCCESS in 6m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7ba867fbe12e42fea16f7272d19eee9c : SUCCESS in 5m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cc084465be464cf3b6d855003c14520d : SUCCESS in 10m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a6288c3f431e40e6b35b8bbf2e6ebb3e : SUCCESS in 1m 58s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6359276c23984f51915c983c3e9440df : SUCCESS in 10m 20s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a801c72c057d4d689546f6d61fe35e6a : SUCCESS in 20m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/32dfa4743fba4fdbbb31429ea39b04f5 : SUCCESS in 7m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0a438aea2e554842850c1d024c5b40b3 : SUCCESS in 12m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b28ad229432049c2882acf1e974ff67f : SUCCESS in 7m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/620e7e28495a40ff8f99c6111071baef : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6cbda0d6a22463cb50aebe32505ce2c : SUCCESS in 22m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cdb685c54aec4593bc0e16e99db4bb2e : SUCCESS in 6m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/579e339123b34803816ebe881a5f2920 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a70af70ab92145cd843d25bd2efee58c : SUCCESS in 7m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/57c5df1505e74568904f36758c66d88e : SUCCESS in 17m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4f3a458dcad24f9c9e8caa14b5f4cf04 : SUCCESS in 6m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b3f430ebc554ffe9f26f545e14e1315 : SUCCESS in 58m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ab7e30cb60f436fba7d9d3038d4d2a8 : SUCCESS in 1h 41m 10s","accounts_in_message":[],"_revision_number":1},{"id":"cf149335d0ffcc51eddbdce03413c69b8625b46a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 16:43:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad68dcb3b7ce6b8ed5c29b9d4504c7cdce875da9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-03 16:44:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7539a5cd5b44a52873b53ca6e002726\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00d40a8957849eebffd4c5091750d37 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85c219e0b981e8a0abb8a0d25bc97779f3edf2ff","revisions":{"85c219e0b981e8a0abb8a0d25bc97779f3edf2ff":{"kind":"REWORK","_number":1,"created":"2025-07-31 16:06:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/956265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/956265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/956265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/956265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/956265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/956265/1"}}},"commit":{"parents":[{"commit":"2b2b12da983a9f3a0ed6559f5d3482e9f85b9a98","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b2b12da983a9f3a0ed6559f5d3482e9f85b9a98"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-31 15:56:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-07-31 15:56:01.000000000","tz":0},"subject":"update constraint for os-collect-config to new release 14.0.1","message":"update constraint for os-collect-config to new release 14.0.1\n\nmeta: version: 14.0.1\nmeta: team: heat\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I6438c3399b6d216feff7c8f21c1d58b82314ecac\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib10b08dfdb2fd93661740207ed3716b3689409cf\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85c219e0b981e8a0abb8a0d25bc97779f3edf2ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85c219e0b981e8a0abb8a0d25bc97779f3edf2ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
