)]}'
{"id":"openstack%2Frequirements~956791","triplet_id":"openstack%2Frequirements~master~I0d36e9751d0b46b336a50a5b43b60b1922e89f5c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0d36e9751d0b46b336a50a5b43b60b1922e89f5c","subject":"update constraint for oslo.db to new release 17.4.0","status":"MERGED","created":"2025-08-07 14:16:49.000000000","updated":"2025-08-08 11:10:13.000000000","submitted":"2025-08-08 11:09:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956791-new-release","meta_rev_id":"b0ea9746e870709e8d6fa2fa277b122e170313d1","_number":956791,"virtual_id_number":956791,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-08 11:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-08-08 09:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-08-07 14:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-08-08 09:15:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-07 14:17:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-08-07 15:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-08 09:15:16.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"2252bb94d4fefe5bbaa3a1540bdc11b8c17e7f72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-07 14:16:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15ac94f49649d6f6e0e862d6f16fec7b3339e6fc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-08-07 14:17:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e22ee4905aeb6e1ecc50e709989a565a33f11449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 15:23:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78ce1471ae0c4ade9dad5be58fd7f5d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143bf44b42bd48d6bd1650ecdb5af393 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ac24dfc24f74f9e9aa52942da47b5cc : SUCCESS in 1m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/742cc3b9e14d4d38a43d9cf512aaaafb : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/866363df435245c597a5b917460a8768 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a44ec44b0e8c4d448347b629ba5a26d5 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1139f3734f54a69b735addd00a8cd3c : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a15531e252b4f43bc57ff690d5df723 : SUCCESS in 3m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/91937d96391343da8fde6d3dab8ff727 : SUCCESS in 16m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/74f48ce12a1a4df19ec6e6eb6d5236a3 : SUCCESS in 13m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1c0df77dc99f4070b677a0ca88c6a570 : SUCCESS in 17m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c3b01a2e84684985aab49987adeb0fa9 : SUCCESS in 7m 59s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/042b1cdd1a674bc19418c7ac8174e57f : SUCCESS in 26m 43s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a73a71d9508849dc94cc3763347d2df7 : SUCCESS in 2m 37s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c850a017b431475c83679fd7708048e8 : SUCCESS in 7m 13s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ef3d5620a06d492da089a691dd30fe1b : SUCCESS in 2m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/47d7322da6744c1ca00ea7cef30de389 : SUCCESS in 11m 46s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5b64b41bde3842b7b346273bf70fdd8b : SUCCESS in 4m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9d07523c80a64e329f8155bbd056a75d : SUCCESS in 2m 51s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9f2a5a198e30464f9fc05c4c05ee82ba : SUCCESS in 8m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d313b8c47e1d469b9f1609476286d401 : SUCCESS in 17m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22f6d77604f54ad7b4d422a4b127eeba : SUCCESS in 6m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/562b22c7f9d54f6eb9af102862c27e2f : SUCCESS in 7m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7d10a6d04ce84b2a9605fbcf6ee7f6b9 : SUCCESS in 7m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ca23888ac24f4c64ad8b132fcf8179b0 : SUCCESS in 3m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9f6199761b854ea18fca7478eea89467 : SUCCESS in 10m 26s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/757ed66d1ed745719d69568385db48db : SUCCESS in 6m 09s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/865672b3030a4e56885127d725b4cd95 : SUCCESS in 5m 12s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/66181c1defa84510995034d952a3b7a6 : SUCCESS in 12m 31s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1a8520d950a04b2bad651093629536c3 : SUCCESS in 18m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/670d0f468e35450f82f0cecdebe1fd28 : SUCCESS in 7m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6dbb7894437049dea04769dfda71cf3c : SUCCESS in 13m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7544257797784ee08fb705314d37bc0a : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4083f5e7aad140f6a3e9b1b712c91765 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37f0fe28fa5d43b8ba7923cffdc915c1 : SUCCESS in 25m 52s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/13cf36d3638841dda75db913646d1032 : SUCCESS in 7m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9055fe5de4dd45a59eefe5b6ed7138a3 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e5d01e79c69d4d73840d8d80a6f87b28 : SUCCESS in 6m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f4f3f6b1c89649adb2e15f8f3c651756 : SUCCESS in 18m 55s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ec2b2613d0e143f0a88aedbb3ca123a5 : SUCCESS in 6m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/530a08c2b2ca46eeaf8aac61978533d6 : SUCCESS in 1h 00m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8beb447da00d4637aad16090dd5140c4 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8afc6bd11b891df982881b6fd9a42c3356b7e5b0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-08 09:15:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"160f3db4fa88d6a2ab231cde2a8ffbded42082c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 09:15:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d221817c9fb87541643d2c615c700967519d724","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 11:09:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/358099335c62456bbb6a050bbe850a01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53653bc32c7d4d8f8c4f7bc10ffe6178 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dffb74c2fd84f25b9fdcb41b5a3e7cb : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f70fcd0ca65e42e2a626a02758ef0c6e : SUCCESS in 2m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2a505e7fefa4732bb73c133ae2504e0 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c38a97dca6246d6a713ddbd4fbdaa81 : SUCCESS in 1m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e919f9b27c2466b801e1224f1aafccc : SUCCESS in 3m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/968ec6f6c0b24b859a2b1f348e6e446d : SUCCESS in 8m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ce49b58aa2e24bada443deb5624ec05e : SUCCESS in 5m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7fb089ede907452e9f621d64cafcf81b : SUCCESS in 8m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/77a1ae881ce54bc3a8fb5c8f3f28ec6c : SUCCESS in 7m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2c4d82d6de0a44a3bdeda0a7b5398329 : SUCCESS in 11m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ad279d5fa27348579a0ddee257f63731 : SUCCESS in 7m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/22c1da37066740f3a241b8b8dc08aa66 : SUCCESS in 6m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/892d2881e212494aafd1fd422391f72e : SUCCESS in 7m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/36c68d79c2eb4183b5c5af408f2f9ca0 : SUCCESS in 18m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e675e22a31fd4701b0cbb25e474be5f3 : SUCCESS in 6m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/afdd86b1653b4ec1a190315f085993f4 : SUCCESS in 7m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/04f78b597a104c978d588016bda13501 : SUCCESS in 10m 38s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/685365dcdba744b98e63fdf2d5e46f8c : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/97a8f8878e0a49c6ae84803d59f5d94e : SUCCESS in 10m 46s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/47ec1ba3a1434df8bcd687f0dfb8d5b6 : SUCCESS in 17m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c40b12e189cb493a89143bdb21369e1d : SUCCESS in 9m 35s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b196fa87689c48fe9efe299827c8190c : SUCCESS in 13m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/00d9f3989f9c4603aa9c6f665549b663 : SUCCESS in 7m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4cfe6b400a8d4e2d80de774b31c426d1 : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/beae396effc44c958f3ebbbf6f0e84b5 : SUCCESS in 24m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/909ff94a31b8482da7118f922a49796a : SUCCESS in 6m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ae7de12b810a4edb86a2fc41039cbccf : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/201ad03872634fb49a44f5b9167413bc : SUCCESS in 5m 52s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4f6adfd8b61946d390cb0bd3a5c48294 : SUCCESS in 17m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5c1ca077422e45c48882a35e2fd282a4 : SUCCESS in 5m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19080270a3ee499cb6bc602b74d39859 : SUCCESS in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa08c263e6bd441f9b79ab66e65593c7 : SUCCESS in 1h 51m 22s","accounts_in_message":[],"_revision_number":1},{"id":"af5ac7cb355b94bcf83182fe89faf9d97cb689c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 11:09:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b0ea9746e870709e8d6fa2fa277b122e170313d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 11:10:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b184085f3181467a877d8755a5f7bf1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6fc3842d2334bfe8f1bd61e33b10cd2 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6ce8972d24ed20c5928c9bbd3bb976a480c8a46","revisions":{"c6ce8972d24ed20c5928c9bbd3bb976a480c8a46":{"kind":"REWORK","_number":1,"created":"2025-08-07 14:16:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/956791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/956791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/956791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/956791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/956791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/956791/1"}}},"commit":{"parents":[{"commit":"6324b28b1bfe61eace77bde3978daec14ed3d176","subject":"Merge \"update constraint for futurist to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6324b28b1bfe61eace77bde3978daec14ed3d176"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-07 14:06:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-07 14:06:47.000000000","tz":0},"subject":"update constraint for oslo.db to new release 17.4.0","message":"update constraint for oslo.db to new release 17.4.0\n\nmeta: version: 17.4.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I7ebcdaa0bc28565d01737c339532dace9a122523\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I0d36e9751d0b46b336a50a5b43b60b1922e89f5c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6ce8972d24ed20c5928c9bbd3bb976a480c8a46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6ce8972d24ed20c5928c9bbd3bb976a480c8a46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
