)]}'
{"id":"openstack%2Frequirements~956868","triplet_id":"openstack%2Frequirements~master~Iefa38873c333677696945dadd65cce3078cab851","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iefa38873c333677696945dadd65cce3078cab851","subject":"update constraint for python-observabilityclient to new release 1.1.0","status":"MERGED","created":"2025-08-08 09:08:22.000000000","updated":"2025-08-13 23:07:31.000000000","submitted":"2025-08-13 23:06:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956868-new-release","meta_rev_id":"e26417bc310dba2bb1142b3969bb02e23389c614","_number":956868,"virtual_id_number":956868,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-13 23:06:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-08-08 09:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-08-08 09:14:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-08 09:14:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-08-08 10:44:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-11 21:06:41.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"CC"}],"messages":[{"id":"34bb13036873145cac616eb50d9a65bda554d753","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-08 09:08:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62378a99cca9694ecb3fa5f46b14a3d576f6c8a5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-08 09:14:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f94aac48f9e0cb83527df3a7416c9281676f4c39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 10:44:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b627a202b4d34e07afb42910c748e0e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0b5de3a6bcb46c1b11c84c59de97e4b : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/098b818a40da4f6da07aba64580ae087 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff8b74ae9186473b9455aa8905093f45 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d4f11e4ac9f45dfadfee5627b8c89c7 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab5d550252fe422c824eadd77ce9ff32 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2866adb5f4cc47929fa13b3bfb81775a : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e22e16fd7b04d4092fa0dad325a9ec6 : SUCCESS in 3m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8ff49e2d084e4ca6834b03538e518d4f : FAILURE in 3m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/72a1e0feeb0a4653ba18e5aafbd6e1f0 : SUCCESS in 7m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/46a03fa87cd44abfa542c0ba4522f99e : SUCCESS in 16m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8584e501ac2c47db9e91846aa34d25ed : SUCCESS in 21m 26s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/81c85a6135a24972a7b687fa9f412ecb : SUCCESS in 24m 07s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2a698d0f7d0d4d51bff036196b5e7077 : SUCCESS in 2m 31s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/790b070e224a41219e80fa42fa178174 : SUCCESS in 4m 49s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/53f52e5aeecf42c68c34863c0ae48e5e : SUCCESS in 6m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d9d13ce8026c4617b48820005222e22e : SUCCESS in 9m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/68a673b6d6a14c4e8583f6295616672e : SUCCESS in 4m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0e69ef089fc549619593a7f5c05dc60d : SUCCESS in 5m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/388b896a94f5496fb0d73aa7414fb424 : SUCCESS in 7m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4e5ca9fcac864aa193727303987cc914 : SUCCESS in 7m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa332efa95d84f64b366f94f444c696c : SUCCESS in 4m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/754a4e535fef42a9b6fdfef9f7f2b125 : SUCCESS in 7m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/292dc81c31794776bb14455b06336458 : SUCCESS in 10m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7236a96944e74817a33d0ecd58152ff3 : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7645baa5175d44d786041dccd750aa59 : SUCCESS in 8m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f81819cc22854af8928e7a98e7d8c293 : SUCCESS in 4m 40s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/dd8cbc6a1c6c462890dee7aca3597159 : SUCCESS in 5m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a9794d6a903648bda8a9056d9465fb8c : SUCCESS in 10m 29s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c221392aa4764320a6511cc1ed57cc3d : SUCCESS in 25m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/236bce29ac55484892c4f60ef7a70dde : SUCCESS in 3m 53s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5f913a70f3834f70bf37d5a39bc4ae6a : SUCCESS in 7m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d091a29becec432ba573f5dfc5f231c2 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eb0a742630274035bf03f3df9b48cfeb : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/68eb6b0f54c54414a03567f519ac237d : SUCCESS in 21m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fa0ebf4da09b453fb2167053cc45f195 : SUCCESS in 3m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/726498b252274da986294c5e7fe3e8bf : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/068435fdd4be4cba95006ad131f29b3f : SUCCESS in 5m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/abfaa188c71e40cc938d3c5268dba10c : SUCCESS in 11m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e65af12aa45e4cdfb9653b787065f90e : SUCCESS in 3m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d22ae233d65f4c54953fa0ccc7e70a34 : SUCCESS in 31m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8115f7a47e0b4b278f4304d52f84c3ae : SUCCESS in 1h 32m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4b18c0086492c2ccfc817a83fc66ce853cd13989","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-08-11 21:06:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efcab2052e8023795cc274020322b882d2208a42","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-13 13:34:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0943f870c99076613e71bcb6c27bb19f19982e3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 19:36:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb76cda1cc1b4ad78ae60ae030493f60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1678bd63358b4d78a759cf5af2911bf2 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e1e7aad4b24381a0b0ac9040af3e83 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f5856660d6b4fae9e3dc76e0ec42ce3 : SUCCESS in 2m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da4097813e394e74887d6a1ccd618041 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3babe80d6c2445cb15c6c6b4c45959c : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b782591479849ff842f1fb3c6fb0126 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4dc15ae5ca6e4ac7bfe2ddace653bb30 : SUCCESS in 1m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6a9756f7c1144485b9054b9c2e095873 : SUCCESS in 4m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4fcd914c4fd3411f81ae38c348a473c8 : SUCCESS in 10m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8008683d1e7f4d9ba180dcad3fd905df : SUCCESS in 19m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6a679113667641959c2d7b28551ea713 : SUCCESS in 28m 14s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/bd30b831fc6743a092e788816dca466f : SUCCESS in 3m 07s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2ba4cf074865415eb0adb5213f30b133 : SUCCESS in 2m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dcda0899e53f494f8a715d0fbbf74cbf : SUCCESS in 2m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3e32b46ab9df46f48bb33074b4d3aba0 : SUCCESS in 12m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0e00f0b8f9e6463aa118cda99cd7d93a : SUCCESS in 4m 22s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ea09bfd0edd948aebf71f2fc80f6b18b : SUCCESS in 6m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2f8df6ba984942d1bc7eabf3c24d6269 : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6da6648a13234e8baa1d1ea2a8063c5a : SUCCESS in 7m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d90f8dbd520e46948ec522021d0448cb : SUCCESS in 5m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d06018927c174dde8bfc0c2295cd50f7 : SUCCESS in 5m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a4a1ff2fdd30401c99abcfaa6e6d48f5 : SUCCESS in 12m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/eca3155410b84f1a9ddf14635a5d9a6c : SUCCESS in 2m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ea7b9c3d0ea24c6d9275d4d7ac7ce11c : SUCCESS in 7m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/55104acf45ef4d8d88b49e5e991668ac : SUCCESS in 2m 21s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6738ffe5f3e3457e99a336da2a296555 : SUCCESS in 4m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4187a2ed039141aba15bd2ed6a3e0f58 : SUCCESS in 12m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/22e0d5faa36343079142353ce6bd2111 : SUCCESS in 13m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/35d8844e681e4473a4d1ca3e7936ecc1 : SUCCESS in 3m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/091e333abda744a4b6ab2c89f9ebc259 : SUCCESS in 7m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/43dd41e14edf43188e4126fea1a0e737 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1ef5a99ac9943b49e31268a8eb8f7a0 : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6dbdc619c7ec41499a4ee2940f33c206 : SUCCESS in 24m 51s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f80a5a221eef478f8956251b6bb75f9d : SUCCESS in 3m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8f8121ab02e941fdb4d3ccd567a585b6 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76ed659f17ae4c3e83685151cf6ee68b : SUCCESS in 6m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0c31404fb40e4a5e99b88817878e9f8c : SUCCESS in 6m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/721dbf8697274ad589f47128d85820fd : SUCCESS in 3m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3df41cf1dfc7486a8d882b2c1349611c : SUCCESS in 1h 04m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afa5c904df1443cba1aa155c95cfbdcf : SUCCESS in 1h 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2ece270fc949c80366f81c9f52582e7fcadf1514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 19:36:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a22f2ab0c2249edf81bd9444097340e568aba506","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 23:06:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df1fdc0431b3468a9621d821be8591eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840d89f688ae454c8f279fc18db83d19 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/924bfada8c8b498584a2614206689f1f : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27367561966445ca8f213ec08b7ed846 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9998311a4d5b4311b73e96b594d2c5b8 : SUCCESS in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4ee171c238c4de0a791839293bf309f : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c28147c48da4d01a9944b410e9c900a : SUCCESS in 4m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b13fc6c40fce4ae3a9ef0a796145baa2 : SUCCESS in 9m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1a6deaaa40cf45ae9c132f9493bb3cf0 : SUCCESS in 9m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/155f5176a0e042e59946235237ff9443 : SUCCESS in 23m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ed1b03adaf76435389aaa66f667fa6d4 : SUCCESS in 12m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7c08482db5024d06943f458a353a4d55 : SUCCESS in 5m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/56a48c41ef75484aaee68771872ef21e : SUCCESS in 7m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/112d732861a4432eaea76308338166e4 : SUCCESS in 8m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/753add305a0f4327a4e060c95e90093c : SUCCESS in 17m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c978352999054a4db35589101cacef42 : SUCCESS in 4m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fb6f9d2895744f45bdda964b13ba5581 : SUCCESS in 5m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d65c0e7ba58244f694f11686c34d544d : SUCCESS in 12m 18s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4bda6fa432da4903b28799d5c619ee75 : SUCCESS in 2m 42s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1a29252ea28244b18a33f475aee6fd8b : SUCCESS in 13m 45s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/192a8526c1204eebbeca3f468f18b4ff : SUCCESS in 14m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8fe5fe1f51584bfb9972a18b046e8753 : SUCCESS in 3m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a645ab1e6eb6403fa79fd6f157dc852e : SUCCESS in 13m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3714f3f9385a4beda0300cf1df1158c0 : SUCCESS in 6m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6d9984bc8a5842528e267492fa23005e : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/defa5352f02c4b50b50eb1316e1860d5 : SUCCESS in 25m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3dd2d0a8bce5474288b7f19cc70491a7 : SUCCESS in 6m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/118cb2fa1efa45f8ad1916389e512055 : SUCCESS in 2m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/44f21fd8ad994b0899531eec44e421c2 : SUCCESS in 6m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6f3aebce75404fad9b0453cc81a8193c : SUCCESS in 16m 06s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d80e1a1ba445478ab5fb2b2e1141c75b : SUCCESS in 5m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/437f2ff3e1ca4da79f467ff820d55b0a : SUCCESS in 53m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/631883813dbd4491adfcba2996d5c112 : SUCCESS in 1h 32m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e52a566838204b1a01164a494533eecbc25f3fce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 23:06:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e26417bc310dba2bb1142b3969bb02e23389c614","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 23:07:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6395b8cd5f748139cd387bbde0c0159\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f08193f69b884f50b0a121f368826ec8 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a07c610da17cb88efdab75b1c5a55753b2d6b606","revisions":{"a07c610da17cb88efdab75b1c5a55753b2d6b606":{"kind":"REWORK","_number":1,"created":"2025-08-08 09:08:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/68/956868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/956868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/956868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/956868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/956868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/956868/1"}}},"commit":{"parents":[{"commit":"6324b28b1bfe61eace77bde3978daec14ed3d176","subject":"Merge \"update constraint for futurist to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6324b28b1bfe61eace77bde3978daec14ed3d176"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-08 08:58:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-08 08:58:20.000000000","tz":0},"subject":"update constraint for python-observabilityclient to new release 1.1.0","message":"update constraint for python-observabilityclient to new release 1.1.0\n\nmeta: version: 1.1.0\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\nmeta: release:Commit: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\nmeta: release:Change-Id: Id105dfc0a73711028f158da26911e489e0164d9b\nmeta: release:Code-Review+1: Juan Larriba \u003cjlarriba@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Iefa38873c333677696945dadd65cce3078cab851\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a07c610da17cb88efdab75b1c5a55753b2d6b606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a07c610da17cb88efdab75b1c5a55753b2d6b606"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
