)]}'
{"id":"openstack%2Frequirements~957587","triplet_id":"openstack%2Frequirements~master~I35412c12870fc6292636a0308dce55a5951e3737","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-10-27 05:06:30.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-10-27 06:30:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35412c12870fc6292636a0308dce55a5951e3737","subject":"Updated from generate-constraints","status":"MERGED","created":"2025-08-16 08:37:20.000000000","updated":"2025-10-27 06:31:00.000000000","submitted":"2025-10-27 06:30:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"957587-openstack/requirements/constraints","meta_rev_id":"0d21a30d4d3344150377c9089a709f7408a1cd58","_number":957587,"virtual_id_number":957587,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 06:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2025-10-27 05:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2025-10-27 05:06:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-16 09:17:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-08 11:43:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-10-22 18:08:08.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2025-10-23 09:54:00.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"77e39b1fdce60d2bd4ccc62b48550e51be142790","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-16 08:37:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a176c15e96e9965813b64ea10ca6973c928f644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-16 09:17:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c55738a690794b3e8cc513919d4c2335\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/419aafbd71b646018f52ab49a1530184 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67dd077632e34d79abb42153f881a83d : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8326f20371b448609e0f3ba6941b56c7 : SUCCESS in 2m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03c9fedf4ff74906b85b37ffc3c3a4bd : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45e939e58ebe470c9424e069f631ddb0 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8770a7b127414f609a4b919df4a5a253 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b8d8e87a8dd4a0bbce2b8dbe54a57e0 : SUCCESS in 2m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fd3c2220e29e478baf80cd52afe93f0f : SUCCESS in 10m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1280be2587e44beaa96f165e63737aef : SUCCESS in 8m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/22736be53a1244b49b9ae7fb55f46888 : SUCCESS in 9m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8106ce599cef4e29bd61a1706ec4c3eb : SUCCESS in 11m 15s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fb5a337ca15448bd959b1ca68f9a76f1 : SUCCESS in 3m 13s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/efb0145bad914dc3b285b6aa4ef76320 : SUCCESS in 6m 46s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/46b0c9db17c9465abe7346ebdd645976 : SUCCESS in 3m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f190b7dad9d64c529034d7ad74e44d03 : SUCCESS in 7m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5a0a321a390e4140b35f61946a129ec6 : FAILURE in 3m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/dec15c35e3d74e9d94537f10d0dec609 : SUCCESS in 3m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/caa6f75b15414e5d815b8826e71a7109 : SUCCESS in 4m 12s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1040b8487c57497aa2850d8f9e8c8333 : SUCCESS in 14m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b382a5a6f6db4deeba2c15cf7bd27a3d : SUCCESS in 5m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/46429bd882a74ca48fcdd652cd1bd448 : FAILURE in 3m 06s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/746c4c5f83064e6eb5ffef4ae05849dd : SUCCESS in 9m 42s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/aa69c5e2d33f432384edebe0af96f8aa : SUCCESS in 3m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/dacaa1dc8d594510aecb445e9b0b29fa : SUCCESS in 11m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/680f2e0a8ae5436cb20d5e8f3431cb90 : SUCCESS in 2m 33s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9d4338f5d6d04fe0a77d671a215076ca : SUCCESS in 7m 06s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6b70e2255251442dad1ff6244916bf65 : SUCCESS in 11m 30s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/497e84e6e6db4e3496d9de61a025c57d : SUCCESS in 22m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/946eaf4a05254175986e534e17921869 : SUCCESS in 5m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/850f5fb05c434140934b6a6c7723188e : SUCCESS in 9m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/99dbf0c6aca845cb86ec006de64a2e18 : SUCCESS in 3m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/86dca419e7ee4bb797f6c83d3a3b4727 : SUCCESS in 2m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3cb11ef48684ff08c48c3aeec3a17ed : SUCCESS in 23m 57s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/061d1e95c1974ae28eb5398394b744ee : FAILURE in 2m 38s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/3204166b64f148c0a67a35a485a165f6 : SUCCESS in 3m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3d8a7d22db93424eb7e942c6371c068f : SUCCESS in 3m 37s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/21bb2ca64fe3433d8768087ec8283d6f : SUCCESS in 16m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dedac89a94104743a078a1e6d82e942d : SUCCESS in 3m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9456539bae914718b4a41431cffa1414 : SUCCESS in 36m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03dbb199b2554c729c8fd6fdd8a815b7 : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"844f842b4dfc7f62e06f6f4b37238728771c76c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-23 08:17:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a60cd7a2794fff5313cc66939c1dedf51ff9e195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 09:11:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f8b8fc8d93e4f82bcd1c1d13ff4e82d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/887aceedb4094f8786b19c7e9c3756e8 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b54f17c3e5224777a721a01d746cbb18 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed71e47f29774ca2aec0c4c521a68224 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97b2ef62432a48c69fd681dd0cbcf0b5 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90ffe6242d3140ddb7e71fca85f9c4cd : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dec7995e4380402d9e04bb8f688d8259 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d7f4ad0375b4b63b1e53731e850aa8b : SUCCESS in 2m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/35dc3e8769194b2db046d2596dcfd022 : SUCCESS in 5m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2eb1a81ddccf4aa29a337ca0b9bcf7cf : SUCCESS in 12m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c54e490cd0414ecbb34cb92c4b5ac452 : SUCCESS in 10m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/984de7fb41ad496b9a2920e102085672 : SUCCESS in 24m 12s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/98b0df06e658403e90c1a54ac516138d : SUCCESS in 7m 33s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a4bd3d122cc64b1aade4304eb0b87c4c : SUCCESS in 5m 08s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/079faef998884a1aa67098ac48d26345 : SUCCESS in 5m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bc2880c44e434994969e4590800c4246 : SUCCESS in 10m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4ce3f24dd5f44ce8b67a61802964ad7e : FAILURE in 6m 29s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2ba0a1e61aef4ede9b86eba8e0460cbf : SUCCESS in 3m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9cbaf5fcbce44b20ba6c56bb793fe323 : SUCCESS in 6m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0a740c6a51b740ce9f6675a7852f8abd : SUCCESS in 17m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f4adae522841434d96cf035849f464de : SUCCESS in 8m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d7ee7aad575844c69e794c67874916d0 : FAILURE in 4m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/db207b6f27bb49539fd2edaf29e8684f : SUCCESS in 10m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4e14e72cc9614b2a9223c71d51414549 : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/791b65cd78e64d6b817566e9a4c2867e : SUCCESS in 7m 18s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4da3e1120213424ab1ecabf8719b217a : SUCCESS in 4m 49s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9eb12d75930f4542a9da765ab1e78fc6 : SUCCESS in 5m 13s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4c9e1d3b1a0244ef932f3e3371bf5737 : SUCCESS in 13m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1752b8180dd54dd1ae31e3f293935134 : SUCCESS in 20m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/81184b284bc14c168a5e29a4e543b526 : SUCCESS in 4m 08s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/38413d488f7c483382f5018ee2ab5a56 : SUCCESS in 12m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0891cb2f828047cf99f1d021d8237ef8 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cac8df20361144e08cf381e1bb6bd797 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63d0c4a219f64095bdc96200cdf6f3f1 : SUCCESS in 26m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/39424bed748f4ed7bc0195eecac87079 : FAILURE in 5m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9ba39ef772d34bd6959507785122e3f5 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ecb06dfd46834e9f8696b2e0804ea708 : SUCCESS in 6m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5232a4026dd34195a4bd1636c4d706a9 : SUCCESS in 19m 06s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4dfc5ad6f9034fc5802be22e3b244e47 : SUCCESS in 5m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c34e05401ac44b6be5bebea3601e679 : SUCCESS in 51m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/245db516a9e24e2ba3fb05ad6a4ef44c : FAILURE in 16m 36s","accounts_in_message":[],"_revision_number":2},{"id":"129e5e083e03b00cf73bfa53dd9bc1786690e758","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-30 08:16:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c852c026cbc4c87036266b510c273bf866c47639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-30 08:50:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6df134e27e174a74b0dc9226f4efd827\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cbac6b4ce0e412ea15d47e27b8d364d : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e6fbc42f38400da96f23fcd49e9356 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ad2de01b6b4456f98ebeba1fd7a1b2f : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c9ba425da9241878983694bc12d5e19 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24a51b888ed44c5da0b6e09abb5922c5 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/075747e3bbe2422bbc8f289bff27646a : SUCCESS in 2m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61912afd318b4a07be35f27efa93c578 : SUCCESS in 4m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e0bb3c059e184215a5528b5621665e57 : SUCCESS in 6m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/63e7963810634aa28aa1a36abedaf10b : SUCCESS in 8m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f25df4e2f2d442be8dd6221440cc8f87 : SUCCESS in 20m 51s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e6a99005065344e48f14e5b916fa711e : SUCCESS in 10m 11s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7df3b49667a14986bc8f7bdb5fcdff1e : SUCCESS in 5m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6d130a8b8afd494889ff2d9352b58d64 : SUCCESS in 5m 54s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0b8c7de4018a4d3ba5141154d1e40578 : SUCCESS in 7m 36s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/732808a246ae4feaa30031f655e634fe : SUCCESS in 11m 34s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3387cb8a72994dee86e82630b95fdf3f : FAILURE in 5m 13s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/27fa0f7002f04eeeb8ea39b84ecc3bdd : SUCCESS in 5m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7789eb53acb145428db83ac3e04fcd6d : SUCCESS in 7m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3e41357f969e494ba0d0b13896a9deda : SUCCESS in 18m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ca11d41a6e3744ae809b7625a9146f8d : SUCCESS in 6m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/eee4dee827424674872c23ef2b4bc83c : FAILURE in 4m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b0486af29f64428abe9f0e96db1e9335 : SUCCESS in 11m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1f1ad6b7a4a14aefb8b4cc15daef3aba : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/18aaae0ff5294f90b6a70be8afb66426 : SUCCESS in 13m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ccaecb5e59e1480db3047c1366f6c211 : SUCCESS in 3m 57s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1eadfd4e20d84fb098ab0536a914d22a : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/61cbe19654564c86b972dd522732c773 : SUCCESS in 11m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e72046087abc4716a40d8cccd3ae5642 : SUCCESS in 19m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3cb44c9fe0a64782beec913603a2db04 : SUCCESS in 10m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/12464e19afab4120b1fb1927d10c3dee : SUCCESS in 10m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2a8f9eeb7db14aefbeeb49c253752e99 : SUCCESS in 6m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/92be4a2fdec04b0ea67a2fa63bfe1423 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2641d0e1920042efbacdf694df104522 : SUCCESS in 21m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/eb357876365f46a9827d154c65c853c7 : FAILURE in 6m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6d61a7b824184bbe9105bc970e091f7e : SUCCESS in 4m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f91ce698acc4f20853d8f8e9d0189f7 : SUCCESS in 3m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dd1450e58fde4964bfc9121694c4fbd2 : SUCCESS in 8m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3ec9b52b32f341e59d6b1fd6450f4c3f : SUCCESS in 5m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b1ca4d915ba4742a4704172299e32c7 : SUCCESS in 32m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/289f7e09c8d64f0691a9819fc09b1822 : FAILURE in 24m 25s","accounts_in_message":[],"_revision_number":3},{"id":"6f27df3c397007abd16796c264d16ddb3261af01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-09-06 08:23:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2aae6c093f36e6a1af4a9d9319c9cde21907ce0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-06 09:13:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cda5b39f0e944be9c1773a2f60624be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f77b683f6e3044a39eca8063dc28ccf4 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/143263a6c0e04d7aa1743ef77adfea58 : SUCCESS in 2m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a743b84e4d9c47b59ac175f96f724f64 : SUCCESS in 2m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a27a6eef34e4e06b3912d2eaeadfbdf : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ba9328c0b5143118b59a1d305f31a8f : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/913f028468e541ffbc27d763b773e88f : SUCCESS in 2m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f15d6d9ff9e743ea91d3dc20d6b4489f : SUCCESS in 4m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dc5d4d1dde8342feb9bc549f09b44eae : SUCCESS in 15m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2e8ac106352341e3bf9a9b6dfdccc92d : SUCCESS in 18m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ed3003fd07954296ae7caab9aa3efc40 : SUCCESS in 21m 39s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/11050e57312e41e7af06c006e4c73a8a : SUCCESS in 26m 06s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2f9c1beaaeff453dab66428f3a1b9f90 : SUCCESS in 6m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/10d9b7ff9e0f491896882f3ddf111e6f : SUCCESS in 6m 34s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d8b6ecd9a82645a3ab9f0dd0dcc368f5 : SUCCESS in 2m 42s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/925ac05626c34c0c9fc2bde80a842117 : SUCCESS in 7m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/89fdcf0863294e2eba2202937ce23358 : FAILURE in 5m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/620a7ce19353406e804f6b45418ea29a : SUCCESS in 5m 27s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4ede356698f64d6890e0a6f12009276a : SUCCESS in 6m 09s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6472db3bc520450ead053ae350a83abe : SUCCESS in 17m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/800e5033d9094ecdaed530c1b65b4823 : SUCCESS in 9m 34s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ae85fcc1e4fc4a94ae0c4431d093aa23 : FAILURE in 2m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8cdf223f9cab4fdab28edfbff9830ef7 : SUCCESS in 7m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dbc5da4b13fe404a9999ad89a48f15a0 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bbfc944e9f3c486e969515bae4800c39 : SUCCESS in 10m 19s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2e0eba4e089d49ef9293deefa7f5adda : SUCCESS in 2m 45s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/193373b22d6c4b85806e448b9e3d8d5c : SUCCESS in 4m 38s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ceaf2869abc04b23b645bcb8b5708eba : SUCCESS in 12m 23s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/22628d1a67af405db1f3b2659049dec3 : SUCCESS in 22m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3dcc214ff5614583bf290ce5de071ea2 : SUCCESS in 7m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a706d4b09d0b4bb4b555f35cf905484b : SUCCESS in 8m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/90ef52c146934e68895b922b28110617 : SUCCESS in 6m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fe8e2152a0ba49f1ac68690b20bf2ce3 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/994f08705db4470f860e8a3872c87d56 : SUCCESS in 25m 29s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/41270a03ea324322acef1fcb954d5487 : FAILURE in 5m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d0065566a1d045b6b251ea91489afd1a : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f7996d84cf5c45f293414673c2e35377 : SUCCESS in 7m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/277e5c279d924de59a6fee9adb7e8cbd : SUCCESS in 18m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c1371c7cb2274f2f8ccdfb14799c3299 : SUCCESS in 7m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/030c87be7cce44968208c7db0fb3c328 : SUCCESS in 48m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77fe306a2d5c4000a07aeeb9c6825120 : FAILURE in 17m 46s","accounts_in_message":[],"_revision_number":4},{"id":"7db317227e2d46e8848af961c884f158b9a24bfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-08 11:41:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9cb29b474b337ffdefda8f77f2588451d3433af4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-08 11:43:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f5694e6cc8a90bbda154e2a385172f04ec7d966b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 13:29:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4d01af32548435693a8c715de1c8b43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae42b300b4846209d68dbd636f42806 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4961b65f544845d9922650fe1f6b236a : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc6c5c2787a74c59b25f075f793467ad : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/873dbbdc9b4b410eacc354b202267089 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6fb2266b8524eeeab48bd254d9a2bcb : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79ee168af7fb45c697e0d646c8e4350e : SUCCESS in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/14939203586445ada3f2ccbe97e735c5 : SUCCESS in 2m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/04a0f36173114cca8b9df4fba037d68c : SUCCESS in 16m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ff64fd8ce6e04d09b50abeaf04e3394b : SUCCESS in 22m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b7722aa2f1d54187a935f3fd6218338a : SUCCESS in 19m 19s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b5569046ec1347d7aa6c629d6c770bb5 : SUCCESS in 24m 22s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5d6d677626b34954a11deb15cd02d577 : SUCCESS in 3m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3d0a7a56fdf04f25ad145aad5287efc8 : SUCCESS in 6m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7fdbf3b4bb1e422298a636cb7bd6cffb : SUCCESS in 4m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3e97b001e28d4336ad6b834b45ab0c2b : SUCCESS in 12m 29s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/410e0f3b2702407fa0a351cf97850cc9 : FAILURE in 6m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2be1e82ef3c34d6d823519bd36b2410c : SUCCESS in 5m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a6ddfee2939e4a73868dea8aaf9c706c : SUCCESS in 9m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cce30b0886654c3f8becda2e1f6c437c : SUCCESS in 17m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b253d839669451292b58ddb08e17fbe : SUCCESS in 8m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/294256822adb45c3a639686c545b12e2 : FAILURE in 4m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/09029d63faa647fd9d10a905e5ae4de8 : SUCCESS in 11m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e90b6940b6d647d9a26a0d24adfe98e9 : SUCCESS in 1m 58s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/10258f624cf74ddbbec73c58696a3219 : SUCCESS in 10m 13s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f916a8dc278e4abd85c334e23c9bea3a : SUCCESS in 4m 30s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1951a24d25f34673bb63c6741e6714d2 : SUCCESS in 4m 52s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/45ee24a232474a85928bd84ffb47c09c : SUCCESS in 10m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b5da0dc60fd740aebf91deb21cee2794 : SUCCESS in 23m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/780a293d17dc4f40a7d97eaf5673387f : SUCCESS in 7m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fd25dd8c3fe848088c1708c4a502b2a5 : SUCCESS in 12m 51s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d2dedc79a8ac435694903e0377ca1106 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/805b775620de488daeba4d37a03ef50e : SUCCESS in 2m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/331f88624efa46349b11376c569c4e7a : SUCCESS in 23m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cd64afccfbe34d61a6c6038b7b1d687e : FAILURE in 4m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5164e58b0f434a2eb045bdebd97b16c9 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a206f536918949df8756fad5288854f8 : SUCCESS in 7m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0b5c24eafde14ed2842ba3dc15d289db : SUCCESS in 11m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d5cc09d0a93c4232b8a5ea062bde6a55 : SUCCESS in 6m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a81848130a74877bda47f77dd254b25 : SUCCESS in 59m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7b9cbc8a05f4ea19654987b5ae27c58 : SUCCESS in 1h 42m 16s","accounts_in_message":[],"_revision_number":5},{"id":"c5e448f2289619c8797d8569e7cc3c5fa4821cec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-09-13 08:18:08.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f1f2456e32bc775f3f96d060113eecc346931cde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-13 09:57:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c248b6f103b4c2187713eeeea840b15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a58e409c3e54d4e97b52917e4e6f66a : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b62c9a109d48099f1e786a5393fc1d : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14857b681bdb4152b1e31132eb84ace8 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85865be526334c7297cbf1adf42170c0 : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d0f2dc74e8c4c57b7fddbeaddf7d0b6 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27fa6a1c169d4e2988fd122254f9975e : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af45639fed8141e8a43516b7a7d9c36b : SUCCESS in 3m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f1ddbeb7b92343e987251b29cea7b754 : SUCCESS in 13m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3d85684ac95b44019fae45e8dd9580db : SUCCESS in 20m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/da8b996f467d4529a639ee24ca6f4a9d : SUCCESS in 24m 57s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d79621f8b86a4e8e96b210c57fc95bd8 : SUCCESS in 23m 44s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/db03ee28b4eb40f6b7cbb56fe2cef4cf : SUCCESS in 5m 13s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2e3cb96c003d4f97be2dc6ff05f19773 : SUCCESS in 5m 11s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/122f9970198a4833a54bb296c20f9158 : SUCCESS in 5m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/72183dc604f44401b217b176b56d676b : SUCCESS in 11m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/cb78e903f14448769dcd245da46ac43b : FAILURE in 6m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b3114e25af85481792a9dd18dacc0fb1 : SUCCESS in 7m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b08d0d0210fe4fbcb8fb43009756fe99 : SUCCESS in 8m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/660374866d3a4992b124e2d07162fae4 : SUCCESS in 18m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e9c729d67284832873ab32811e7b474 : SUCCESS in 7m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/20030e694f3f44ee98aa27e311b3202d : FAILURE in 4m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9d85c0d7c6e7494fa2dca2d972e89191 : SUCCESS in 11m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9d029b2b5d1f49d4875925ecf9f49209 : SUCCESS in 2m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9bd4e56353294992a7b4fd4c4db2cbbe : SUCCESS in 11m 08s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ff0c6648aa2448728282c17b59bc8dc6 : SUCCESS in 6m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/555cf1da2df947faa0397d9a92c6dd91 : SUCCESS in 7m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9546d203c23f4e8b832a5f1a51c08c9e : SUCCESS in 11m 04s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d66802537a15475da652fd1cda8e3bc1 : SUCCESS in 20m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9ac20feed0fc44ec8bd510c4a89d85e9 : SUCCESS in 7m 35s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8260e5535e2145c986434f24e88b7dc1 : SUCCESS in 11m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b93368a3076b4da8857a9e2c63bc2237 : SUCCESS in 7m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/283af32f4ff8454a88b14868adef6db8 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ef43da36ccc49a280e9083e295b29ca : SUCCESS in 26m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ff3423473100426bae323121645916dc : FAILURE in 5m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5a354e5c952f480d84c15880218c4e42 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a69a925af9ea4da99cddb20a9496b592 : SUCCESS in 8m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6dc19cc42a084fd8b61cf9dda19bae99 : SUCCESS in 18m 17s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d4c6022d741f44b1bfba7936d370d67b : SUCCESS in 6m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d065cefd1486407c8f11189ab2304010 : SUCCESS in 1h 07m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e377cd38b7554a5ebd2a5cef433b8a76 : SUCCESS in 1h 34m 02s","accounts_in_message":[],"_revision_number":6},{"id":"f060332744bb0c9210a932391662ddde048899d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-09-20 08:30:56.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"e670b7bd6d656c9cb80c889f5b82282e2769452f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-20 09:36:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffb547b8159e4ab9a29c3539d4b42b6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22523cc40de4335ad738d9bca6427b7 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dec9a1098814e61bb7d5c67131c8242 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85e9ce66527b417f87912a84199c6f6c : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c51d523874a44b928be58c4fb3d845f7 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d01ada26eb34a5da5213bdcb83e687e : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0cb89ab1f70c4dc1bd0f30ce4f51a85e : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5fbe4d4a21324d2bb8a7c4912beabd85 : SUCCESS in 4m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b633e7be59cd4877b6e5c31612a37922 : SUCCESS in 16m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ac11e510fe494216b128fc5b9fd9dca6 : SUCCESS in 18m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/393f3697768c499c83c702757bfd95cd : SUCCESS in 20m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b0aa48302aeb4db5acec087c73685ccd : SUCCESS in 25m 56s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6746e4a0bf664dd1a590b3548e2c12d1 : SUCCESS in 4m 12s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3d4b85ab06fe45359b664fe5be6fc043 : SUCCESS in 5m 25s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8fc7dadb64ca4408acde4adc37de310a : SUCCESS in 4m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d399e858a1a74732b7772f3d8773dd5c : SUCCESS in 10m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bccf5dd15fb84b3697d242408b77ee43 : FAILURE in 5m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e897ee4386b3455db2bff4c0c468b03d : SUCCESS in 5m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/62e4248a9fa34fbc81d704111342be7b : SUCCESS in 8m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c3b17437896e466f822f9b57b64f5443 : SUCCESS in 15m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5f38592cb1042d19111be9141faf7f5 : SUCCESS in 11m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f2c5e58a18f84645b910410657a39cda : FAILURE in 4m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/571084e74fe14f3ebf1b3f14029e17ce : SUCCESS in 12m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f906dcca14e3407e912b6ff68213615d : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/022c6d1074e0449994ed2d19702f70a6 : SUCCESS in 8m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8987bef0a7e74acc81fc0b6e26039a6a : SUCCESS in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/edf301e6b2954374996de5876cedf9e1 : SUCCESS in 4m 59s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/9daf5ca2530f4d9cbc8317b5f050adc1 : SUCCESS in 11m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/46263042a08d4556bbad37a58b7f2a93 : SUCCESS in 19m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/42e0e81a9a124f269cf5a4b876c59869 : SUCCESS in 9m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/da2be0a6c52045f0b7fa9ddc4ce00adb : SUCCESS in 10m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/512ff69a552147ffbcd0be4c239b273b : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79feb7eca558445e80d97b447a9b1117 : SUCCESS in 5m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad497a06ade74c30ad44689d39a3bd09 : SUCCESS in 22m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d135ae1f0fd34daf8c98a498bc91981e : FAILURE in 5m 14s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/637e9a43a3564373b839d3a4b95377ab : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dbe10a0d41f84e10b1f110016e82e035 : SUCCESS in 6m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3bda2590fe29405b9572c90ae5980d96 : SUCCESS in 19m 28s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7e879ff0e6d24c80bd1299671bddaac7 : SUCCESS in 4m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fccd92a7838646ceaa1d02646911bcda : SUCCESS in 37m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48c1be065cf74eafb92606d2c0e09dcd : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":7},{"id":"d251d5f7107bc49c836bff8583243a11fd110487","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-09-27 08:20:11.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"7b6e9a41b0382505ffa7a49899c2d8a92f192407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 10:11:35.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e16356c19fb446709694b2bdcbc081e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70d098b1898b4da9a187e1467a217af2 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fabbeb3d856d4c5c8c8af98c106a7732 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02723902db01488b9a046263bc28f898 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05498179ccff4a45b25357431a084c0c : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdba120795a444f9bc3676a5d3318b53 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/169161f119d64058bd50e555b2a1908a : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61f395ca02414b09bc318ac695d62cd1 : SUCCESS in 3m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a87227c9999b4b7190c36ee615294496 : SUCCESS in 14m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c0b54ae1a37f4225b8c054f1ab7e5d11 : SUCCESS in 18m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/29d37caa377e4085aaa4a917ea329583 : SUCCESS in 18m 17s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/252311ead39b449e912be0cd8df18e1a : SUCCESS in 23m 39s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a62c00cbbd724e629c15d880e290e0be : SUCCESS in 4m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2b6b81ec46ea4a29bee9b63063b1d29d : SUCCESS in 5m 11s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/66062f852c69411d985f1c8f7d54d816 : SUCCESS in 6m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cc16a4294d394e08a4478c8796107989 : SUCCESS in 11m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0bd4bc33f7cc43de897351d51db6ac0c : FAILURE in 3m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a25fdff1a9164d7691c913fd2a8688ec : SUCCESS in 3m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1b1d01309d654427aab4ac69d5321c01 : SUCCESS in 4m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/40988eb42bae494da2bc1eebdff27c08 : SUCCESS in 16m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e4ca1736a72471583e672eeaca87fc2 : SUCCESS in 8m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a5e3f589f0d049d6840ec2f8cbabab8f : FAILURE in 4m 24s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/52d31c46d60c4d418c233d89644c7377 : SUCCESS in 11m 52s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5d06a3f134664eb08f53e5865ac16f43 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9ffb684619674251be05e72dc39d2666 : SUCCESS in 12m 01s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3fabc9d2cc214af1a7ad054073db66ce : SUCCESS in 4m 30s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e16a9d1916164227982a2a9fdd579124 : SUCCESS in 5m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2364fd910a024850a0bb06b77f0dc486 : SUCCESS in 9m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5861cabd8a564a51bfeb5e9f38e44af9 : SUCCESS in 18m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/739043ba17a34e15a215459a45128c4c : SUCCESS in 8m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/314f665574204ee7b890cf29e11f4942 : SUCCESS in 11m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9be13510eaba40828d8ea80c5340d184 : SUCCESS in 5m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0c860e055313410691990b59099e7ea7 : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4f16ccddf20463aaaf00cb16c27b456 : SUCCESS in 27m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8a02f8e2aeff43b1af9fa5d50dc74588 : FAILURE in 5m 57s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/cb4075c5389942aabb42d28a23f4c2bd : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4deacf61c8844779aab0a4d55367783c : SUCCESS in 6m 44s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/aeef8815210d4c7fb1ef249b2c712116 : SUCCESS in 21m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/39aba1e3f4824c79bc06c1f7100c9ff7 : SUCCESS in 5m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a1e46784ebb48d292875bf93121f37e : SUCCESS in 56m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63a16d0517ec4fb7abdaceb32dadf869 : SUCCESS in 1h 50m 20s","accounts_in_message":[],"_revision_number":8},{"id":"69674bd7559c43de0dedaccf540d2c2d25069962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-04 11:05:26.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"05f5cacc860e78314c919599e298c2c983e462e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-04 12:16:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b44f774301b146f8b5b072ccd73ea9f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eb3ee8b8e3e42799f517cacb1f27069 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/983551c66ba14fd784807c06a6975866 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/136b7bb3b1984807bbad452db553bb31 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e28c8d703214afa8322355dea802b2a : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e645fb369d64f93ac4de1ba18cfba1e : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/689668118a83434e973efbf232f5ff2d : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f81a59a3632b4871ae697f45ddf4e605 : SUCCESS in 2m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6c0b4cbb02c1440f98b773026f13a19b : SUCCESS in 15m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1fd0ace5c0564da999c27eb94516767c : SUCCESS in 11m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/da1f9757bce249efa7ee032c9b96491b : SUCCESS in 19m 35s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8b01c37b0c094fdfbc7e17e4c5c1f7b9 : SUCCESS in 23m 46s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5aafff5a24254ad39cf37c2a33f42bc1 : SUCCESS in 6m 17s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1ae17fd6f77a4bc79b90d21ae16847cb : SUCCESS in 5m 09s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b8354dd1c03a4d7a82b1d8bd85920310 : SUCCESS in 5m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/febdc645ec61407799ece61fd66b9f2e : SUCCESS in 12m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0312e1ff2e79402984fbb32424bfd6b8 : FAILURE in 5m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ab08ced4e0a94b14b7e7e9c4b841f7a6 : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a6f1ec0b67e74e1ea2183c4e74ffb8f0 : SUCCESS in 7m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7730b4e67cd64674b97a2ede9978d7f9 : SUCCESS in 14m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8d2c39160994b63a49f0366e57e06c9 : SUCCESS in 7m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c38a8fcaf315453a9259e36f49ce8783 : FAILURE in 4m 15s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/705032c87c634b0dacfaa174110e1d4c : SUCCESS in 13m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/53d25a452bfe413d898801f3e4d6f79a : SUCCESS in 2m 03s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1d6bc219d1114575ac979f9589d198af : SUCCESS in 8m 42s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/eb53115ff9b941c8b3a47b342f52d07f : SUCCESS in 4m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ea06e77d9c524272b3020aeaa2c7d0b1 : SUCCESS in 5m 15s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4c35559504a2436b913663a70869afe5 : SUCCESS in 10m 20s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7ab9807a7c5f4827842103316349dbdb : SUCCESS in 20m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dc8e1c1d606947db8398f5910f002b3a : SUCCESS in 7m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/520cf949ce9a481a98ad4e2d7b36b80a : SUCCESS in 13m 12s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7004e11e12814541a92b0314f71a60ca : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1df799aec18f4041b638752f97e3b079 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1f2e14413234db79b008c874ee98941 : SUCCESS in 29m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2b8ccb221b2f4d6f89c4226b02e02481 : FAILURE in 4m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a6e5112c99244e2d84e7b9bd116efe7d : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e111d54b08c4451aeb7d87228e0e6e8 : SUCCESS in 6m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f58913252d73422986d7c8fa37145fb4 : SUCCESS in 19m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/841051be421d49248d8303bd42f2c8d7 : SUCCESS in 5m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b90d538f158431ca3eb805faa04b320 : SUCCESS in 1h 05m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5decf7b6ce24ba0a9741fefab01c6a6 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":9},{"id":"b2cdd2bb543d00cfc02750bb8c825c6d3b998394","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-11 08:25:26.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"558098920d5e30d339add81daff78470172012dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 10:36:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c308842c3244391950920d7b4550481\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dd77975f3a049328584abf602fa861b : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7e1694b60e84d319c1611b6a419b271 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4592f4ed82ce45559daaee28550ca016 : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da426fc820d941b88216fd737951ffdb : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cdc0fedd771a493f95b010d024474e79 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6ea356c91374f5682cfd7f1b23bde50 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/33228a0cbfdf4540b436721b6acf5de2 : SUCCESS in 3m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1a77e8caf04a4241b2b0798d24f7a624 : SUCCESS in 10m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/493514f542f04cc8b3720c75dea54d4c : SUCCESS in 7m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6c1952e4a74c4e6fa760e868676162fd : SUCCESS in 18m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ab81fb33ed354b0c9acb0875de8997a2 : SUCCESS in 17m 41s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/308115c80d9a40edb81be82ffe3bc075 : SUCCESS in 3m 00s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e24d5270c9654a89af73ea9a6b5a5f57 : SUCCESS in 5m 27s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/cb88fb566e404c35b5fc3c3d03b7608b : SUCCESS in 4m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/089b02b37be5475a963901a9a0def448 : SUCCESS in 6m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1100ab5d37874ed1a1520f0483d73db2 : FAILURE in 4m 22s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/22b49bf7fb424374a10d75c070189f5f : SUCCESS in 3m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f97ca5dad1cb4072b7554d38ca55e052 : SUCCESS in 6m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0b851b3aab6643948d96e6de509703fd : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/def1a4410bcd4881a9898926bd1eb2d5 : SUCCESS in 7m 38s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2df5bca47ade46d39c22139ddec5868d : FAILURE in 4m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7080b7770167443d9701e6c650c8aed0 : SUCCESS in 7m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9d61ea6000234d86a13d2d543a7eed79 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8ecc1f05348c413aab628bba4286196b : SUCCESS in 9m 41s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c20560c28b2548358469980f591aacb1 : SUCCESS in 4m 38s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d8a51e7b7d3d4aafaf77c7a3a17e8ecf : SUCCESS in 3m 31s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d060bcecdc614f4bbe968a9432f9a13b : SUCCESS in 8m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e67f73629bf44ead91b192707cdb45d8 : SUCCESS in 19m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/290d6ca3ccc9432dbd5b05d2a0bd1bee : SUCCESS in 7m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7cf6393716474326b366f5b23b385883 : SUCCESS in 8m 27s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/93a93ee3cebb46aa97c9017833bde29e : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/267831f3dc06437e92e6aa7122ba73d7 : SUCCESS in 5m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0848b53c6fe5459bafb981ea357732da : SUCCESS in 23m 55s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1bd2a33a69a14040a6ecfab7d27ea7bb : FAILURE in 5m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a4b156dc929d4406b64ef16fa007e9ff : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b2fff849cb594e919373201022b5a676 : SUCCESS in 5m 30s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b2324780c213475e8ae46121eec1a316 : SUCCESS in 19m 05s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/761a872d51fa4b63853272b106c06775 : SUCCESS in 5m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2d4b27d5a1945e19a1693527e65249d : SUCCESS in 59m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5162705aaa3a4800bd08e1ce18351d1f : SUCCESS in 1h 47m 20s","accounts_in_message":[],"_revision_number":10},{"id":"620cec486b378210212b08816a0c01651abb1a64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-18 08:21:45.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"764aa14afdcce10727ed04a4871871ce431f10c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-18 10:12:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec7e068f8fb345ad82bf1aed39f3da5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fa73eb1e53c4563a1895e42ed4095ca : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a14c4f39d1349b5bd5addf2a2c609be : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f9bdd4063ad45fca1b71889eb0983c8 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83cf45bf74cd4fdbbfbaccfe7e2c1404 : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14fb4a0a1d604acd8742bc1c7f3b51eb : FAILURE in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc61ee5aff354eaeb1075bae506914e1 : SUCCESS in 4m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/424649d989514a4f92e0a55278f7520d : SUCCESS in 18m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3aff47897fe54c3eb01e682b3e52de99 : SUCCESS in 19m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/30dbb018d021430cb30f0ab0431d9efb : SUCCESS in 14m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3b492c6100114dc79d613787cd74657b : SUCCESS in 19m 31s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8f22bae5e1a34152afc5b98db959fb5a : SUCCESS in 2m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fbaf778aaa8b4e9b9cb45fab0d188e46 : SUCCESS in 3m 33s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/da03e4a547824f7693e3c4f66094bd29 : SUCCESS in 4m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4cf883e616324fd380b874447e4c1ce6 : SUCCESS in 10m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8f61c7bae49c43478e9be069bc2ca533 : FAILURE in 6m 28s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c02d6c451b204b68a7fb97cfe1a9f702 : SUCCESS in 7m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/31ffa7ab08a84315b897ff609d4b612f : SUCCESS in 3m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/425adee424e04b2f943d14f05e09d4af : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b0b99cd16e2444f989a9094dc9cb8fb : SUCCESS in 6m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bc480f47fbd7424b8bd7105ec4b05dc0 : FAILURE in 4m 35s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4970709900444a1285dc3351faaf4b41 : SUCCESS in 11m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e9f2b19fbf314574a853bbacd1c7ddb8 : SUCCESS in 3m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/20ad03c1ffed4044bcbf3b3f183fd930 : SUCCESS in 6m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/cd44b918d1b845b8b66e7797a99ae0dd : SUCCESS in 3m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7c69e37c79cc4e0eb7d9dfd026338963 : SUCCESS in 3m 35s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/045e4eb9c2544aaaae8c67d01159c332 : SUCCESS in 9m 12s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8c44ae077ff24ce98f9f60eb88862fb9 : SUCCESS in 27m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9b545911057549539bcb62ddee12658a : SUCCESS in 5m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ffbca84f565c429580aaf34c56f622fc : SUCCESS in 13m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ec0e597392e14c2aa7b8a864d82232d1 : SUCCESS in 2m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/78f1e47c707a434594990615624b8175 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ef2fb88eaa34e41aaa252253056d8e2 : SUCCESS in 23m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6dc05d8f96c7411383b80e5003b9ee56 : FAILURE in 4m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/1153c184fb824b4cbbf97f7a3e5ac435 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbd063e1f5bc4623b352b3c6d7b51d93 : SUCCESS in 5m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/77ea7afb3723446cb6c7b77d92601c9f : SUCCESS in 14m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bd5da9b0b4f14d8382d38f51012e364e : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2662ce8faf644b52adbbd26ef0f1a969 : SUCCESS in 54m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0f3fc913f74470eb414651bbcf42c6b : SUCCESS in 1h 45m 36s","accounts_in_message":[],"_revision_number":11},{"id":"ae4d1e7d9713634990717aa8e094eb90927a53bc","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-10-22 18:08:08.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4da490b6b463e4806f11681106cb64d1b7315312","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-23 09:54:00.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"a88612a3df8bff0cdb862932b8dfb004f2f7fae5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-23 10:37:07.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"f9d2daa52eac36ad3999dd424b8e0b2d133f8b4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 12:31:57.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13d3ed9635264987bcd89246e5ee9448\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b93ab9dc7ff24b249ba2333988207685 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/912b3d74c6644ea9ba82bcd3377d17ec : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d94a520ddff649cfaa30cfc1afdd3b7a : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/338d4f5c50f04f2488cb25ce2787a4e5 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75631420264947a8813bd00855cb631e : FAILURE in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8519da871f2449159b97add8af9121b4 : SUCCESS in 2m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4c35ad225e2344478c18fd472fc3ebb0 : FAILURE in 3m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8f17914a59024160b7ebd68733457cd3 : FAILURE in 4m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/18a00bba087948aca18ac94221ae2099 : FAILURE in 3m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c96456bbd2354a689c3edb6981109c0a : FAILURE in 4m 44s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0a9d916f5b8f4bbf9259aff06620b8dc : SUCCESS in 4m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d867f1cbe6bf4a75b716a0cb8b6a5b60 : SUCCESS in 5m 13s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/84a85f158a89464aa1fe91fa90741ce2 : SUCCESS in 4m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/00aec8e978d74ce5aa9db3fc439fd0a5 : SUCCESS in 10m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/3947e59d389d4a62815c8bf34786d64c : FAILURE in 5m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f6e8ddd9ae2b497f8bc8dc43a6087763 : SUCCESS in 5m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8c99d6e5fa2e4099ac4693811a029727 : SUCCESS in 7m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e927d5958e8b45489ee87d12b704ba02 : SUCCESS in 17m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4f5b8b23e3944c49a299b40abb71649 : SUCCESS in 6m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1d5cbc456095440fbac8af19849e7251 : FAILURE in 4m 25s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d8cd7083159449b3b1496b220ae33257 : SUCCESS in 11m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/daf5db673641470cbcbb1cd124e4532f : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a016c4f471f24ada99a762d53687a3d9 : SUCCESS in 11m 46s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9efa4130a95842c08faf088c7c0ff222 : SUCCESS in 4m 13s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/15560543bced4b93beb25155c4aea307 : SUCCESS in 4m 56s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/556ea5f39fa94737bf4f7e1d1d880dbf : SUCCESS in 11m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/43f0a5e0ccad42b89a5fb092a31dc1f7 : SUCCESS in 28m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/01d554018dab42939c66b2bb73d6a4be : SUCCESS in 8m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a2bcc63cc9c14c859f45a6fd50b96701 : SUCCESS in 13m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d55e3100127046afb8d2c2fbda047482 : SUCCESS in 6m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a676edd746b9431797d532ce011fa422 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b2a85eb95e94243ad8cc7464076cd1a : SUCCESS in 29m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e179d2e2e4474a23baaa8f28648d1563 : FAILURE in 5m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6a2042f81f484b45b914b57370977652 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/34a2642777b341edb5ff09d761fbf173 : SUCCESS in 6m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/93ba9e9bf86646988378e30084ef4ea5 : SUCCESS in 19m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7b86b9debaf642e393a2688932a28e36 : SUCCESS in 5m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e2f3cffe7504e2386c6668314b22c3c : SUCCESS in 1h 01m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7f46c3cd45c470383d51cd3f9293226 : SUCCESS in 1h 48m 53s","accounts_in_message":[],"_revision_number":13},{"id":"cc0aa16d8963624fa9eb03c3ef5958d2329e4353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-23 21:31:09.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"73b6c7e314aef9a6b2aace76ca5e904e7f66c16f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 23:23:10.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fcbbf2c0a4a4a508512df9e5a26e960\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27e251f63eec4209a0a4781f4c3c2038 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/487ca8161acc48c5b2554629a83baab8 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5c5088319a34af1890538187ffc643f : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9caca731cf24e358cf531ea490c2b82 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d43a9ea02c841fb905ffa9c669e64d2 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aed2445c621f46f295daca0fbbf5cb17 : SUCCESS in 3m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/865ab46f03f24050929a31f75905f9e8 : FAILURE in 3m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/78986fe609a0409692c05885a4b34699 : FAILURE in 2m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8db838b975934d248b1625f14f7850dd : FAILURE in 2m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/599468d8d28e40ee9eb89dd1ac6c53bf : FAILURE in 4m 52s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4273ff85830a47e4a5b1261497c617e9 : SUCCESS in 5m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bcf6f129ad144b7f8b32f28072875a87 : SUCCESS in 5m 23s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0a38470ad6374d6fb545cfce79a7ffb8 : SUCCESS in 4m 19s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c3b3d88f84db4accbbf107b62fdff372 : SUCCESS in 8m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7700b760705b44fcbdfdf5fc726013cc : FAILURE in 3m 13s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0f6c65e6acb847839b9a6490365bf721 : SUCCESS in 5m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d2f9666abfbd4e8e860269b1971ab850 : SUCCESS in 7m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0b4eca03e31647d88b65c4ebdf5d8a55 : SUCCESS in 15m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b39b5b45ec94b5e94f122db8d1328ed : SUCCESS in 8m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fff3162a3927494d8f0bc122a2ce61a6 : FAILURE in 4m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6e6032281c2848e5a57f5d212466dcee : SUCCESS in 11m 45s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a49d0be3b1dc404ab54c02fed8563c72 : SUCCESS in 1m 35s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/307050f5fedd497d89615eb3d6d4228a : SUCCESS in 10m 07s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/83ed55542ca64e50a9c20b02f5f315a6 : SUCCESS in 4m 39s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/836c46c302fa4d3d94d834885d22e1d8 : SUCCESS in 4m 59s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2f0bf88a1182421cbfccedb4d6486923 : SUCCESS in 10m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/35daa478e7334210a0d207837ba52770 : SUCCESS in 31m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/abb028a7a4c44a858479ae14409e28b4 : SUCCESS in 8m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a6ab754c791c4121b62fdff8efc711cb : SUCCESS in 14m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7256be1219d040cc821b8014ba971901 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/978741dfea434c7faa1512e2ef148274 : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/35939fbe3b3a44409d2c391d6e3dbd7d : SUCCESS in 26m 23s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2ea01712ccdb445082270c0c16cbf9f4 : FAILURE in 4m 57s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/44b9a20f2be9475da0cd0604e04d0b17 : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9bfe724c5f964359ab5437666f6ba5e8 : SUCCESS in 4m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/764f40d46ed1427ca4e97f9427c0fe38 : SUCCESS in 18m 44s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d12d4111f1404ae9a4ab6e655c64c261 : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b827473615b14d87b2a40d78e7101b83 : SUCCESS in 1h 04m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/456a28d33cf14e2484eab6a2ce7c4f36 : SUCCESS in 1h 46m 42s","accounts_in_message":[],"_revision_number":14},{"id":"ab1b785b92028a3f32f1521bd43d189a3c72d80d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-24 00:28:35.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"e9acef1ab480f86c963da0a661077e92a46215ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-24 00:49:02.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"0b569aff8beb8e0dbdcc6064dc09577203dfab92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 02:34:35.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22d8975f56b545fa8bddf9ce00fcdb12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5172b07d7bbf4bf2abc838482e1d709f : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a999e800fc453281c848f79cd58e8f : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5912aa4479b74347896989ccdfa52f11 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9845a2c677643548f14d598c1a4f7df : SUCCESS in 7m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f037864c70e4057ad9c2e193aee0425 : SUCCESS in 2m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1678bf135d8d4301958fa418ce87ea5e : SUCCESS in 3m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/da5b77ebfc3a418492e37fd28e658779 : SUCCESS in 6m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/458384f424a74e4a8fd857c9b07e155e : SUCCESS in 5m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/857b7d20469445c995926017f7bbb0e4 : SUCCESS in 12m 32s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/db24cb45d60d4f7d914f21b2a6db57a4 : SUCCESS in 4m 58s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/afbaf7baf4b14a03957291de92269b10 : SUCCESS in 3m 15s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/46f2b489246b413d8b558d5b513e8762 : SUCCESS in 3m 34s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a6ff596787f543bb99a5fab931133197 : SUCCESS in 3m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/89ad8cb113d94578b3e01059e12bde5f : SUCCESS in 13m 05s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/eb081056455a426e8e9bf5a83209c451 : FAILURE in 4m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3a3f995486c048d8b36b0968fe05a743 : SUCCESS in 7m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4446866ee4be4675a5fd97a9ed63f430 : SUCCESS in 9m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1738634eb6434dc5823eb12a4e88142e : SUCCESS in 17m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56b79b74ff3c47d8b8e9e73ba28f44b2 : SUCCESS in 6m 18s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b825bba05c19460f934d7ee94547c798 : FAILURE in 4m 41s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/77576e8c8c524b12ab7470b98f19fa68 : SUCCESS in 7m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b0d99fef5fc34a75acdb2e5861e863db : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7b3dd33cece34e7a88d0f59e23ea947b : SUCCESS in 10m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/534b3d8c1a1e4ffc9531bd069f02f85e : SUCCESS in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2bf3d118ecd64487864e424fc121de76 : SUCCESS in 6m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/afb66b26a3044a949f65eaf33ea5e66f : SUCCESS in 10m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b1051be4aa3040aab019404894e15645 : SUCCESS in 26m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b5e2e25d9ace47ca9925973b3db566ec : SUCCESS in 7m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/694c7d5b108849f99d4e4975867cb531 : SUCCESS in 10m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d189bc54a7b54897baaf5384fd8bd79d : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0dc78c12a886461796b4caed026f8693 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e72317e5455a48f598fdeb1a07fbe959 : SUCCESS in 24m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/19d0de620c614cfdafdaca923fbebabe : FAILURE in 6m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/517310d3b4a04d1d9a973ed18216815b : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/580343b02bec404ab866e5f949c746ad : SUCCESS in 5m 24s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3f396dc3b322458a8e35c9f24e62d6f4 : SUCCESS in 19m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c585637c0f1c46949ed1001143b97272 : SUCCESS in 3m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de8f8a05cce8429484cfc47d4708c30d : SUCCESS in 56m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e7ec32114864f6fa25eb01681fd0dde : SUCCESS in 1h 40m 51s","accounts_in_message":[],"_revision_number":16},{"id":"44abf5aff80e10e2dbf6b8a9af49143926b259a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-25 08:15:34.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"5b2c49fcec3304223a789c6f0a53233173de1492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-25 10:04:21.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11ef7dd3063b43ff98b17360fa9dd8f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc52cd80e78644eb91dbe88b6af20d1f : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae1993bb30d740248ba0cee506013084 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a23c17b8fa0d4893a8a0513eca3bdd0c : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/933c6467500a4b6ab13df9436ae39f2f : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b2f6a9aef564e36a3cc6351cfe8a085 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c0f7b1d3e7d4a1883c40cd5b5cbc5c7 : SUCCESS in 2m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0267a6abeae54869b92d87c83f0cb27f : FAILURE in 2m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/465689b55f6c4162b980cf6bc1ab1ba0 : FAILURE in 2m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dd0e514216ee4bee8e6411503e7dd74b : FAILURE in 3m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4d14814428d44ae794473489a76fccc1 : FAILURE in 4m 50s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9f67fb460c3a455daebed343b9470f77 : SUCCESS in 4m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/65b3bc4ca6694ad1befbd209992b7527 : SUCCESS in 4m 42s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b4e1af5d1dba4fa49d11cbff5404333e : SUCCESS in 5m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4db99e86c17d48fa9d25c9ff2dffccb0 : SUCCESS in 9m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e50da102274d4bd5a27ef8e22ab01276 : FAILURE in 4m 44s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5d2e6ed4184f49a58bca9a30e991ee5a : SUCCESS in 5m 38s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cf3c4ee7b0444120acd744544fb10350 : SUCCESS in 6m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bbcb5ed8f91b49f6966f92c807f547ec : SUCCESS in 15m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6331261e31f94d3785444b17bd625ca9 : SUCCESS in 8m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9f00f1b2c6304dfa94612e766c3d326f : FAILURE in 2m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1cd05853b94f4223b4f06caddf168b57 : SUCCESS in 12m 31s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1d5f9d4fc6cd4ee1bca613a6732211b1 : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/696b8277b1ce4c25a3c7713b93b04a11 : SUCCESS in 12m 13s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7d6d471beaca4525bf577eee527b8b9d : SUCCESS in 6m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/866463e6d903441db72462ed90d4ed2b : SUCCESS in 2m 56s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6399d918ae9d4e17b38e756f4e3d7053 : SUCCESS in 10m 50s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/018971fb62a44431ad8b3b1db6d577cc : SUCCESS in 21m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/403632e6833d4f7c93c880e30dfbe1c5 : SUCCESS in 7m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b818828747d847ccb37f391f620d0fac : SUCCESS in 14m 31s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/bb82aa97b19247d3b5a1721eeaa901b7 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2120c9566cf405882f764929e36ef4d : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8b086cba3a274596a53c707a7a851e38 : SUCCESS in 24m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e037cfbf50a946a58d0ce8f51f8334cd : FAILURE in 5m 09s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bbfdbce4e4f243dd98ccf122fa527450 : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/964ab835da8d4c06a394dea14b9793ce : SUCCESS in 6m 06s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a9df67d661b7424483fe6ad6ac837aee : SUCCESS in 16m 00s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3a4a5465c0424550a28920bcfbc96e8e : SUCCESS in 6m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f179bc2d04d4508a456d0abab336674 : SUCCESS in 1h 05m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88d53c257e4c4dfdaa92d393218c352e : SUCCESS in 1h 47m 55s","accounts_in_message":[],"_revision_number":17},{"id":"8a12cda9e2b2ba3dfd11e13bb77c38649cda95fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-27 03:25:49.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"6b003f64f3798fc3692747b382247512827cb507","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-27 05:06:30.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"c0aeb144f491391850db94889e5aa7a169cca641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 05:17:40.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccfcf4daa4a3441eb6ebbdda9c9136d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f665dcee2440c4999a17402e46779d : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b0b93204f434939b15a11bf5735575e : SUCCESS in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e473ee41a624c1095276b4c7516925a : SUCCESS in 2m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21abac7427214d5fa9aee1454e6579b2 : SUCCESS in 5m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/837c39de3fce48fc964fd4393b35690b : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f762676465a543379aec892186a6ac4d : SUCCESS in 4m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e54002bf150d48ff906b5c309ad8a5b3 : SUCCESS in 6m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/93a685d424e64456b72b716893c52410 : SUCCESS in 3m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/16c98a8e6080468481ab1f7bbb177237 : SUCCESS in 12m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/9b5367cccda646f9b7da7d8c2eb76e66 : SUCCESS in 12m 48s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2243ac1de46b4cee938460a84aa56298 : SUCCESS in 5m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e1b0e5e3d40f4ffb800515a9066b41d0 : SUCCESS in 4m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2ca0cf7ef2264dc7828e4abc9ccd6613 : SUCCESS in 4m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c6be1597384b40e990740dac94a2a704 : SUCCESS in 11m 38s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/aa84fd415bde4670b9132ada0f7eb6c0 : SUCCESS in 6m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c89a5741dd824262a65c3e3afa6da085 : SUCCESS in 5m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4c5f472e266c41c9a06e2b7826a81f8b : SUCCESS in 3m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f918f47996744ba990e72191881cc98d : SUCCESS in 18m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1374fc4992a54833bd4a16eb7c025a36 : SUCCESS in 7m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/eaad0bbffcd2456aa6d1be7bfbf922dc : SUCCESS in 9m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2d152df5d99347cb94155c70b6b75d1d : SUCCESS in 10m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f6b091cd4167485fbcb560f35bf9466f : SUCCESS in 3m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9bfc91967c3c4d68a3de44d57abc2c4f : SUCCESS in 10m 46s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/623eb4543ae34bcb9343ffbc6ff59284 : SUCCESS in 4m 45s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/24d8f571624a4aa985f84b9635a2f833 : SUCCESS in 5m 08s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4d4dc77392d34e5a874241cd2207fdd0 : SUCCESS in 9m 57s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b2d2231756cf40a0bb0aaeef59f8088d : SUCCESS in 18m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f086f740b6044413b5a378d2b9b918d3 : SUCCESS in 7m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/195c08f57e7944129eccab2850353f9f : SUCCESS in 8m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/50007c0b83f64262909846ac011e2cf9 : SUCCESS in 3m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e0dbca23091148bcbd42ee98f310e497 : SUCCESS in 2m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4bbf30534c724012bd8d309e0761aa3d : SUCCESS in 22m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fcd58d37e69f43b5842b450fa327216f : SUCCESS in 7m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b8863a22018f486c98d7bd826751b4b0 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f639f68da20d4f54b353ba30b8d60014 : SUCCESS in 6m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3e0553443d4d42188c451ec6366f8881 : SUCCESS in 12m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7fe4adb52ff041dea987f7fbc488bf74 : SUCCESS in 7m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5981bfd58cfc4fd1ac749200cd97b728 : SUCCESS in 59m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d74a98dd28b4624a5c467f9eef58bf3 : SUCCESS in 1h 45m 58s","accounts_in_message":[],"_revision_number":18},{"id":"822d07fc55bf712481af5dd12b22ddd1b49e9ce1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 05:18:15.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"f250492f50b09e7299f68fb913bfd0d13ba4dba9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 06:30:14.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f36901a5f76e4b14bac014ea44c50166\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e54e1b7eccf949069c9065300e7c7c1a : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcaf1c3b8ba44d83a5e8c9c14a9e375a : SUCCESS in 2m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb93266aeede445184649b6617a3211b : SUCCESS in 2m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/edf4a68ce7c94bf68b046d628a4ff768 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c739acb9f114abc839b2f5b564b6969 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f61ff3692824157a796e598aeef5c89 : SUCCESS in 3m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/463d5c0f736a4c79b9345b680c76cb46 : SUCCESS in 3m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6088aad5edc3410b9d2f4c37e3e82012 : SUCCESS in 7m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/66eec75005254ff09629f9300cd29aa9 : SUCCESS in 7m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2adb596cd40940bbae570850e9a9df34 : SUCCESS in 7m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4585407bbdfd41fab5620566c8b35e1d : SUCCESS in 5m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/19ea1d29b1c4449b9317720ac4dd6c9b : SUCCESS in 5m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2e78a3a40f8a41d8b53623a099f0bc26 : SUCCESS in 3m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9d5693e6aac845b0b17a499deba995e5 : SUCCESS in 15m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b8b876d50bd4961852d07f904920afb : SUCCESS in 6m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/18a2dd33b7b64bdd82c0a2075239766d : SUCCESS in 9m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a17d63bd516f4599bcff77b6f549ee40 : SUCCESS in 13m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/52ccc4dac83347baa71828e3d8a04d16 : SUCCESS in 3m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9d377c7bae8d4aeab33ba6641522271c : SUCCESS in 10m 06s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/365a82863cf549f9a8c05097da8fd0c4 : SUCCESS in 22m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b6a2895c08c947e88f4452b2e03569a5 : SUCCESS in 9m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/54732fb89d49488da0f5b2ba955c3b73 : SUCCESS in 13m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d1a330986acd44f7b69de26fbc14c725 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3f95b2fd4d5e41999e59e8c2d3084fd2 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f96de73a3b004704941166d2c05e1ca0 : SUCCESS in 26m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/573154a81f40461d8874ea23c1b566f3 : SUCCESS in 3m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2bc1024e154d42f4873e0b0cdfa02f8d : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3e3b517c30645fb98e95b58d67e468b : SUCCESS in 6m 04s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/402c7228d1dd44408e3d90674f3895f0 : SUCCESS in 6m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/79a67cef66604aa3b31efef51e1cc4ba : SUCCESS in 5m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ca9945d79814db0868b2529a02ed74b : SUCCESS in 31m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf0a052c5d8c4ea6a7b250a77e6888ba : SUCCESS in 1h 11m 34s","accounts_in_message":[],"_revision_number":18},{"id":"163180a8414dad6ef564e12bf351a33def7cdcdc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 06:30:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"0d21a30d4d3344150377c9089a709f7408a1cd58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 06:31:00.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0aa81c871dc44d4187cbe9adeededecb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c523131823472282fef16a8ddca17e : SUCCESS in 33s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"6759ad0f43ed1cdd3e30e01c98ba756ac4589559","revisions":{"623c90e492c2710d557991f1bbe1a5d245312924":{"kind":"REWORK","_number":1,"created":"2025-08-16 08:37:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/957587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/1"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-16 08:37:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-16 08:37:19.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/623c90e492c2710d557991f1bbe1a5d245312924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/623c90e492c2710d557991f1bbe1a5d245312924"}]},"branch":"refs/heads/master"},"428e6f42c9227f9296ced32f3d9fd8a2939de42b":{"kind":"REWORK","_number":2,"created":"2025-08-23 08:17:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/957587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/2"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-23 08:17:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-23 08:17:55.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/428e6f42c9227f9296ced32f3d9fd8a2939de42b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/428e6f42c9227f9296ced32f3d9fd8a2939de42b"}]},"branch":"refs/heads/master"},"099f7c8f6ce26889b7b21e38c360aed1ee100846":{"kind":"REWORK","_number":3,"created":"2025-08-30 08:16:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/957587/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/3"}}},"commit":{"parents":[{"commit":"56798df8bac5995e09a8bf3ce8820e0d4f8710fe","subject":"update constraint for futurist to new release 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56798df8bac5995e09a8bf3ce8820e0d4f8710fe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-30 08:16:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-30 08:16:44.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/099f7c8f6ce26889b7b21e38c360aed1ee100846"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/099f7c8f6ce26889b7b21e38c360aed1ee100846"}]},"branch":"refs/heads/master"},"301e230d501e956d2ba8632605f0eacfd130cebb":{"kind":"REWORK","_number":4,"created":"2025-09-06 08:23:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/957587/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/4"}}},"commit":{"parents":[{"commit":"46a7e3927152884c05f1e257c528af593d67b84d","subject":"Merge \"update constraint for python-ironicclient to new release 5.13.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46a7e3927152884c05f1e257c528af593d67b84d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-06 08:23:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-06 08:23:11.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/301e230d501e956d2ba8632605f0eacfd130cebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/301e230d501e956d2ba8632605f0eacfd130cebb"}]},"branch":"refs/heads/master"},"9952983d7bd531ca34643aa6372927c343edc1c1":{"kind":"REWORK","_number":5,"created":"2025-09-08 11:41:39.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/87/957587/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/5"}}},"commit":{"parents":[{"commit":"46a7e3927152884c05f1e257c528af593d67b84d","subject":"Merge \"update constraint for python-ironicclient to new release 5.13.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46a7e3927152884c05f1e257c528af593d67b84d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-06 08:23:11.000000000","tz":0},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-09-08 11:41:29.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9952983d7bd531ca34643aa6372927c343edc1c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9952983d7bd531ca34643aa6372927c343edc1c1"}]},"branch":"refs/heads/master"},"e2b389b729259e3d0556299914a83d526da0c570":{"kind":"REWORK","_number":6,"created":"2025-09-13 08:18:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/957587/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/6"}}},"commit":{"parents":[{"commit":"3741dcc257f6a481e719a4428fa29ec2ec5a4988","subject":"Merge \"update constraint for python-zaqarclient to new release 4.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3741dcc257f6a481e719a4428fa29ec2ec5a4988"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-13 08:18:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-13 08:18:07.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2b389b729259e3d0556299914a83d526da0c570"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2b389b729259e3d0556299914a83d526da0c570"}]},"branch":"refs/heads/master"},"450b1ff76172db088f9172bee0b00dcc85dc52af":{"kind":"REWORK","_number":7,"created":"2025-09-20 08:30:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/957587/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/7"}}},"commit":{"parents":[{"commit":"361dc703c6d20f1b78c141800c8f2c7ea79715a3","subject":"update constraint for horizon to new release 25.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/361dc703c6d20f1b78c141800c8f2c7ea79715a3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-20 08:30:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-20 08:30:55.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/450b1ff76172db088f9172bee0b00dcc85dc52af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/450b1ff76172db088f9172bee0b00dcc85dc52af"}]},"branch":"refs/heads/master"},"0eeea7d21f0c56559ee93d8b3d65ba85f7194269":{"kind":"REWORK","_number":8,"created":"2025-09-27 08:20:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/957587/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/8"}}},"commit":{"parents":[{"commit":"af5fa8a98f1f093d01d87f8e57579e14e7ca5f61","subject":"Merge \"Deprecate ironic-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af5fa8a98f1f093d01d87f8e57579e14e7ca5f61"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-27 08:20:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-27 08:20:10.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0eeea7d21f0c56559ee93d8b3d65ba85f7194269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0eeea7d21f0c56559ee93d8b3d65ba85f7194269"}]},"branch":"refs/heads/master"},"edc2985f9245ee196801c23d96a303f7af0faac2":{"kind":"REWORK","_number":9,"created":"2025-10-04 11:05:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/957587/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/9"}}},"commit":{"parents":[{"commit":"af5fa8a98f1f093d01d87f8e57579e14e7ca5f61","subject":"Merge \"Deprecate ironic-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af5fa8a98f1f093d01d87f8e57579e14e7ca5f61"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-04 11:05:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-04 11:05:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/edc2985f9245ee196801c23d96a303f7af0faac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/edc2985f9245ee196801c23d96a303f7af0faac2"}]},"branch":"refs/heads/master"},"59c2b0a16202534817eed4259367cf757289c41d":{"kind":"REWORK","_number":10,"created":"2025-10-11 08:25:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/957587/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/10"}}},"commit":{"parents":[{"commit":"37cac3c01ae91e02996dae57d65703f806212249","subject":"update constraint for neutron-lib to new release 3.22.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37cac3c01ae91e02996dae57d65703f806212249"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-11 08:25:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-11 08:25:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59c2b0a16202534817eed4259367cf757289c41d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59c2b0a16202534817eed4259367cf757289c41d"}]},"branch":"refs/heads/master"},"583fc95aa38e9a3f9643c32f2a553bc186696526":{"kind":"REWORK","_number":11,"created":"2025-10-18 08:21:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/957587/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/11"}}},"commit":{"parents":[{"commit":"01e147e1627debe7f4020132732d7ae733d1f3b4","subject":"Merge \"Add lark to global-requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01e147e1627debe7f4020132732d7ae733d1f3b4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-18 08:21:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-18 08:21:44.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/583fc95aa38e9a3f9643c32f2a553bc186696526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/583fc95aa38e9a3f9643c32f2a553bc186696526"}]},"branch":"refs/heads/master"},"b3aee6ea5fe17a8db8427ea1ba7841f238b339f5":{"kind":"REWORK","_number":12,"created":"2025-10-23 09:54:00.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/87/957587/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/12"}}},"commit":{"parents":[{"commit":"01e147e1627debe7f4020132732d7ae733d1f3b4","subject":"Merge \"Add lark to global-requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01e147e1627debe7f4020132732d7ae733d1f3b4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-18 08:21:44.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 09:53:49.000000000","tz":660},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3aee6ea5fe17a8db8427ea1ba7841f238b339f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3aee6ea5fe17a8db8427ea1ba7841f238b339f5"}]},"branch":"refs/heads/master"},"40dac68b6c26a04535a4cdc5572861e17005c0d8":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-10-23 10:37:07.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/87/957587/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/13"}}},"commit":{"parents":[{"commit":"fdff4d7149e263fc866c36c702268b36ada62cc2","subject":"Merge \"Speed up pyN-check-uc jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fdff4d7149e263fc866c36c702268b36ada62cc2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-18 08:21:44.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 09:53:59.000000000","tz":660},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40dac68b6c26a04535a4cdc5572861e17005c0d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40dac68b6c26a04535a4cdc5572861e17005c0d8"}]},"branch":"refs/heads/master"},"e6b8663d53b8c5eb3087277a4f89e6d78e7804ac":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":14,"created":"2025-10-23 21:31:09.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/87/957587/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/14"}}},"commit":{"parents":[{"commit":"e767f28c26872e7147f18b3c0b4d3ddbf7b6eda7","subject":"Skip comments, empty lines when parsing reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e767f28c26872e7147f18b3c0b4d3ddbf7b6eda7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-18 08:21:44.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-23 21:29:05.000000000","tz":660},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/964724\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6b8663d53b8c5eb3087277a4f89e6d78e7804ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6b8663d53b8c5eb3087277a4f89e6d78e7804ac"}]},"branch":"refs/heads/master"},"6918edd52478cf64ff566639f31770a7c467946f":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-10-24 00:28:35.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/87/957587/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/15"}}},"commit":{"parents":[{"commit":"da5964bc14d19bf92678de71bddc745eb84e9fb7","subject":"[check-uc*] Create essentially a \"no-op\" dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da5964bc14d19bf92678de71bddc745eb84e9fb7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-18 08:21:44.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-24 00:28:11.000000000","tz":660},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/964724\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6918edd52478cf64ff566639f31770a7c467946f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6918edd52478cf64ff566639f31770a7c467946f"}]},"branch":"refs/heads/master"},"bb61e2844b141f27e7cc11b5c67cdb393e0abfe5":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-10-24 00:49:02.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/87/957587/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/16"}}},"commit":{"parents":[{"commit":"6b3dda33d305df109b9cb50e7c0a9de6bd6dd567","subject":"[check-uc*] Create essentially a \"no-op\" dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b3dda33d305df109b9cb50e7c0a9de6bd6dd567"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-18 08:21:44.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-24 00:48:44.000000000","tz":660},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/964724\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb61e2844b141f27e7cc11b5c67cdb393e0abfe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb61e2844b141f27e7cc11b5c67cdb393e0abfe5"}]},"branch":"refs/heads/master"},"039cd329b0b23b48ae0b160eb51becb3bbce9916":{"kind":"REWORK","_number":17,"created":"2025-10-25 08:15:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/957587/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/17"}}},"commit":{"parents":[{"commit":"e767f28c26872e7147f18b3c0b4d3ddbf7b6eda7","subject":"Skip comments, empty lines when parsing reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e767f28c26872e7147f18b3c0b4d3ddbf7b6eda7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-25 08:15:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-25 08:15:31.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/039cd329b0b23b48ae0b160eb51becb3bbce9916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/039cd329b0b23b48ae0b160eb51becb3bbce9916"}]},"branch":"refs/heads/master"},"6759ad0f43ed1cdd3e30e01c98ba756ac4589559":{"kind":"REWORK","_number":18,"created":"2025-10-27 03:25:49.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/87/957587/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/957587/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/957587/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/957587/18"}}},"commit":{"parents":[{"commit":"756a1a94446d58ba06c907c1ec578182d0be8943","subject":"Pin boto3 and botocore to 1.35.99","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/756a1a94446d58ba06c907c1ec578182d0be8943"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-25 08:15:31.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-10-27 03:25:02.000000000","tz":660},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- gabbi:\n    placement: \"AttributeError: \u0027NoneType\u0027 object has no attribute \u0027api\u0027\"\n- infoblox-client:\n    designate: |-\n      [designate.plugin] Loaded plugin backend:infoblox\n      [infoblox_client.connector] Cloud WAPI version detected: 2.10\n      [infoblox_client.connector] Configuring session\n      [urllib3.util.retry] Converted retries value: 3 -\u003e Retry(total\u003d3, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)\n      [infoblox_client.connector] Authenticating with username and password.\n      [infoblox_client.connector] SSL verification is False\n      [infoblox_client.connector] Connector initialized with options: {\u0027host\u0027: \u0027192.0.2.1\u0027, \u0027username\u0027: \u0027****\u0027, \u0027password\u0027: \u0027****\u0027, \u0027http_pool_connections\u0027: None, \u0027http_pool_maxsize\u0027: None, \u0027wapi_version\u0027: \u00272.10\u0027, \u0027ssl_verify\u0027: None, \u0027cert\u0027: \u0027****\u0027, \u0027key\u0027: \u0027****\u0027}\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n- boto3, botocore, s3transfer:\n    glance:\n    - https://review.opendev.org/c/openstack/glance-specs/+/962248\n\nChange-Id: I35412c12870fc6292636a0308dce55a5951e3737\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6759ad0f43ed1cdd3e30e01c98ba756ac4589559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6759ad0f43ed1cdd3e30e01c98ba756ac4589559"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
