)]}'
{"id":"openstack%2Frequirements~957657","triplet_id":"openstack%2Frequirements~master~I2e9cbb337bdede3624329b749c073f78f9cf17f1","project":"openstack/requirements","branch":"master","topic":"bug/2109396","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2025-08-27 17:57:09.000000000","reason":"Change was submitted"},"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-08-26 14:20:47.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-08-26 06:16:35.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-08-27 17:57:09.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-08-27 11:57:36.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-08-27 17:57:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e9cbb337bdede3624329b749c073f78f9cf17f1","subject":"Bump pyroute2 to 0.9.4","status":"MERGED","created":"2025-08-18 09:25:06.000000000","updated":"2025-08-27 17:59:45.000000000","submitted":"2025-08-27 17:57:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957657-bug/2109396","meta_rev_id":"4b837cb509ca415ae4118b21bbfc299fc755c9bb","_number":957657,"virtual_id_number":957657,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-27 17:57:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-08-26 06:16:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-08-26 14:20:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-08-27 12:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-08-26 07:26:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-08-27 14:17:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 11:22:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-25 14:36:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-08-26 07:26:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-08-26 08:50:57.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-08-26 08:50:57.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-08-27 12:33:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-08-27 12:33:35.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2025-08-27 12:33:35.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"8f407eb4e3c36e2adaec4af246f304a44366c704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-18 09:25:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f833b36353595782c0113764cb514e18bdc3b9f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 11:22:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e4477175534051b378796d63e104c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b4806b13d2540a4a08ae77379e19ad2 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b04055830f3341e599a3de4e8d43f567 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0649b9f7af34156a76e3290853129fe : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd7ef060581247a1a75bddacac59cd81 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1767fe3d63024e96853b4547b117a680 : SUCCESS in 7m 12s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/57d48fb13ae94b7ba420584b266faa1f : SUCCESS in 1h 52m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c7be6faef0aa41ec9ece49859b61ba3b : SUCCESS in 1h 56m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cb4f19fc55ea7a7137ae2dd7e4c232912a617760","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-18 13:00:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0012142d5810173f0594c242af7af1c8a7f2c675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 16:07:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e689b78355ce4575b2480b8ca8b77318\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45fe2408cdb4f93acad357e9620411d : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e595b09fb3a3471f97c6f7d180783eb2 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6acf5f961604c53b9838c44010af91b : SUCCESS in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79789a8388274523bb4e76a2d020c19c : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7749687e96a460d9e2f7231eb60d45d : SUCCESS in 5m 48s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/293ec1236ed74012821bebfe0f5dda07 : SUCCESS in 58m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/176053b1ad6c4641a34eb45ed75178ab : SUCCESS in 1h 37m 49s","accounts_in_message":[],"_revision_number":1},{"id":"44f64c4ae5d71dafa635427a2b50da14b4e451a3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-19 06:35:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc93f12097e03a47974aee7c8baf7a9893d69733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 07:50:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1b21ff52be0436f8b02117d794f5d92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f4b8077ded04c7e8b608e0ac3e50e15 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7930f3b6fdb458ebbf2ccc6ded8326c : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d310a24d0ef4777a201080293bda104 : SUCCESS in 2m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d26ade3e27124e67889482e4d7d7024d : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f42882465f24666a6782dd7187e5303 : SUCCESS in 4m 33s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0af0c7b937d44615b5682a8b717dd292 : SUCCESS in 1h 13m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c8cb16990ea5432cb0f18233dde61838 : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"246a631f5bbd3af02abb4041c290552b01169d4c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-25 10:13:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e35abfb72babaadf892162d8a1474903d41a87ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 11:36:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a2d98c228a54188aea095908a47e8e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180e65b6d74f413e9bb82c872a2dc678 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa7d9bc2710c4730a5e46983aa629b60 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/838609706fcf4c08aaa4b913f59e5db4 : SUCCESS in 2m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6b6a45e3afaf4034b1c643e161054f29 : SUCCESS in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59ad85463fb145558d22a03bef72adfb : SUCCESS in 7m 21s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/09a9e0a647984a338ad3cd3773637f25 : SUCCESS in 1h 21m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9d86f85fda2d4cf5b022a04eb7bb0617 : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c6fc3595e0f751341c95d8b8f286a6397baa1f9f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-25 13:09:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f86d32bc489b4bea478e59bead64495f5979fb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-25 13:20:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9a53680e4094c6bd03e9026efdbd07e5f4663f21","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-08-25 14:36:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"03d8bd9d2d13478177930bfde19a603678871ede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 15:36:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66e0339e23114a8690a7377032750aee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa41fd11c6442c6b616623c4da3a87b : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/041eaab999c24dc3bf570f0f800aae7e : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/502906554bba4b048fcc75d64f87eae2 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/046f1052f6d245aa84325fec57e7dc7d : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb450d74fb964aa3a3e1581aa124d9b6 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f7e00f871094b9093f70e9897daaad9 : FAILURE in 2m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/291a95499fcf4414a1f49cd3c2c7ceff : SUCCESS in 3m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/895bfc94a7b043839a6f3e3f04ccd94e : SUCCESS in 9m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c5026188db474b6ca0f553137b557045 : SUCCESS in 16m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/47ff1d24b58f493f955017e4f81ac8f8 : SUCCESS in 10m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/239119db6fb242b2be8ae42945fb01be : SUCCESS in 21m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e5567b9e4ba2498899ebe70f3a6c00e6 : SUCCESS in 5m 32s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7201ba7fc82e4037ae882a4d3f5bbf2d : SUCCESS in 5m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4a784207aa89462da67d0135240efe79 : SUCCESS in 5m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/be114ea0accf4f4b998933936fe08306 : SUCCESS in 9m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/cd61ab9d9d894cc98d398df7d85929a6 : SUCCESS in 6m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/eeed89a452e34533ae28347ecff92d49 : SUCCESS in 7m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/682a5981fe7945d2b3f55c17fc4007cf : SUCCESS in 4m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3ab9337f3c4347e29908e70d64576011 : SUCCESS in 16m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/463bdaa83d9b4edca2633d229af5a56e : SUCCESS in 6m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9fd1adbae7f94a1c97cf9a3e7b720f8b : SUCCESS in 7m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1912d7d04ac547ef9132105590585cf9 : SUCCESS in 10m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/449edb9c347c4814b7bf501da84fd4af : SUCCESS in 2m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/66ba270e8c5d4797bba7caa651299a4f : SUCCESS in 10m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/1dfcc97db21441c59bac55029e87bab6 : SUCCESS in 5m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7adc75c51b154ede969b1808a4db3a98 : SUCCESS in 4m 41s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/1697f1811e464de992b65478bb09c8da : SUCCESS in 12m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/92e1f61b29a74df2a38bfe4e5b15071c : SUCCESS in 21m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7d75a1da531446be9a52a26edd122429 : SUCCESS in 7m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/962535a4f82d4db3970c57b7965f7de5 : SUCCESS in 12m 34s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/73579c2a2be24a5ebcc73cdb8314ec0f : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d59c23ac2443498abaca7aa7032bd4c0 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19b6fc8983fd4be295a36413dd0919a3 : SUCCESS in 25m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e02d8acdabde466db3d33d8598ed4f5f : SUCCESS in 6m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4216e3a9022b4f559baa474af5e950d8 : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/492bdcc8d0c242888aead9d0facfa462 : SUCCESS in 5m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3dc80007960342b09fead30725a55a76 : SUCCESS in 19m 59s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a49ebdec888149059c223d576b731f9d : SUCCESS in 5m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d148d104164c4ad0baa1cf13ad7a1bf3 : SUCCESS in 58m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fe89bca5168446e87089dfc0c4bfa0a : SUCCESS in 2h 10m 22s","accounts_in_message":[],"_revision_number":2},{"id":"cef70077fd18c72e9b443abc8de926eac4190e61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-25 16:53:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5009a63c91b7f0031bc480dce6671c0a6e6f67b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 18:06:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17729b1d9f4c4b5a9585d41dee4a06ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea17f781405c4f39b0242fda8b0ad2d8 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc0575cb14345f8aaea2b7a7c4231fc : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/254094edd47a4b79879f3d20ba6f7adc : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4670cac7c95847eabfffed7ffcb62bdc : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d7ed86038e046638a0953db3fff2991 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c839336b1ccc46c3a1911e85c2cde7af : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16f0b3d31db5422daf91059edeaa638c : SUCCESS in 2m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/934f1f61e29b43d4a974cf90a64acc27 : SUCCESS in 11m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ed4fd57fc8af4f01a0d89bae03e98421 : SUCCESS in 17m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c669db883dfc476aa76ac0ee3a2c3053 : SUCCESS in 18m 35s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f203211a8fc94c2898bce0015aa29dac : SUCCESS in 24m 33s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/723d7bb9df664d649a3208d59a43df87 : SUCCESS in 5m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/44590a7caf5d4b9cb5f8711ae11f13c2 : SUCCESS in 4m 21s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9d10add2b70b4581b1344a9f270bdddf : SUCCESS in 3m 18s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/25e8293fb457422494d2e2099cf5e099 : SUCCESS in 12m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4b4ecc384df6417c8ffbe6809f7f4d1c : SUCCESS in 3m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/aa2c60676e2e43b2b21f602a8fe5e823 : SUCCESS in 4m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/18e0f90d5b7446d5b0fcbd8e1bdd2b33 : SUCCESS in 4m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a4d0dbe7dcff45e286d58eb657888f01 : SUCCESS in 16m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97a21efa74dc4dd08ea3412b481b64b3 : SUCCESS in 7m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5aa6a7b5773b40fc9c35a0a90895a0fd : SUCCESS in 6m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e06a2929dda440b7b4a1be3985fe961d : SUCCESS in 7m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a0f35ed1563f4071a98f94db32a0539f : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2a27b923e4fd4f708ea92e6021d41dee : SUCCESS in 8m 32s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c382c21ed4114c06ace52f0cbd875638 : SUCCESS in 4m 45s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cccd137c0c764ff58097adec5b2f39b8 : SUCCESS in 6m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f9037a8c639347318af2788bd5d7a75f : SUCCESS in 12m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2d76615700a947bf8866e1bcb8a56245 : SUCCESS in 17m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2a7d7596f2c34fd5adff21456b808f7b : SUCCESS in 8m 23s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/72e7baf981c342208267d9132adbd068 : SUCCESS in 13m 51s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e80ac074602f45a3a2ff3aa526074a57 : SUCCESS in 3m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1882c1b5ec544d583e0bdfc99ec5591 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dfc5ddaedbd84e84ae206c181f73d970 : SUCCESS in 24m 46s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f28cb697e2a4447989389fd14a6d53a6 : SUCCESS in 8m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/eeffad226df04ed5a643c1d5cdfdd3ec : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61d107b26f4846e597e0652a229bfd5c : SUCCESS in 6m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/37961291066b4ffe88bee7e8637260f9 : SUCCESS in 8m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/db0fb9a6f72b4a1b988ff7a06c444d70 : SUCCESS in 7m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ca9a767649c44a0a532a5bbed69788d : SUCCESS in 1h 02m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc3ed6ca4b0d47078e30b8af6d019da7 : SUCCESS in 1h 11m 09s","accounts_in_message":[],"_revision_number":3},{"id":"2c8a0dfe8e035ad975597bd732a1f7a77b4cf7d8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-08-26 06:16:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f1d186891e4bfc4ab7a6eabb189b736fb8631239","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-08-26 07:26:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3af9320431f9a259ab8d7c31bb6e4c666e2d1b6d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-08-26 14:20:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2f69ef3f24c63342500f286be79232294bea4cee","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-27 12:33:12.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a4d98bf2a4829c35c0f06fa5f27bccf9f1a5c16","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-08-27 14:17:14.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4921621251ec3321658fbc32202ca65d35080f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 14:18:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"688af36daf2c393abbe1d070f00f8b39942d42c2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-27 14:49:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16858146463c3cde2eda916b706bee446a3d4c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 17:57:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9d51c6d0a9e4120aa29574b61804d2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/895f23573fc645a38eee7cba98683424 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83eb0146d5c644e484bdfa61d3d40609 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e42f7ed6a6174a45bf8f6ac0af9db05d : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/438f3b4dc3d8414f83df90731d2aed51 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ac70060ccae4978abc11a9106bc102f : SUCCESS in 2m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9bac87a5a6e641ccb08a93c07d78c084 : SUCCESS in 2m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5a8351bf2ddd426ba5cbdbf70bb0d28e : SUCCESS in 10m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d4e64da973e04dd299cb8b57608ac7a0 : SUCCESS in 8m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d64dcd2077e849f6853185a2ee5be362 : SUCCESS in 10m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3157a8ca9492400a9455fb2aa0ce30b0 : SUCCESS in 7m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/404ca1d45b4b4e5085dacbb84222bd48 : SUCCESS in 4m 28s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/051e24668805437baf1f1999d24adc51 : SUCCESS in 5m 52s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/40e69153d72c4649b7763a1121921f33 : SUCCESS in 5m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/46c401fcda38464c882a05b01ed140d1 : SUCCESS in 17m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e5bbb05b8bd4db5aabce6108a928ebf : SUCCESS in 5m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b5c767d868da4586b5068624cbd13fd8 : SUCCESS in 9m 00s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/81e74e58cc2b40c993334395660d478c : SUCCESS in 11m 31s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6b166a6b0b9d4430b975e83c8686e370 : SUCCESS in 17m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/05dbea57048746e6904cc0e536099b97 : SUCCESS in 8m 01s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6eeb1b6e75204f969d872a1f2615fe0a : SUCCESS in 14m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9813defb58274a2bb6a47112989b0529 : SUCCESS in 8m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/bfd38565e59349d69aa8195e3999e04d : SUCCESS in 15m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ab30d6a9a2974661b70cba16733d4e6a : SUCCESS in 5m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8a7cb460ea448e99ec9bed22acee432 : SUCCESS in 2m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8283b7ed007648f4bdbfa316f17dd5cf : SUCCESS in 28m 09s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5cc58b5aae4e48669fefce8754bf52bf : SUCCESS in 14m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/11f94b7692234300a8bdfc92044d6d45 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b588358fd9747d997e82995b6e1f7cc : SUCCESS in 7m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/08c4ea1308cd4f5aa141730f1d176a91 : SUCCESS in 8m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/912e80bc79454d29bb29089e44170ea6 : SUCCESS in 5m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a356d3f28e0f400787b5450b82c7d108 : SUCCESS in 55m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab93ccc73dd64c31a2da4143197155b1 : SUCCESS in 1h 37m 27s","accounts_in_message":[],"_revision_number":3},{"id":"3515e84b6ff50127595b60fe83d8daa19f4142f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 17:57:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4b837cb509ca415ae4118b21bbfc299fc755c9bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 17:59:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0032b6a1570e4c8a81d64bec94f91b2e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785cf8db2c7d405a91eb97c63f29bc93 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"edaec0fa2d0c777f75c890096b24aea303b38bff","revisions":{"62b53ae9fb34d5895f09a2bb8ee3c9c1ab8fdc21":{"kind":"REWORK","_number":1,"created":"2025-08-18 09:25:06.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/57/957657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/957657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/957657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/957657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/957657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/957657/1"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-08-18 09:22:17.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-08-18 09:24:58.000000000","tz":120},"subject":"DNM: check if pyroute2 0.9.4 resolved bug/2109396","message":"DNM: check if pyroute2 0.9.4 resolved bug/2109396\n\nRelated-Bug: #2109396\n\nChange-Id: I2e9cbb337bdede3624329b749c073f78f9cf17f1\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62b53ae9fb34d5895f09a2bb8ee3c9c1ab8fdc21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62b53ae9fb34d5895f09a2bb8ee3c9c1ab8fdc21"}]},"branch":"refs/heads/master"},"e6ff7ad82098b1b100c7e7b1364bda354103f8a0":{"kind":"REWORK","_number":2,"created":"2025-08-25 13:20:51.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/57/957657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/957657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/957657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/957657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/957657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/957657/2"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-08-18 09:22:17.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-08-25 13:18:15.000000000","tz":120},"subject":"Bump pyroute2 to 0.9.4","message":"Bump pyroute2 to 0.9.4\n\nThe bug we seen in 0.9.1. is resolved in 0.9.4\n\nCloses-Bug: #2109396\n\nChange-Id: I2e9cbb337bdede3624329b749c073f78f9cf17f1\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6ff7ad82098b1b100c7e7b1364bda354103f8a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6ff7ad82098b1b100c7e7b1364bda354103f8a0"}]},"branch":"refs/heads/master"},"edaec0fa2d0c777f75c890096b24aea303b38bff":{"kind":"REWORK","_number":3,"created":"2025-08-25 16:53:43.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/57/957657/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/957657/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/957657/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/957657/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/957657/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/957657/3"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-08-18 09:22:17.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-08-25 16:53:37.000000000","tz":120},"subject":"Bump pyroute2 to 0.9.4","message":"Bump pyroute2 to 0.9.4\n\nThe bug we seen in 0.9.1. is resolved in 0.9.4\n\nCloses-Bug: #2109396\n\nChange-Id: I2e9cbb337bdede3624329b749c073f78f9cf17f1\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/edaec0fa2d0c777f75c890096b24aea303b38bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/edaec0fa2d0c777f75c890096b24aea303b38bff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
