)]}'
{"id":"openstack%2Frequirements~958191","triplet_id":"openstack%2Frequirements~master~I72b67be4104f3d5191b4941a80b2dd9434090f12","project":"openstack/requirements","branch":"master","topic":"master_cve_fix","attention_set":{"9649":{"account":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"last_update":"2025-12-17 15:09:04.000000000","reason":"A robot voted negatively on a label"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-01-07 15:50:42.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2025-12-17 15:08:16.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-01-07 15:50:42.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I72b67be4104f3d5191b4941a80b2dd9434090f12","subject":"python: cve: use latest cryptography and urllib3","status":"NEW","created":"2025-08-21 12:40:39.000000000","updated":"2026-01-07 15:50:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":16,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"726f90b9f84373622ae05e42fc9d9248e5e35438","_number":958191,"virtual_id_number":958191,"owner":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-05 19:47:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-21 13:11:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-08 08:04:56.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-09-11 13:14:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2025-11-03 10:40:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-01-07 12:41:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"e0441830f7f18e782e72fcf2be5cd8039bfcd08d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2025-08-21 12:40:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1dcb9b16cfa82fa14479a4556387a8b6470a1e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 13:11:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5af4f23b050443369845ab3a12897b76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bc8502ab3654069bd49dbac1bfd5ab5 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99ad2a665b4c47818c8fd9f9d18c2c03 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4bfa48d4e564970a0b7d419622a8870 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43a322703dcb49dd900c924fbb760820 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afcff26154084a07ac991782f5b0da2c : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e107f8773a5a4f2c8f726e5e68a264cb : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fa5ca3534742438d9dbb7f7d4496244b : SUCCESS in 4m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aa33b48a70db4e438a15244ea07fc37a : FAILURE in 2m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4e8e88adfbaf40919a39be4c56d1168b : FAILURE in 6m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4e3d0c014eda4e338073bffb136bcf4e : FAILURE in 11m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3dd0500271d34e659bad28c20233f10a : FAILURE in 17m 30s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e2fe5c6ec2c04b2c8706e22f8e748c5f : SUCCESS in 3m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9246924bd7bc4708ab2dbeead32bcb01 : SUCCESS in 4m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/45dd7de03f8043ce9bd07063e23428d3 : FAILURE in 4m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2a31713ebca84177bd48c976a67b35c1 : FAILURE in 16m 44s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2c4fcd9818ce4f36950682d81a578e05 : SUCCESS in 5m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/128476ccd9a14bd8b708aaf336bebc39 : SUCCESS in 6m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9dab8adb37df45acaec15bca5ce64f36 : FAILURE in 4m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/940bf3d5d90d4c8da18381f9f942ec24 : FAILURE in 2m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/127692e3e3ef4005a0bdea68c5bb21f1 : FAILURE in 8m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/44210112f1f8489bbc783905c7125505 : SUCCESS in 5m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0d68c7796dd342019d5f0f82dc539f0f : FAILURE in 5m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1fb1fccd3be343adb76cfd2466ccc7da : SUCCESS in 1m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e2dc3217691a4467882909b6bc68c0ed : FAILURE in 3m 46s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/37af917b73114f2c80653d668a48c2b1 : FAILURE in 4m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9122fe114ac349b598802245821c376e : SUCCESS in 4m 43s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/867eaa677ab34702b793b1746f3bda93 : SUCCESS in 13m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2ffc65e8a8d84909aaae1ef3c8a3bcc3 : FAILURE in 3m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f8a81516555f4e38967df8e056d9a80c : FAILURE in 4m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/325fe3e434fa444787dd5847ef5e5eae : FAILURE in 5m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b1e21f78d16e4be4a261b0fbcc5c5265 : FAILURE in 4m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b0aefdf10f7f424987cc6dfc54c401a8 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ab7a69b5302492eb83e9298964fd471 : FAILURE in 4m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/be39530c222d462a8cfed106f3e52b31 : FAILURE in 4m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/03e25b01e5b04c9f9124dfbc2a315a35 : FAILURE in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/213f7eb03ca64e898c7e83201b4c05c6 : SUCCESS in 5m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5357dd3e2360442d9171dfc00f1faefc : SUCCESS in 23m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8fa8191a1d114584afdc669c58f6a787 : FAILURE in 3m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46220771e1e94d85be62d4daf0125878 : POST_FAILURE in 14m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/693af5ffaf5340a08e26af1260b2acb9 : FAILURE in 11m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5513ff8c7d4f735b1ead45687a56116a7d73ba1b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-08 08:04:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9cb83a89d23cda98c5882c218b5c2737bb75fc67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-10 07:51:56.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"413dc213150c6207644cfd44bcb6d301c7714542","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-10 07:52:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd4c1884465115eb955affc3652a19a478a14fff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 08:18:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b4e3ba5c4ff470099b78c4c2905906e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5fe66616a86421e830aca30b49f2941 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a15d6e4adf85487294e34f6d6f6088d5 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/595ee2c765ae4a17848a92b702c9fc23 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4762b49a9854ec9ad0f70b55a45e01e : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7bb25b2f7334a109671d4af029208c7 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3447385453394bef8105e83b95aac509 : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b2a043e8543648619e45359b88100353 : SUCCESS in 3m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5d67608776e4416cba8d868b420518ab : FAILURE in 4m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/31246b056dae4a719d25b1e2d3b1fb40 : FAILURE in 7m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dd7175f26c924507b7522216cceb77b9 : FAILURE in 10m 50s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/37f5ecd31740453db7c679d237d580ad : FAILURE in 9m 42s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/315b962b24234c94afb4646a71e838cd : SUCCESS in 6m 52s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6156f40af2d44810916a5d67afc1ac3b : SUCCESS in 6m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8058c618546744f1830b59987691d521 : SUCCESS in 4m 36s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b5bddedb08134a52a1d93943c613d414 : SUCCESS in 10m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/6a1951b55af54f40a29f9675ef982dee : SUCCESS in 7m 00s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/972ead0422f74505833d7211b223171a : SUCCESS in 3m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0c90c50957a443a4911747bedbc5c573 : SUCCESS in 8m 20s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/02267219d10e4c17a891ba13974603ab : SUCCESS in 18m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/64c704657ebc4f14a845e4beff93e222 : SUCCESS in 7m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1d8e1891e735423b809011da1bbfd91b : SUCCESS in 6m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f06218179ac54cb69915cdb363b910d0 : FAILURE in 3m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/45d5d7f303fd488f937f737edbe9d828 : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9371ee0378584360a003be2085b7d482 : SUCCESS in 9m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d9f3b673ec1a4408a22b5f9cd8b1a3ae : SUCCESS in 3m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8774b383577a4570949c07479189aef1 : SUCCESS in 3m 16s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/84ad736ba95344799be3da9ad078107b : SUCCESS in 9m 48s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/68e29925852445e59b5cf1ddfa9c57a9 : SUCCESS in 20m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/89a38e930f684bcaa5d906d963823300 : SUCCESS in 8m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7393696d82fe45e1b221624baf7824a4 : FAILURE in 2m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ae1b79f2cd8e4dbe980dfc796f6591b9 : FAILURE in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b5f5668cc6f7427595e761b0703caeec : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aaae67220c854ecbae50f2875dceb584 : FAILURE in 3m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2f961b3302ae49be961e68377b14b60b : FAILURE in 4m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6feef8cb5ea14280898a8fd59c140175 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3371e58b87664a5689b0e15bfe695621 : SUCCESS in 7m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b0febf6890724cf9b79117406c967714 : SUCCESS in 18m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e14402d38ef845d480e0b9e528f91ae0 : SUCCESS in 3m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6a45f21ed3a4fcda3a892a403a48526 : POST_FAILURE in 7m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f3e8affc2254e4994d1162f3f6c367d : FAILURE in 14m 45s","accounts_in_message":[],"_revision_number":2},{"id":"d2f60e5acdc9d2addf02609f59e3663d7fe23818","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-10 09:01:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca85dc054b1b0ed9af1f67604eed578aee709f9f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-10 10:26:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff5f1e46dd331a4d0248dc06f7891eeb0bf94dd2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-09-11 13:14:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"428dcc6eb2b0a6ac3d92ee5c8259495a9645d8b2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-12 07:17:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7848a79afc9a90d2f9a3fdee05c824d5a23d7e8b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-12 13:16:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec021f0fe337782d7d7ef85b6af87f55ddf4ee49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-12 13:17:34.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f144c808604137fcd16a1ac8124c9b4e1051227d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 13:57:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/258ea466f75e44b6b183a8b1d5180336\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5867c1b7734140bf9d78139c7c02ad : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da80cc7785ef4d51a623a96279ea5077 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46359a6a2bdb40f0b04eaadcba04d19a : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c8e679ba631448b84b7cc812368f370 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24795984efd34069aec797f82558149f : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a1c3edb299144e7b9193255446de6ba : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/29ad86313b064393a93e1701517cefc1 : SUCCESS in 3m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/66e27d91c857415da4a441e73a4506bb : FAILURE in 4m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/40c73959784b48ef9e6638bf1ae592de : FAILURE in 7m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/42ec23e7965a4dac8e6fbf829d011667 : FAILURE in 7m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/bdbfc8a8652f4e3e9a9f96fdcc863803 : FAILURE in 10m 25s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/61e30dc7c8444831a9359b7362ee9964 : SUCCESS in 5m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d370e66ccde74ef387bd219ae2eab020 : SUCCESS in 4m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/05e7a46a19cf48f9a1a30ebec5ce2d07 : SUCCESS in 4m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2a2ab10f4f024fd9a3ddc9f7fd1ec8f8 : SUCCESS in 9m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4beaef2eeb6c41b9aa679196287bd854 : SUCCESS in 5m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d9336d3699104b4cbaef80cf73dded81 : SUCCESS in 3m 59s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1f3dc88ebb454e8da1f0b851535195cd : SUCCESS in 8m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e32baf3386c747e78dbdf2d0cbaf41b6 : SUCCESS in 17m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da38d1c662fc4879a9ddfe30224ebd7f : SUCCESS in 7m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9578b092dc344428a0f539a27f90e75c : SUCCESS in 8m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1089a791406a4c09b23ddeb197ef130f : FAILURE in 4m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dd5324f7c00744f98693b76f041f336e : SUCCESS in 3m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e0048b3735194d3f88ae7e12f1b78674 : SUCCESS in 7m 12s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/afa7c18022e74264ba664c3663c331be : SUCCESS in 7m 10s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5985fd607c064f9d9f8009f1552c684e : SUCCESS in 6m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5e9b7e256aa24d6c85a6a499704c3c4d : SUCCESS in 9m 29s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0f809fdf7ccd468799cf6dff84880ecd : SUCCESS in 22m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c9e90e65a9c0413a94a02069edbafbd9 : SUCCESS in 7m 42s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6f36891379c04257a2321b0888a00966 : SUCCESS in 9m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7090adae24c24ab0a5ec4e57be2eaee9 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0bc8b54147b14355a2ce4953fe0cf9ce : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c482c366de2843bba480c5ab50efcda6 : SUCCESS in 27m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/45b01334c1f04bab8f48c50687aad9ab : FAILURE in 4m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8318f5bc138d460eb56b172ee7db77d9 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd524aaa0e0845348217ca8fda0b1f38 : SUCCESS in 6m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7ac9f6bc5a114c76a196acb6ee862d57 : SUCCESS in 14m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f78168f1ac4c42f28b11bb77b543a403 : SUCCESS in 5m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80d0ce1f537c46a7963b1be0f8f2fc1b : POST_FAILURE in 9m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b53357914044ff49770c8d14ed87ecd : FAILURE in 7m 47s","accounts_in_message":[],"_revision_number":3},{"id":"c929c3bbe0da338971094151eee6148a79b0001e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-10-06 09:05:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"65a152165b9647b94bf0eb850697d6988cb6a755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-10-06 09:05:20.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5ef4535fb2e985de7527cde5c5cdd01da0a3cb78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 09:31:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c68db3abff03410aae15cbcb12d87555\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ed3e859f0045b593f0bb7ae53f5c22 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2176b2afa664401b4ee6b0a2edccb9c : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27277cdad97d4076833696838be965b2 : SUCCESS in 2m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48518bb9f9bc4b94bd3d40682a4f027e : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/375ce3808b8d4a68a64fc19301636fc8 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a0f340577cc4986a91a35201a2f697f : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd2e6edbbeea4ac7802d79ae06d01513 : SUCCESS in 3m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a4a2c8e9f0a444838e7833753130fe02 : FAILURE in 4m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/317ee093e27d469bbe46d715c6e1b305 : FAILURE in 7m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8eefbfc029ae4da4bdfaa06cd20cae45 : FAILURE in 10m 11s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/374297fd9a6348d5964112b7930fcf10 : FAILURE in 11m 46s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/63b6e8ad5ab547c19a3e60ef718ee476 : SUCCESS in 2m 58s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/93693a81f5d74083bfe68518400ff3ad : SUCCESS in 5m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2fd133dfe6694e979e14c73f9c905d33 : SUCCESS in 4m 46s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2158eaebe5c147eaa77c8da7dcb1a159 : SUCCESS in 7m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7c8b6c8768ef4dd0bca6a75ae9187053 : SUCCESS in 3m 13s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/db776bace1954798b9ecaf1c9ec0b2ab : SUCCESS in 3m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/773a8a8fd8a149d1a7d14c484c3ce996 : SUCCESS in 3m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9197eedbe8a540a4b37dfa80fb04824f : SUCCESS in 7m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0423c8efc5664db8be702ffe4e287edf : SUCCESS in 6m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/314436ffd05a4ad4b804eeb19841e0cc : SUCCESS in 9m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c13b76c5185f4949adfc1470b3054b81 : FAILURE in 3m 44s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a511294a061843638c135837324c79ea : SUCCESS in 1m 57s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a3700d6319544a9bbd6e16296525025f : SUCCESS in 9m 36s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2befcc5a5a344f5e8c5b854f4242c0e1 : SUCCESS in 4m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a6b943c12cac4fbe8850d0500887aff0 : SUCCESS in 5m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/24453095d20341aead53a1b9887cb958 : SUCCESS in 10m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1184dff81cdd4f32a3f4483407648c6d : SUCCESS in 15m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a8e589cde3574925bbbb384d816522e8 : SUCCESS in 7m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4b9ce91148b9434e95bc0fdebe124576 : SUCCESS in 13m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3a1f3db3d6d44b6f9caba00a6d3bcc0e : SUCCESS in 3m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1f530a95a38a43c4b7f45424fb9fd3f2 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a9d4684863934578aef62a9b2791b82b : SUCCESS in 23m 29s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ea0d26bc1aec4d97bb3a4836142665a5 : FAILURE in 8m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/614c8b0eb726402e821dec7fd142ab40 : SUCCESS in 3m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/41d443cdaf4843589460157feb890ae0 : SUCCESS in 3m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/80bdd9c8e397460b8e886a2e828efc5a : SUCCESS in 18m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/45f7e6c5299743b3b0ecd919e115f6ac : SUCCESS in 6m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56906091c5cc45ee8a5601884c978bbd : POST_FAILURE in 12m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/609125a778e8498f849b0c5ccbc3e0a4 : FAILURE in 14m 17s","accounts_in_message":[],"_revision_number":4},{"id":"1361318d5e829a871d111c1b70950972c8821b26","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-03 10:28:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1e6a5cc05a6a8f7875b61151be29f14b53117d7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-03 10:40:01.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d346579725c37c0bfa5b1d4248618413a173040b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 11:20:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99e9444ae347425ca82f65b4bfd9cafc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b5c39646964a67b96a3b3e613e19f7 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/905b20a8438f4a69b798bcffdcf871a0 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d3f822a9e0c44b08768ee6609a52526 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2f35303de444b03aa185180168a08dc : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2540142413e455286204b292f80e996 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f816a5ce4c4642d9a8ddd4c5ae677939 : SUCCESS in 3m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/05c16283ef054b5584cea65b3bc9a15e : FAILURE in 5m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b3f9fb4e365b4ae8b60ffa48bd4351ad : FAILURE in 6m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f25c20137a4245c9a1de5047a6406933 : FAILURE in 9m 46s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/018df80faa234cc3ad0f9a6ac047d050 : FAILURE in 10m 32s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f308c25e40074fe0a198a1e1a449b2fe : SUCCESS in 5m 12s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3a7025fd5cb24219bcb0186f1e32a00b : SUCCESS in 5m 15s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7cfb827ec8ac407f99f372cc2426225b : SUCCESS in 4m 23s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/aaa40cbc748e43c0abd3505379cd6f34 : SUCCESS in 11m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/daa417567d81484d8429079dab7cd83e : SUCCESS in 5m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/39f1a154ed294a6389c727aeff09edb1 : SUCCESS in 5m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/565a14abb7584075887c0c5aa6f3a48a : SUCCESS in 7m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/973cf5b79419433199f0e7a7f2675aa5 : SUCCESS in 8m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e8508ea32bf427aba5455973bd5e9fd : SUCCESS in 4m 19s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4890fcbaf7ee4b4eaf67a38e4740e678 : SUCCESS in 8m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6fc29e91280246db871493f6511dc09f : FAILURE in 3m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5968827d2be8433c8089839030c65ad1 : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a49e39b2a8ff4ca88290f90c936e2dd4 : SUCCESS in 10m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fd35d6edffbb4d5c95add14eda39db37 : SUCCESS in 5m 19s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/fc0debe6ea8645258fa8af13a6f5d9f0 : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/64471f8895824ea383383f22edb54e71 : SUCCESS in 10m 35s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cc9ae4f282fd4b1299cd8e9eaa3225a1 : SUCCESS in 23m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fe68018db60d408389b9b040e692041d : SUCCESS in 7m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/31dfc8025edd4e57aae79c0d82508a76 : SUCCESS in 14m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/38047360a2cf4f039416464dfd159c26 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3c4ed13e0b9f459e89eeaf0963cebd27 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/36bcc40615cf440fa589b13f5b7d00a7 : FAILURE in 24m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6ed79916074744a08d0ec3abeb6b5703 : FAILURE in 5m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ac12cac159434fceb7e30fd3454f93c0 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad1aa3aa72764a30865f1d5b59e1b17f : SUCCESS in 6m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d92054fd2ac044e393b88928b257e8b9 : SUCCESS in 18m 59s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e1db242fbc974a1f82eefd821f8d3a3e : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2de422309af4bc69f4a079e8a70f241 : POST_FAILURE in 8m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb2576a2b21b4c6798619bab66452260 : FAILURE in 10m 19s","accounts_in_message":[],"_revision_number":5},{"id":"c3b7c17b556ec6f790373ce822753a49a46f8cd1","author":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"date":"2025-12-17 13:17:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1f66bfc6eedebc432a1b248a19c721ade28c0718","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-17 15:08:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5dda3690fb85987165a20841a87f59b16927cc78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 15:09:04.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 958191,5","accounts_in_message":[],"_revision_number":5},{"id":"ebe1b4cc4b33635df4d488f991576b67cd65a7dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 15:09:11.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 958191,5","accounts_in_message":[],"_revision_number":5},{"id":"1fa431452d11dc91ceaf79d4d64a3bb3c4edbbd1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-19 09:16:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2d318f2b20c1f7d2d222667f153796b92f250ae2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-05 10:03:14.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"be0d75277e6fc265ba854586b19edf9bfc43242b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-05 10:07:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9ba6e30e6ebcd3ac24865f0a75cdc2a5e8fc00b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 11:04:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fbeda50263d470384eaf8f70c197dc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520ada6052574d5a85128ea1ae1ef021 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/507a67868cb14b248d44005e99437def : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7695e44ee5814461bc64982976c18f6c : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6b9df811f0e4d84a7aef78363f8699c : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a462b879910a4c548e1ba97d4e6b5c47 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f196e81abddb44e6aa8fc6798dd078b2 : SUCCESS in 3m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/622becff79ed4073b3c6e1279ae74e6b : FAILURE in 3m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e45aeb9072f64e67a9ab7a0267fdc924 : FAILURE in 5m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5053c5d0dc234c85866e3b12f111d91a : FAILURE in 12m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8cfb9bf59bea4a49bf8edc83e338f30b : FAILURE in 10m 25s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/51f6e10409cf41de9a1bf2b5b93d2442 : SUCCESS in 6m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/eef7be91fdc84667a160c448f5c89c04 : SUCCESS in 3m 20s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9c45afdd39e446389400ee325922fda7 : SUCCESS in 4m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/65b0cad3efe94a3d88a1981f0b65254e : SUCCESS in 11m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d4083ba3e3284865a12d20f018d8c300 : SUCCESS in 4m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1c8dbbbfdee4404d9ad5a3f5e45eaaac : SUCCESS in 4m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7b512e3cfe2340cbb66c6c338e62bd3b : SUCCESS in 7m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/693119a9efc94956a561207f97e1b760 : SUCCESS in 7m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/889c3cc7560c4c99bd229899e0722650 : SUCCESS in 9m 20s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7b3396fc88ac42c3ac7648ec1400ebe3 : SUCCESS in 8m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6d615cf5f5ef4339bfd0a63aa35ee53f : FAILURE in 3m 31s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3857c9303d6f4e0b8f8d1564c527e1f3 : SUCCESS in 3m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f9573ab7f6a742f0aa42eaf80a28c9de : SUCCESS in 10m 08s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/83bd6abd9ed04cfcbee590f03950b57f : SUCCESS in 4m 21s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2d3417be72714b23b9e7580e937f9245 : SUCCESS in 4m 49s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5524ab18dddc4fc48528287b658318c0 : SUCCESS in 9m 28s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/18090cd181c8444d87b5de5d00ac2a21 : SUCCESS in 26m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0e79a1d1a8b14ec4b56518313bd7709b : SUCCESS in 7m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/bcb00da2caf446419f7c0f69f64f14d7 : SUCCESS in 13m 16s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3f9a5f3b6bd24986ac091e934a959234 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/82ec118352e845518fdb5cb4974f46a8 : SUCCESS in 1m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c369a6f48934b8984f2df4d561c7286 : SUCCESS in 25m 51s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/32c1c7eb327442b9886562fb420bc660 : FAILURE in 2m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/349433f669f4454f989f77ba6ce46a92 : SUCCESS in 3m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f18e9def85f4e73bb01745d753eb6c4 : SUCCESS in 7m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f3d54503ccf1451f93d47f63a3ced463 : SUCCESS in 7m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c2a3c66ce2534b3e91ec5b281de98090 : SUCCESS in 5m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c4f886febdf4c539a261aa63a156592 : POST_FAILURE in 6m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c047c3b9da9a43b59de43ed4e4b942d7 : FAILURE in 7m 32s","accounts_in_message":[],"_revision_number":6},{"id":"04486380e1f084c0705f81ed162c65fea17f0754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-05 19:02:59.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"c6204f3c6d882a0d4ba91a3a165c9d6f57aa23a9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-05 19:04:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"67180f03ac8c76621a7c721ecdc7e9d1369bd20d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 19:47:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3d573f3f7e2403bbd87955b1af98f5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d3007786dd4ffdb8b3115211571001 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ffc016abc54971b7ab030ba8f0c2ca : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c631dcf0fc7547998c1a03f76d31a9f4 : SUCCESS in 6m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57cfae69af04437d813708adf6fc92d5 : SUCCESS in 3m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e54ccb086cec485994df244ce5a3649d : FAILURE in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55c8e0d7d9c74518990c7a40f6b19def : SUCCESS in 2m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/75e0539c74bb4bc094ee8d360b603862 : FAILURE in 8m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1e304d9a203e4e6f9258af687e5c59b0 : FAILURE in 6m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bfb6be34fd9b4b839dd914c9a5f2f3f7 : FAILURE in 3m 42s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b48454ceee1f4ade8d2fc9e546661f19 : FAILURE in 9m 50s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4f4c8c0d87924d7090da37a60eb042eb : SUCCESS in 5m 25s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/60b0d07eb73b43a8b707b8022b6c2e0e : SUCCESS in 5m 23s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9388c4fc5e9e430a8946cb6fefd210c5 : SUCCESS in 4m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f99326332a43472c892acc085ccb158a : SUCCESS in 35m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8291a8d7b5f54ad6a7b09cd14c313ad5 : SUCCESS in 4m 53s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/68131e90a9c94771b4b752e3edb3a637 : SUCCESS in 5m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/759aeb849fb74a0baef22c078a36ef31 : SUCCESS in 7m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7efe0df404eb4546bc4ffb73c4c2cf29 : SUCCESS in 17m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/236c700451a34ef1adf7362422ca48e8 : SUCCESS in 6m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/05e38af0db644eb3b0d4e94ed71061d8 : SUCCESS in 8m 35s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/eef0bc7ae8e5448cab423cfd026f65bb : FAILURE in 3m 42s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0bcdf75c67c849d48d7f766d44b7a833 : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/40ba72dba73d46b0926dd1e0e922f803 : SUCCESS in 5m 37s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/5bd9dc95b5d4493ba0530f634c2f723d : SUCCESS in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/48955be5cc6e4ad68f322343933a1fbd : SUCCESS in 2m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8a2e7db47e4248c18aa49bf34cd4b05c : SUCCESS in 12m 01s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/39c8f0d302bd4cddab30f1397266177b : SUCCESS in 28m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b9cbe83b185c4567acb6b0464cdd64fa : SUCCESS in 4m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8f6d80e859764108ab2def3ac5f5ab8b : SUCCESS in 8m 37s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7660979df33a4301af79381b8ba5ac04 : SUCCESS in 3m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/25f01b81578343e888a7868e6086c911 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/461109d86b694a6bb0742f14ef90654b : SUCCESS in 20m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9320d0e4f53545519054861f3c84c7ca : FAILURE in 5m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c743e2077a98475394eb1af24530f96a : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7816447c1a3e4b4ba7bb3b727ed493f0 : SUCCESS in 6m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c263db9d0f054709b2e55ce3e4892dda : SUCCESS in 12m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b394fb48bbb142b79b0c483fb6a231c7 : SUCCESS in 7m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/feb53dbe897d4c9b986f5af4d1fccdcc : POST_FAILURE in 6m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49e3a286936549cdb72941878872eb4d : FAILURE in 7m 00s","accounts_in_message":[],"_revision_number":7},{"id":"8883c62f580a1b81185b1f04e63327fa63c22306","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-07 12:41:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"726f90b9f84373622ae05e42fc9d9248e5e35438","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-07 15:50:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"277cbe320a9b12309d4e6bde595e66aca71b05a1","revisions":{"6b1f4edc570f147822ddc3e5469c0e01c4b60be0":{"kind":"REWORK","_number":1,"created":"2025-08-21 12:40:39.000000000","uploader":{"_account_id":9649,"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","username":"avladu"},"ref":"refs/changes/91/958191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/958191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/958191/1"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2025-08-21 12:40:25.000000000","tz":180},"committer":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2025-08-21 12:40:25.000000000","tz":180},"subject":"python: cve: use latest cryptography and urllib3","message":"python: cve: use latest cryptography and urllib3\n\nFixes the below pip-audit failures:\n\n```\nFound 2 known vulnerabilities in 2 packages\n------------ ------- ------------------- -------\ncryptography 43.0.3  GHSA-79v4-65xg-pq4g 44.0.1\nurllib3      1.26.20 GHSA-pq67-6m6q-mj2v 2.5.0\n```\n\nChange-Id: I72b67be4104f3d5191b4941a80b2dd9434090f12\nSigned-off-by: Adrian Vladu \u003cavladu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b1f4edc570f147822ddc3e5469c0e01c4b60be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b1f4edc570f147822ddc3e5469c0e01c4b60be0"}]},"branch":"refs/heads/master"},"18f04a9d8f6322b1ba3c38036270549cb14dedb3":{"kind":"REWORK","_number":2,"created":"2025-09-10 07:51:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/958191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/958191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/958191/2"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2025-08-21 12:40:25.000000000","tz":180},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2025-09-10 07:51:56.000000000","tz":0},"subject":"python: cve: use latest cryptography and urllib3","message":"python: cve: use latest cryptography and urllib3\n\nFixes the below pip-audit failures:\n\n```\nFound 2 known vulnerabilities in 2 packages\n------------ ------- ------------------- -------\ncryptography 43.0.3  GHSA-79v4-65xg-pq4g 44.0.1\nurllib3      1.26.20 GHSA-pq67-6m6q-mj2v 2.5.0\n```\n\nChange-Id: I72b67be4104f3d5191b4941a80b2dd9434090f12\nSigned-off-by: Adrian Vladu \u003cavladu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18f04a9d8f6322b1ba3c38036270549cb14dedb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18f04a9d8f6322b1ba3c38036270549cb14dedb3"}]},"branch":"refs/heads/master"},"e1348f093c2f30c01e6bf4b904a5b9a30d1dfd53":{"kind":"REWORK","_number":3,"created":"2025-09-12 13:17:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/958191/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/958191/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/958191/3"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2025-08-21 12:40:25.000000000","tz":180},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2025-09-12 13:17:33.000000000","tz":0},"subject":"python: cve: use latest cryptography and urllib3","message":"python: cve: use latest cryptography and urllib3\n\nFixes the below pip-audit failures:\n\n```\nFound 2 known vulnerabilities in 2 packages\n------------ ------- ------------------- -------\ncryptography 43.0.3  GHSA-79v4-65xg-pq4g 44.0.1\nurllib3      1.26.20 GHSA-pq67-6m6q-mj2v 2.5.0\n```\n\nChange-Id: I72b67be4104f3d5191b4941a80b2dd9434090f12\nSigned-off-by: Adrian Vladu \u003cavladu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1348f093c2f30c01e6bf4b904a5b9a30d1dfd53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1348f093c2f30c01e6bf4b904a5b9a30d1dfd53"}]},"branch":"refs/heads/master"},"52cebbaca15b6e8bfa45cda1b404b7ae449e8089":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-06 09:05:20.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/958191/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/958191/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/958191/4"}}},"commit":{"parents":[{"commit":"af5fa8a98f1f093d01d87f8e57579e14e7ca5f61","subject":"Merge \"Deprecate ironic-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af5fa8a98f1f093d01d87f8e57579e14e7ca5f61"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2025-08-21 12:40:25.000000000","tz":180},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2025-10-06 09:05:20.000000000","tz":0},"subject":"python: cve: use latest cryptography and urllib3","message":"python: cve: use latest cryptography and urllib3\n\nFixes the below pip-audit failures:\n\n```\nFound 2 known vulnerabilities in 2 packages\n------------ ------- ------------------- -------\ncryptography 43.0.3  GHSA-79v4-65xg-pq4g 44.0.1\nurllib3      1.26.20 GHSA-pq67-6m6q-mj2v 2.5.0\n```\n\nChange-Id: I72b67be4104f3d5191b4941a80b2dd9434090f12\nSigned-off-by: Adrian Vladu \u003cavladu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cebbaca15b6e8bfa45cda1b404b7ae449e8089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cebbaca15b6e8bfa45cda1b404b7ae449e8089"}]},"branch":"refs/heads/master","description":"Rebase"},"7b1fac3c643ba008750e1bf87233a1dc0d758014":{"kind":"REWORK","_number":5,"created":"2025-11-03 10:40:01.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/958191/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/958191/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/958191/5"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2025-08-21 12:40:25.000000000","tz":180},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-03 10:39:45.000000000","tz":60},"subject":"python: cve: use latest cryptography and urllib3","message":"python: cve: use latest cryptography and urllib3\n\nFixes the below pip-audit failures:\n\n```\nFound 2 known vulnerabilities in 2 packages\n------------ ------- ------------------- -------\ncryptography 43.0.3  GHSA-79v4-65xg-pq4g 44.0.1\nurllib3      1.26.20 GHSA-pq67-6m6q-mj2v 2.5.0\n```\n\nChange-Id: I72b67be4104f3d5191b4941a80b2dd9434090f12\nSigned-off-by: Adrian Vladu \u003cavladu@cloudbasesolutions.com\u003e\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b1fac3c643ba008750e1bf87233a1dc0d758014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b1fac3c643ba008750e1bf87233a1dc0d758014"}]},"branch":"refs/heads/master"},"9e9a0538296b087acea2cce1112e3a97c8945661":{"kind":"REWORK","_number":6,"created":"2026-01-05 10:03:14.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/958191/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/958191/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/958191/6"}}},"commit":{"parents":[{"commit":"0cd36c4f8da2ad8c6a02ef71ab9f239734593760","subject":"Merge \"update constraint for glance_store to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cd36c4f8da2ad8c6a02ef71ab9f239734593760"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2025-08-21 12:40:25.000000000","tz":180},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-01-05 10:02:56.000000000","tz":60},"subject":"python: cve: use latest cryptography and urllib3","message":"python: cve: use latest cryptography and urllib3\n\nFixes the below pip-audit failures:\n\n```\nFound 2 known vulnerabilities in 2 packages\n------------ ------- ------------------- -------\ncryptography 43.0.3  GHSA-79v4-65xg-pq4g 44.0.1\nurllib3      1.26.20 GHSA-pq67-6m6q-mj2v 2.5.0\n```\n\nChange-Id: I72b67be4104f3d5191b4941a80b2dd9434090f12\nSigned-off-by: Adrian Vladu \u003cavladu@cloudbasesolutions.com\u003e\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e9a0538296b087acea2cce1112e3a97c8945661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e9a0538296b087acea2cce1112e3a97c8945661"}]},"branch":"refs/heads/master"},"277cbe320a9b12309d4e6bde595e66aca71b05a1":{"kind":"REWORK","_number":7,"created":"2026-01-05 19:02:59.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/91/958191/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/958191/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/958191/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/958191/7"}}},"commit":{"parents":[{"commit":"0cd36c4f8da2ad8c6a02ef71ab9f239734593760","subject":"Merge \"update constraint for glance_store to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cd36c4f8da2ad8c6a02ef71ab9f239734593760"}]}],"author":{"name":"Adrian Vladu","email":"avladu@cloudbasesolutions.com","date":"2025-08-21 12:40:25.000000000","tz":180},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-01-05 19:02:36.000000000","tz":60},"subject":"python: cve: use latest cryptography and urllib3","message":"python: cve: use latest cryptography and urllib3\n\nFixes the below pip-audit failures:\n\n```\nFound 2 known vulnerabilities in 2 packages\n------------ ------- ------------------- -------\ncryptography 43.0.3  GHSA-79v4-65xg-pq4g 44.0.1\nurllib3      1.26.20 GHSA-pq67-6m6q-mj2v 2.5.0\n```\n\nChange-Id: I72b67be4104f3d5191b4941a80b2dd9434090f12\nSigned-off-by: Adrian Vladu \u003cavladu@cloudbasesolutions.com\u003e\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/277cbe320a9b12309d4e6bde595e66aca71b05a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/277cbe320a9b12309d4e6bde595e66aca71b05a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
