)]}'
{"id":"openstack%2Frequirements~958440","triplet_id":"openstack%2Frequirements~master~Ib85335df6d3167b186e3b0e3c2fb124a12d2044c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib85335df6d3167b186e3b0e3c2fb124a12d2044c","subject":"update constraint for oslo.rootwrap to new release 7.7.0","status":"MERGED","created":"2025-08-25 13:08:06.000000000","updated":"2025-08-27 14:42:29.000000000","submitted":"2025-08-27 14:40:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958440-new-release","meta_rev_id":"0ce6789b0f381d0726388bf0091a6880fd1aa384","_number":958440,"virtual_id_number":958440,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-27 14:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-08-27 12:00:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-08-27 12:00:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-25 14:01:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-27 09:17:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2025-08-27 12:00:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"154888403223a7eb1818e58128207f6d10bc54ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-25 13:08:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14da29d83cbaab1ab4ed9dd3ede833345789d3c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 14:01:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb12b0a9e2694124ae94d2d21ffe3f4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93500120b99949d8a9e5452aa443350e : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e76b6e5a89c94743b211f088ef7f3153 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc0489a3f46b4c63bd5ab8ea1913e9f4 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc0aa64e9496440c9cc867a00263ab11 : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d54dcb485a544c5b471c05f8ed638c9 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ae4c9d18e8f4e6d92821378ac78019c : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b21ee481c4414af0b5d15533ac8c8894 : SUCCESS in 3m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b60deb6a49dd482cbb963acfe52c8ac3 : SUCCESS in 4m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e436ed757ccb4ef8b4bcff438f3b7be9 : SUCCESS in 9m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a9170fed6aeb4a24872c92a619e3c2d6 : SUCCESS in 24m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/502817600a5a49abb0b7e7f9f1a53cd7 : SUCCESS in 10m 43s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/cb9cd0a56c1c40aeb0752b8778fbda02 : SUCCESS in 6m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bacf0f35ad1e4404b0afd8b493cecef5 : SUCCESS in 4m 35s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/44e6dd661b414eb9a39397ac5a1464ec : SUCCESS in 2m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9a5de193d7b94bd7a0afb05305a3be51 : SUCCESS in 10m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e3883fb315b94ccdbd7855b691504a38 : SUCCESS in 5m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4f2ec0ea355f49fe8d698d98aaed2e6b : SUCCESS in 7m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0a45d820c94c49f4a14156ea9209c7c0 : SUCCESS in 5m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ccc052d032c544c09e4288a1790d3fa4 : SUCCESS in 17m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5eee9a5bbbad4750852ae6ced933b32a : SUCCESS in 7m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cafa964be6124b5db59876c0d7cd9132 : SUCCESS in 8m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9b4fcd64907b4b46982bacca4385128a : SUCCESS in 10m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5ac5bb5cd19c46d99fd70ab350e6081a : SUCCESS in 2m 53s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/eb2085cf70ff476e88dd1a05b2b17350 : SUCCESS in 8m 36s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/798f53f2f30a4591b1bbaedd955f6570 : SUCCESS in 4m 38s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f64116c575af42bba8a2d4ab0d996ba4 : SUCCESS in 5m 28s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3d7887d8a77f40bebf296de7bbd2feb1 : SUCCESS in 13m 36s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/db3b02b2dc47439a988abde384f42cab : SUCCESS in 19m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9eb14da00f7249feb64c32ebbc93e2db : SUCCESS in 4m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/159ec7a32bb94340b118d79165b55429 : SUCCESS in 8m 47s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3c3fb008c4e043058821540f97893896 : SUCCESS in 5m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/00734c916b2346e182e26d13cd112f4e : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2d3a46c9c0d4c86bffce2d807c7f511 : SUCCESS in 24m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b106985cca474a478fd63b7a4b2ce347 : SUCCESS in 6m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b59a7d28d0764623aed2d4512d1cb696 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6940ff5ac83b4e2f91d9aa7a02891830 : SUCCESS in 3m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b46a6ec8f8ed4e299692903d1bb2254d : SUCCESS in 19m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/64657fc4b8084c5cb101f1553207e946 : SUCCESS in 3m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a386a7efeef42c6af44b826d3be4f1d : SUCCESS in 33m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7e94b8665d84e0ea0beafb4bb40b99e : FAILURE in 45m 31s","accounts_in_message":[],"_revision_number":1},{"id":"627b4bfd2046de5673d628e8b17e2230a3cacc48","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-27 09:17:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0298ded24598971f50c0e7e8cd1f34c012473c17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 11:22:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d4c97af08884de9a1b2aa02040f3b5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7abcbe6e07aa4239a0d414524f7eaca4 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d598cb3fd1642568daaee6f388ef0cf : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3be6e7aaaec3470787b7f07b6ecd3d85 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e66ae862f344e68889502b23f5e1290 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0bc64e6eb0e34ad19444ad57b9e1f624 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d00ac17e83904f5d965675b6f6c6e84c : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/46f5e9dae8a04cefbac3c8e7e421fac4 : SUCCESS in 3m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1d31ee163bf8424e86b127efec591b80 : SUCCESS in 10m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8aa7a40552604f83bd903f749b41a8c8 : SUCCESS in 10m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/530bd3cde2084af9b123437b0a49d2d4 : SUCCESS in 18m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/bb627e652a154253a5941528175faf5f : SUCCESS in 24m 06s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f645a3dfd592425bb73a412bda0200d8 : SUCCESS in 5m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/34fb0246883e4444afcb5df7ca577df4 : SUCCESS in 5m 28s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/bf48977286394b24a2b48ed95d15549d : SUCCESS in 4m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/11325045285745599164fa9d6a708697 : SUCCESS in 12m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/cfc75b0d342d49dfa98cc42a2391f9f2 : SUCCESS in 3m 14s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/aff61bad93e6430193e66694c8e34a12 : SUCCESS in 6m 16s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/69ea47b68087464dafd26b707a6c1467 : SUCCESS in 8m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/144594e84b724373b8b7e6229716c8f3 : SUCCESS in 15m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad61f7fad386450583ba983f78c86272 : SUCCESS in 6m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6c5f8dc4b52946c1a8e9f01bccd6cb0c : SUCCESS in 8m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f380ef3ad0844f528c59722535e5fbce : SUCCESS in 11m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0b37b17dbb7a4c53b680b59cf9e50273 : SUCCESS in 2m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/74a1d0ef2ac74754baf4a2a738d8cbf8 : SUCCESS in 12m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/366963f67f8e47fa85157f89a4940327 : SUCCESS in 5m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/148aaa7f497549538aceb2d4fe52bc25 : SUCCESS in 5m 58s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2d58141ba1804edeab6671e595adf541 : SUCCESS in 12m 23s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/adf027c947314080b91c070ef9ff4c4a : SUCCESS in 19m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bf57cb9b77e54cee86e684447e085bf3 : SUCCESS in 4m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4b2bd79d54e74294834ce5461e75da6c : SUCCESS in 13m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d56da327c66b483fa2c9ffee3c09612c : SUCCESS in 7m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/59e68cb4ed0642ad807c4fa059e0b8a3 : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a2fa10ccaf74baa8e62b4d81a340548 : SUCCESS in 24m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/22ccff3c7f544137b3404dfce3573c32 : SUCCESS in 3m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c7ea44aa2d704ba1b87f7e2fd36d3bf9 : SUCCESS in 7m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d7ace8873a464596bc19a318b0f78ea1 : SUCCESS in 6m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/522675564b294eabb7ef4ed25e8a0782 : SUCCESS in 17m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/64d87389670c480097c1d6a26cb6ea0a : SUCCESS in 5m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/922509e0fd7e43d6bbf50a6d82ff15de : SUCCESS in 1h 00m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7843eb60f7cb43d29b210659713b958f : SUCCESS in 1h 50m 50s","accounts_in_message":[],"_revision_number":1},{"id":"889a3d8e57941cdd659a3f3302e435e1243beeb3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-27 12:00:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5856517f87db923f03365168fddc190d23ca7f31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 12:02:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2bba7f071278081965cb9980a74115093d613384","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 14:40:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/328acd7729db4852ad5c5e18b2019840\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/675b084796f2489093bf4ff97d5aa332 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cdb3e1f4f8348c79bb17494a28f22fa : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/075e110aa6964b50b44ebc1b39cfe9d5 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3963bdb4691a43d394aa8e3dc3bfedc7 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/582727943ef74935a449828fdf47ec86 : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cca4d89cd9e64203b1f699dd1a27b725 : SUCCESS in 3m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a0ac180b726d408fa5840b09a9d11d95 : SUCCESS in 10m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c1846844757a48aea374358713d24fb4 : SUCCESS in 9m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e04679776ebb4c5396ebe42d8d5e2e9c : SUCCESS in 14m 27s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4bd4baef7f154b72b9e147ca922156a5 : SUCCESS in 11m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1038780e78b34d5da1a07d4b7ace8950 : SUCCESS in 3m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/83e62ea86da94c6caa730c6d93515b5e : SUCCESS in 3m 41s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9828480bef394860bc13a1a36e8963c8 : SUCCESS in 4m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a2e53faed1d0499bbf93f57c8ad6c8b5 : SUCCESS in 19m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f439785a681643eebac1bfa3ead8be52 : SUCCESS in 7m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fc5666a6e3ff4030a8c98cfd187d17f3 : SUCCESS in 8m 06s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/160ad9cda6bb47b0b9f892d3414d8024 : SUCCESS in 10m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e61e14c4f5e94aca878b2bd404971b2a : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d70363124fe9454fab217ee48d11ab2a : SUCCESS in 11m 33s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b8322c4ee68c4af3abae5b8ecade5187 : SUCCESS in 14m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/328c748e2b2c4bfba9a20e1a7c85ced4 : SUCCESS in 8m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/15dcf1a7856b4bb0b4a39c67a463cff8 : SUCCESS in 14m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/99c81a61df9b48be8e4decfcfa53890b : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8983cb470811495c83eef271e4e1ccd6 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a564db1c0a1453ca1afb4a922b3099e : SUCCESS in 29m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7b27345d0f6f42d49af1f61083733bac : SUCCESS in 4m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dbe481ed01d54ac6bd12bf614edb2d26 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6a05b0faec84395b1504a8cfd919c74 : SUCCESS in 5m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9f4c47906a52484cb44ac63603bc313f : SUCCESS in 13m 03s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4ac5a328074e4fcca8aeb4efe118400e : SUCCESS in 3m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efdfbac525224ccf9c7a8f2c20295109 : SUCCESS in 1h 05m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/685093bd876c4167a712c11858870644 : SUCCESS in 1h 44m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0ba3d5c71b44100a58b0517b877cd6c1ad7802eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 14:40:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0ce6789b0f381d0726388bf0091a6880fd1aa384","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 14:42:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e10162eca9b84d34b68ea9a5f33724e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f9b01b13654e1f95350ae72a119cc1 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a729b0e3f0f72ea715719c9565e8c392401d51d4","revisions":{"a729b0e3f0f72ea715719c9565e8c392401d51d4":{"kind":"REWORK","_number":1,"created":"2025-08-25 13:08:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/958440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/958440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/958440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/958440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/958440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/958440/1"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-25 12:58:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-25 12:58:05.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 7.7.0","message":"update constraint for oslo.rootwrap to new release 7.7.0\n\nmeta: version: 7.7.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I0a3aa00afa038677c0e400ee5e06c3f6319e0fd4\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ib85335df6d3167b186e3b0e3c2fb124a12d2044c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a729b0e3f0f72ea715719c9565e8c392401d51d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a729b0e3f0f72ea715719c9565e8c392401d51d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
