)]}'
{"id":"openstack%2Frequirements~958441","triplet_id":"openstack%2Frequirements~master~I47e14ec7994ddcba87461dbddd1c8650c59f63f6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I47e14ec7994ddcba87461dbddd1c8650c59f63f6","subject":"update constraint for kuryr-lib to new release 3.3.0","status":"MERGED","created":"2025-08-25 13:08:45.000000000","updated":"2025-08-27 14:02:50.000000000","submitted":"2025-08-27 14:00:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958441-new-release","meta_rev_id":"5c8bc5950d8dc45dfb837cc9c0b2b5589d686957","_number":958441,"virtual_id_number":958441,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-27 14:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-08-27 11:59:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-08-27 11:59:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-25 14:52:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-27 09:15:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2025-08-27 11:59:07.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"abe199c2709afa6a45a61b91e9de9470d47a1a5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-25 13:08:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d379c1ad3033a37b0c571e0275b561875523617a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 14:52:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a96a5e1d07b449f9aaacde2c8b7da02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3367ff0c6c484ec1a3e42f6912df9499 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee7588a56bbd493baabb3c8cea0ba455 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eafc72acb12f4122ab30a38662753471 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d7000271051494a9ae65fe40bf56156 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b60eb592fa14ed2a285081c0cca4b77 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9edc3611085f4d179d592f60f24b4144 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a82c40059a040bbb1eef936ccf0ea92 : SUCCESS in 3m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/432290d2ae01473b9556ea6175be4c7b : SUCCESS in 9m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/83676f85d6fe481294cf5b3d7571e2f3 : SUCCESS in 17m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/89bf6f5fb66a4190a46373f9ec2d6c12 : SUCCESS in 19m 57s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/707258b1053a446791d4bd6ca3e30f15 : SUCCESS in 24m 43s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/04a26669e90340169ea2dd875ffcff12 : SUCCESS in 4m 37s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/badc619eadea407593fda4fa156b0bc7 : SUCCESS in 3m 23s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/31eb0aa719c44de3b1c5ec25501529df : SUCCESS in 4m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/87c8a62fadaa4746b76c24e9dac354d3 : SUCCESS in 11m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/404b96d8ec1248e39ad086a0807ce1ad : SUCCESS in 6m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6434476e41e649b9a37286630a7d38dc : SUCCESS in 5m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/97e719c721744a5f90f2fef970e64930 : SUCCESS in 7m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f6d6e07abf9e42e08873f64a1a795de4 : SUCCESS in 16m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/402138d5617d465a91b17de573ab0f7a : SUCCESS in 10m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/22653853fa5442e5bcc338d15c5b08a4 : SUCCESS in 9m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c6dec7cb4c3d407e901e9b036064cb6d : SUCCESS in 13m 28s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3c9c049bfb654c10b87fc6b7d0329d4a : SUCCESS in 2m 00s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fe22ba2c85554fa1a1eff6027604c9d3 : SUCCESS in 9m 53s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fe78df878a5d483ba56b6f819d0287b0 : SUCCESS in 4m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9043e35ef8e249bb85f8626f0c8631c0 : SUCCESS in 7m 01s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ca2308b35f33433a8f45681427ae2d09 : SUCCESS in 13m 20s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5f4307842c1644239f4b5cf76d0d7235 : SUCCESS in 26m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4eb6a366c2764715a4ced458d0fac494 : SUCCESS in 8m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4b5031514ab74b8eb383ab3be0b9d3ea : SUCCESS in 14m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e3ff2e6eb46a4abc90cdcaa32f9809db : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b969b4af6c664029b82318e1ec7e507f : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a9b728be6c1f4d029232635850dea034 : SUCCESS in 23m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2a9f3cfb5ba3406390524c6f49b2778e : SUCCESS in 7m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/285087fa3c4546a8a8eabfc07da17677 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2fe8eee530ae4d98b3da06e2178f1bad : SUCCESS in 7m 18s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/11e3a40917d242468160c1f38edafe4a : SUCCESS in 18m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f6f3f4e14cce4a138d1cdd804fa57a5e : SUCCESS in 6m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff94ddece0d7460fb9220fdac752fa07 : SUCCESS in 53m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac187f941778476badbc684e004cae15 : POST_FAILURE in 1h 35m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ed9462a6dcade574dea7c37902017039e4e4d0bb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-27 09:15:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd713bff19834040bb34183e16403ba5d9b6df28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 10:21:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f01efa42beaf4528826b4bc84b78dbdf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db675c4127784ba88420251fe01fd6c2 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b5f3fcd833a4605b306aafe66594189 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71a9b7b889d346559a8040e162af7462 : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cdfb6bdfb8f74d94b8377903ecf2e0fc : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a24b1b9a50447a78b57cca9769854da : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c8a5ba22e96434ab64351d246eddc21 : SUCCESS in 2m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b5db889d17594e8fbc9a0ce9cd6c40ab : SUCCESS in 3m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3054d579d55640a18c3f9808de474394 : SUCCESS in 4m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b481edb65c0a43ad97b5dee5f87df358 : SUCCESS in 16m 14s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dddad075bf584d5b814951c310eb7991 : SUCCESS in 17m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8d09026bdc7044b08197532f68207297 : SUCCESS in 24m 07s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1871eff5e37a47caa7f9b00905a5d6f7 : SUCCESS in 4m 43s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/487f952bf0424e4e9f035a91cf2f15d3 : SUCCESS in 5m 17s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/79b30c4bf5574a86a9cb11e7da18652d : SUCCESS in 4m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a939fd2136a041729ed85dbd19bcdab3 : SUCCESS in 11m 25s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c5de53fb168f47b2acf0482671211bf3 : SUCCESS in 5m 56s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cc522d2e62344bc79ee41b16ef5bbc56 : SUCCESS in 6m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e504df55039748f68bc351dc952f9792 : SUCCESS in 8m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d4dd49a385b94601b9b00c2f776c533e : SUCCESS in 14m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/317a988e79344268aec0eee9cfac7337 : SUCCESS in 7m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/430602fcecf54c20bfea6d17e3a6c4cc : SUCCESS in 6m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/8f168ddf4c50407697317a453a98362f : SUCCESS in 12m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/15fc360c421e403396897756ca6bfe94 : SUCCESS in 3m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/aebfd6c056984b1b9a87c1a0c780b34d : SUCCESS in 11m 56s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d9b4b35614da43e0a30e388e90e3bb04 : SUCCESS in 4m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9cbc42687ef0448482357793f47b78e8 : SUCCESS in 5m 09s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a628a78387da4d4fb13db853c20ae385 : SUCCESS in 11m 38s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/516683b7d4cc43e9aa6cebdb399278ba : SUCCESS in 19m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e8bf751196f447f3ac516d4908515dd1 : SUCCESS in 4m 18s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/163bba45881d4c44b6ba0453baeaa8aa : SUCCESS in 10m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/946cced3820b4d3aab82fbe22603f706 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dfb189d73bd1453fabda0755979c5541 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/83587192642a4e0c8071fbd69f59c8f8 : SUCCESS in 27m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6973b20e1eb642b680f7b093571db5f0 : SUCCESS in 6m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f1b54900b1264b908862a90e7d585b67 : SUCCESS in 7m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ae921b850094c3a99dfa4834d3dbc52 : SUCCESS in 6m 54s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1ee867d522ac4239958103bbe2031f00 : SUCCESS in 12m 03s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d78599c9a1e940d186ae2cfb8419100c : SUCCESS in 7m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/494b027547a94fe9a3169d85b08e9e02 : SUCCESS in 49m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a0b963021ae4e68a2bebb4ed0749d67 : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5f1ba3edfcd0eae92b9d8f56d8a2560cd8031c63","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-27 11:59:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ca6c238e66c8cbbe0b6135cc19b8eea4c3ce444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 12:00:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f7fb5fc9108ab407296a39db679b3fd3c78c400","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 14:00:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a232e1294284e66be84a46d88de16d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60103c4ba0f4a0eb9963a8bd7424244 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c06dc4526401466db8a1fe089de57687 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d66b2fb0d7044fb481220819315f17c7 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/39ea02db734a4f53837b1190f4e89cd9 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c57da888a7f24e8b87a3ec958e3219a2 : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f28c7128dfa46d4a4550e35e1ab2aa4 : SUCCESS in 3m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5e2a8e5a003b4c5ca6e985cf538db050 : SUCCESS in 4m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0d9b885a177f47d59f2790c1aa4495bd : SUCCESS in 20m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/50cec1dec9cb4e3e9126400e469f2cee : SUCCESS in 24m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c8301800a5a94ef789e4e8e57cd5a0f9 : SUCCESS in 12m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/672da154fdad46b7ab161f035f0497d0 : SUCCESS in 6m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e770a52dc7fc46058436ae2226efb6ab : SUCCESS in 3m 42s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c12dab0c20154bf78493649942162aaa : SUCCESS in 7m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7b96fe34725b43ec848f11a541f815f0 : SUCCESS in 8m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b3c3f4654dd4c3cbe7ae84d8765c359 : SUCCESS in 5m 47s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4105f9e306f74d469c7e31e73c5ae520 : SUCCESS in 7m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/58b1314f6500453685df4d6b48c7fb11 : SUCCESS in 10m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9cc1782110884328af380bb0dba8beea : SUCCESS in 2m 44s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cc959c79c5a043079ebae307032bb2d2 : SUCCESS in 10m 07s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/07177fb8cd6a43f4b4ecc740cc071547 : SUCCESS in 20m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4e720fbec64b47ae99813e0418f1296b : SUCCESS in 4m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cc0d4b0db4704ae1ac280923866d2b13 : SUCCESS in 14m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2d8ebab7d80a4c71a878a2b47f7f628b : SUCCESS in 3m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f8f511344c9244198c558112c4f95708 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/faa9aa4e056d42dcab276306fd55ec43 : SUCCESS in 26m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/127f5558e8c048bdbe2d2a081ea7fc81 : SUCCESS in 6m 11s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8361790186dd4913baf5d4e580a4fc3c : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9352e57f09342378409f39fb28a9e1f : SUCCESS in 5m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/55cf9729e85a40ba90b06c1e69592792 : SUCCESS in 8m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/23482ccf6df44379a92908a2e9daa8e0 : SUCCESS in 5m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cde646cd3d6243d4b81b9a6495b00779 : SUCCESS in 47m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/754261d9eb52485583c084be807db0c8 : SUCCESS in 1h 55m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2ae6631a7003b75bfd7243875ebaead9260187d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 14:00:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5c8bc5950d8dc45dfb837cc9c0b2b5589d686957","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 14:02:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83970765810242299ef8a75f05d93b80\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c92bff2181a45c084df7e63a8d3f974 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b5d34f488332198716f42c6958cccc4868e8122","revisions":{"2b5d34f488332198716f42c6958cccc4868e8122":{"kind":"REWORK","_number":1,"created":"2025-08-25 13:08:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/958441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/958441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/958441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/958441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/958441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/958441/1"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-25 12:58:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-25 12:58:44.000000000","tz":0},"subject":"update constraint for kuryr-lib to new release 3.3.0","message":"update constraint for kuryr-lib to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: team: zun\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I414313e98f4479a00c2707f6685447ffe1808150\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: hongbin \u003chongbin034@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I47e14ec7994ddcba87461dbddd1c8650c59f63f6\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b5d34f488332198716f42c6958cccc4868e8122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b5d34f488332198716f42c6958cccc4868e8122"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
