)]}'
{"id":"openstack%2Frequirements~958445","triplet_id":"openstack%2Frequirements~master~I53909075b845576110333c8a267f03c384bde624","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-08-27 11:39:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I53909075b845576110333c8a267f03c384bde624","subject":"update constraint for neutron-lib to new release 3.21.1","status":"MERGED","created":"2025-08-25 13:13:32.000000000","updated":"2025-08-27 11:39:02.000000000","submitted":"2025-08-27 11:39:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958445-new-release","meta_rev_id":"c5392f64ff73fb1d7a3b4735eef8d64738ea5f07","_number":958445,"virtual_id_number":958445,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-27 11:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-08-26 06:09:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-08-27 09:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-08-26 07:05:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-08-27 09:08:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-25 14:41:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 06:09:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-08-26 07:05:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-08-26 07:07:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-08-27 09:08:40.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"ed2d0e0ef995bfb8afca62d697a7485cb1d5a8de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-25 13:13:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f296d58f32cffa7152e83595201ca11cfb34668d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 14:41:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b548ce4f1644ed9b4c79bde32b70cf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d37c853b9f43449000536fa90f369f : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca5e6735d3ee40bea5e603ba012ed784 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70aa2db6dde7492e82be72ea9c2bb7ed : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba65594e99c14983ac39b61585ee5ce7 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da547f7bd8c5486ebf4490a705cda631 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bdd49cbd1e5f4afdb10d5267ffdc732e : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ab8cf6bb3a64ac79e00ccebcb686f59 : SUCCESS in 3m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/45cc81e6cde64533b65c2f40dbbf5a34 : SUCCESS in 6m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b142a6ec017a4c1886f818fe878c5920 : SUCCESS in 16m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/caab5687689f43b4ad623b1603a79d66 : SUCCESS in 23m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f64a8879cbcf486a93c5556e80a780e1 : SUCCESS in 13m 46s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/78aacdb20ac540a9aae1b065ac4543cc : SUCCESS in 6m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/71733c748318499587883c8f86a6cbf1 : SUCCESS in 5m 25s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4b92b54055a6495abec56661b912d7da : SUCCESS in 6m 45s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a0691cbe91ab4e35a7752f1e3e01bf12 : SUCCESS in 7m 16s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ddc20449eec74ad6a6b83da2352c95d0 : SUCCESS in 6m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8b9ad1f91d824b228698f08cfb371be6 : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b9276c374b3245eb9ca04f3022301f41 : SUCCESS in 10m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/68d71d23e38d4e02bb898704812b622b : SUCCESS in 14m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed167d5dad464e238635e69ff1ad5768 : SUCCESS in 8m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/10538e7a1faf4defb5748c8dd6fa0db8 : SUCCESS in 12m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ec554057ed934caeab4a601865abe8c1 : SUCCESS in 13m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1ea7e0bcf2fd41999f4b9600cc5302fc : SUCCESS in 2m 37s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cc2e86908596412fb46408ab76de8ee0 : SUCCESS in 6m 54s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/74668c2e405e41a58c7cc4ba7542e283 : SUCCESS in 4m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a373ab2626aa4a8fb6d051930f5404a2 : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/1769ee916b8b42ed8d06df5802da7db5 : SUCCESS in 14m 12s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/66a208954f284e2184b1c5995d664d4c : SUCCESS in 19m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/33ef77fef166430d9d90d2d6bafb0d3f : SUCCESS in 7m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b4946d76962a493d8962135049efc655 : SUCCESS in 11m 47s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6dad2dbbd9dc4330a97277fa067da91a : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7508ef62a92a455cb9c09d1d9ddce7d2 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ccc13a004ee840fab75821ed9d6783a4 : SUCCESS in 27m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/66811076fddf4ca9a91cd7e5f6ec7800 : SUCCESS in 6m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/eec93277e71748c1bc3e7580304cf5f2 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2499bfe6161049cc9747bc9a9f3db35f : SUCCESS in 5m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/737e3b65603f4825b5794583a883760c : SUCCESS in 15m 09s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/492ce0568c394fc9b18288401471a3fc : SUCCESS in 6m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4935d98f80b74729b343ae7be535b051 : SUCCESS in 33m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d32368cec56c46b68c9ec8ca697bad10 : SUCCESS in 1h 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a62454fcf241d3c98f288974aaf7cd7ac9022087","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-08-26 06:09:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b0c87b15e520284f6e233b095343796b33bf470d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-08-26 07:05:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d44b3cd5a9c91f6bcb0cdc5d376c25a661ea98d6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-27 09:08:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cff7e217704e7e312c871e6e99ebcb6726ce3b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 09:11:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fab128ac4f05761f62e65445443ff5c57b06b32f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 11:39:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b649179e6e94a5bbab66fee1911a6fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a0205296f44b5c82c5695db3bcaa8c : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c1e3f728b5242c3a4fe405db5290bbb : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a87169930f7a46e28aad3636f9b8e529 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a295677e582641ab883525854d456c42 : SUCCESS in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a182554017924650a85d13ca98bf08cb : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8577b3ce9ba443029c84d2ec87e6e75f : SUCCESS in 3m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c091563ba1c142a69468fdd35e9889ea : SUCCESS in 9m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/37d153bd9f2c4d808c4a6a5960129762 : SUCCESS in 20m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0de9b02e8fca42c98e91a7db043ec934 : SUCCESS in 19m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e90f7151c8e3442eaf451ffdd461793a : SUCCESS in 8m 00s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0dfb04d12be94313b22cef032b8956e7 : SUCCESS in 5m 59s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/46156b7b748640d994e61e7e34726c10 : SUCCESS in 3m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1c591e5a653446b28ed9bedc523a693b : SUCCESS in 8m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/302cbb1940a246a79cf5ebe89bac3aaf : SUCCESS in 8m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ce782a3d8024bae815ea5fff9675d0d : SUCCESS in 7m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9994824e66c44f08bb3e65f340879016 : SUCCESS in 10m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bef83280c893436baf2aa23ab8dbc78f : SUCCESS in 16m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/df55add0e2d848c6952f23df8963e120 : SUCCESS in 2m 39s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b3f9e8347a114b89ade56ae5dc2e951a : SUCCESS in 7m 26s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e1068dcd905d4d08ad851aca3c2ab9f6 : SUCCESS in 25m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/798462e0054b438dadce6129b8d97f49 : SUCCESS in 8m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1f4b40f8f1514d59820eedd9ae3440bc : SUCCESS in 9m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9b4d4c2b92904778b933b3ff14e6ed34 : SUCCESS in 4m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/35a0ed5cf85b41b6b9d2eeebe31f1443 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7a80d784874477b9bab08255882304f : SUCCESS in 24m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9e19de4bb2c1462cb08d99f3ea42a1b6 : SUCCESS in 8m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/220d9813830f4128abdeedb3e475e55e : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f60bde60de2a49c99495e48948dd523e : SUCCESS in 6m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7dc6ec2d81ff479b8519a22ee0e7f0d7 : SUCCESS in 9m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a7905ce8572e4b7a9bee36ee952973a5 : SUCCESS in 3m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/167924dbd94f40d1b5fc4d4845dbc194 : SUCCESS in 55m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/acbe13f654a34969b4bcf1dcacf54663 : SUCCESS in 1h 42m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c5392f64ff73fb1d7a3b4735eef8d64738ea5f07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 11:39:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1736d7ee7ed4053f0a7fee7354037a53aa419559","revisions":{"1736d7ee7ed4053f0a7fee7354037a53aa419559":{"kind":"REWORK","_number":1,"created":"2025-08-25 13:13:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/958445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/958445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/958445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/958445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/958445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/958445/1"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-25 13:03:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-25 13:03:31.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.21.1","message":"update constraint for neutron-lib to new release 3.21.1\n\nmeta: version: 3.21.1\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I7a3a7dd5f8a569f661d09b3fcb4b53c524469976\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I53909075b845576110333c8a267f03c384bde624\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1736d7ee7ed4053f0a7fee7354037a53aa419559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1736d7ee7ed4053f0a7fee7354037a53aa419559"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
