)]}'
{"id":"openstack%2Frequirements~958449","triplet_id":"openstack%2Frequirements~master~I30eb3ff0480fadd26333b811c4011a565bff56e1","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-08-27 11:37:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I30eb3ff0480fadd26333b811c4011a565bff56e1","subject":"update constraint for ovsdbapp to new release 2.13.0","status":"MERGED","created":"2025-08-25 13:19:55.000000000","updated":"2025-08-27 11:39:00.000000000","submitted":"2025-08-27 11:37:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958449-new-release","meta_rev_id":"c6471856f3c61c443100e9bd0d641b378b28fd5e","_number":958449,"virtual_id_number":958449,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-27 11:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-08-26 06:09:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-08-27 09:08:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-08-26 07:06:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-08-27 09:08:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-25 15:02:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 06:09:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-08-26 07:06:59.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-08-26 07:07:15.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-08-27 09:08:11.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"4155d3a699000d05e6f748e132b5d4171bcca30f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-25 13:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b74cc6813acf6188659120920b11a23a794b2ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 15:02:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c79ddd5e4feb4f379f09ba035d591c2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e2896e5d2fd4255a1c1341040a271dd : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f6ad3a2bc22418880e74f901a58289e : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50c5772938d54d948c78e148d8dd1e3c : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2d3f9af240e443ebecdf32be1bdd04e : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83c02ca461c24eeb814d437555f33f03 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5aa1a25354a344b189f577c1b5719eba : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b400abe54324e9e99ca5da7db057976 : SUCCESS in 2m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7d7c1cb9f0c0421f98bb8ca66cd7f466 : SUCCESS in 9m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cb95530d2ddb4fc995d3f2b34045a9b0 : SUCCESS in 17m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/281187c3544b4c3ebba3364e6941528d : SUCCESS in 11m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/11c93669e4964834a4ec4a5938783e4c : SUCCESS in 14m 07s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5be02c57763c454fa9c3d8333e50f469 : SUCCESS in 5m 24s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e63e831d51f6401a928a02b4765bffad : SUCCESS in 3m 28s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/71a72c8c7e584dfdafbc2f59bd0de1cd : SUCCESS in 4m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0876d6629f1c44d1bbda986ade1ff74f : SUCCESS in 10m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f9b3e3a7837344c68634524c5877b117 : SUCCESS in 7m 00s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bca169f7587143bbb38bb53c6e04484c : SUCCESS in 7m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c746ad5243b649be87f2f316350dfd23 : SUCCESS in 7m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/587f9c78eafb41b28323e604c8aa5564 : SUCCESS in 8m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6a78d420ca9445e0a68fc6eafe4f32c0 : SUCCESS in 4m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1b90475ab47d40fdaed0f060abe51083 : SUCCESS in 6m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/50bdd557648342d79a3f73f6828990d4 : SUCCESS in 11m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/91d627da17f84c5a92a38cf132f782c3 : SUCCESS in 1m 48s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9cfb241c47964aaa95db10bc3a910f34 : SUCCESS in 7m 10s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9372d7c6399246fd900ebbab4fafbae0 : SUCCESS in 5m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/efb2f5b8f17d4694bacb80600e6e308d : SUCCESS in 4m 08s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/989104e2a1c243b28438a59a95cd38dd : SUCCESS in 11m 29s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e30c44dee1fb44f7ba48894c50790914 : SUCCESS in 19m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2c888c3bc7234235a277145b7cab4ca2 : SUCCESS in 4m 25s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3d9b4c183a6a414f9fc862d9b07a9a24 : SUCCESS in 13m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a3b5bc43bd8e4c10bb461adbd2f3ea4d : SUCCESS in 6m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b2ff768be8de477092c87b81b7c29f3f : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67888799e7d348029048bb267f7e4113 : SUCCESS in 24m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a06065dd13c746ac9861efea245a28af : SUCCESS in 6m 21s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/98aba064ad1e41599b21fdc9e21ea55a : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c725be505284ebfa2f7d8490673fe6a : SUCCESS in 5m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/34f282629682426eb8adc550c2427c26 : SUCCESS in 16m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/72fff6ff1abd434aaf81d871bf554465 : SUCCESS in 5m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b03d9b506ca4735ba11b48ead429eaf : SUCCESS in 56m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e3bb1fc4c90463c81d837445fbd2b02 : SUCCESS in 1h 36m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e2d74c1edddd68faf173f5644baf5912814f8309","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-08-26 06:09:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c4d13d2f1732c140b6d754e188b9ad14535f1aaa","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-08-26 07:06:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"67bd5ce7cc8a4bf6b647ae03cc761c2024a7b7f7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-27 09:08:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8dd0cc57372b8466e07d98779351286e516a3b07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 09:11:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0f5cf5a962f6027cf5c1aff1fa195ec7ef7821c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 11:37:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23d293d545a5493b9e0b4e62741153d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d485daf9b4497d85c0b744572f89c4 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72fb397321e241019e41856bf9687f2a : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b136ab68f2274be981a7d78ea5460ef9 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e0bfba0a6e845879419a2c6c895a286 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/510a2d7c021640bc822bbab6862eca8c : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff486f48c6134f0983fef959f01ccde7 : SUCCESS in 3m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/708e79d1c27342109468f7df6804484a : SUCCESS in 10m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2f2344ca54b94fafb6165ea1989ac42a : SUCCESS in 17m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c9dda979444b4023ae39e9cf77cc1d1a : SUCCESS in 19m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/783af815a8c34daebde6cf07a5f6b4f1 : SUCCESS in 11m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/fa8cb597add44bdea7e3b9379ca26512 : SUCCESS in 5m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9ae02da393cc417abf1714f6c678efc9 : SUCCESS in 8m 06s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/28289bbf996941a2b61ac6d202074dd5 : SUCCESS in 7m 49s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/92d99192db434e7daf18bdb2787bc85e : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3a1a7059bef24937bc04c4964f0ee90c : SUCCESS in 6m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/999095dfefe04649b59210c6bd79e1d6 : SUCCESS in 10m 35s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/617ef129a79a4911abd7a8be4b82f90a : SUCCESS in 12m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ff401221be83423f9ece12cebdd10ebe : SUCCESS in 3m 57s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/09609c0d6bcc4bcd9adaebe608ad20bd : SUCCESS in 6m 36s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4137c6f2e4c04e32a94d2e2478ea9ae2 : SUCCESS in 24m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7bff00a6d81644c6934c1422cfb5ced5 : SUCCESS in 9m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fe7061c9f8694b90a2224d481e26fce2 : SUCCESS in 17m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/82d968e30df54bc99b24f2a40c823759 : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/13c66bbba3284927b3dd5de59ccf41f1 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6d35763cab34b6a9bb026f841daa1fc : SUCCESS in 31m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/be5e81f10b7b400bafdfa1b58d490e07 : SUCCESS in 6m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/44665146ef0e497fa488df14a42716c9 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17d25707982a44b48c35f0dcf81181a0 : SUCCESS in 6m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a2023ec7725c46fd96c96fafadb4067f : SUCCESS in 15m 44s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4c3ddebcbe63454391a0cfafd842bb53 : SUCCESS in 6m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f48c2c1d3da4bc88c42633ebb5b91bb : SUCCESS in 56m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55cae4a5e8b149ac9ce6716af39debc3 : SUCCESS in 1h 52m 53s","accounts_in_message":[],"_revision_number":1},{"id":"aa0bbfed23e3d49227a8185f1c65fe67d6468535","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 11:37:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c6471856f3c61c443100e9bd0d641b378b28fd5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 11:39:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13ea60cb97684434880e66e77ad61ca4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc90279c83004429a91a06762a800492 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7093581f21f25137bf7b73fae8a802a5c93401b7","revisions":{"7093581f21f25137bf7b73fae8a802a5c93401b7":{"kind":"REWORK","_number":1,"created":"2025-08-25 13:19:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/958449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/958449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/958449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/958449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/958449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/958449/1"}}},"commit":{"parents":[{"commit":"8d8121c57bf26115a335f82120014d4cfd0473c4","subject":"Merge \"update constraint for python-observabilityclient to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d8121c57bf26115a335f82120014d4cfd0473c4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-25 13:09:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-25 13:09:53.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 2.13.0","message":"update constraint for ovsdbapp to new release 2.13.0\n\nmeta: version: 2.13.0\nmeta: team: neutron\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I39992a5a1f16b453ff70a649d1c559fa49514dc0\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nChange-Id: I30eb3ff0480fadd26333b811c4011a565bff56e1\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7093581f21f25137bf7b73fae8a802a5c93401b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7093581f21f25137bf7b73fae8a802a5c93401b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
