)]}'
{"id":"openstack%2Frequirements~958743","triplet_id":"openstack%2Frequirements~master~I8a2e2af7a0e527ff52f118396d601c7b7f79ba79","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8a2e2af7a0e527ff52f118396d601c7b7f79ba79","subject":"update constraint for python-observabilityclient to new release 1.2.0","status":"MERGED","created":"2025-08-28 10:04:25.000000000","updated":"2025-08-28 17:53:15.000000000","submitted":"2025-08-28 17:53:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958743-new-release","meta_rev_id":"f82e5101dbe778a840c0e67f24e934ecaba9aa72","_number":958743,"virtual_id_number":958743,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-28 17:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-08-28 12:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-08-28 12:13:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-28 11:33:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-28 12:13:15.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"beeece0444e4b23cbc3d5847394d488bad8b5d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-28 10:04:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3da7e98fdf178434628c72c98ca030f78b95d558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 11:33:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7d6bb18fec14746a5d83fcc0378fc53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7255361da1a40718eef28ca702766f9 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e10dd0b8b5124d0784a1536cc23e9ac0 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/955c70f240554a8facaf7db74e034d7f : SUCCESS in 2m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20b366c2bfe140afa4ce8ef146e7e430 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c077b7840a3a458ba1b98145147dce1a : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1b54710b92a24b7781c410222c758a0a : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7413a8ec0714ab4bd1b6ca335ad183f : SUCCESS in 3m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c9abaee292f2494a868c4770e2e3b7f6 : SUCCESS in 4m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d6b4cabaa4b046caa1ec7b0b34b832da : SUCCESS in 16m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/47ff330865cb496d9298b051a17fd8c0 : SUCCESS in 21m 30s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/852353c4efad432687c5f2ac2673d48b : SUCCESS in 10m 23s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/51c768661437461d94fb14f98bdfe753 : SUCCESS in 5m 20s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/71e666fd82a340aeb7e1fd09649ae3a9 : SUCCESS in 5m 36s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d3216ad9282746d383e3d5e674f66587 : SUCCESS in 5m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/24ffe10d3ea84acf9d01d3d4098d130d : SUCCESS in 8m 14s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/84d0a82023454dab91294ff1da43fdc2 : SUCCESS in 4m 49s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0b334516df1245cca30f6cbf65a7f512 : SUCCESS in 3m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0d3d0e2c042b4dc0a6a2cd9d7c47fa94 : SUCCESS in 8m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1a589e021df94a38a843cb2cf4cefd74 : SUCCESS in 8m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6f29493650e4f9293f722946cbfdd6e : SUCCESS in 8m 08s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c3d9d54520aa489187856929f3da58c6 : SUCCESS in 10m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/626037e3edf6433a81332432de08a84d : SUCCESS in 13m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/abae1c650e0b47aaaf380f89da189d2c : SUCCESS in 3m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4552b893aa704170b375281c7bd7fb98 : SUCCESS in 7m 59s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ea4652db848e40e8929f26efdde25bfc : SUCCESS in 5m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/cf9ad33d723044deb5746b7d8910329d : SUCCESS in 6m 17s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/998a6aa9fd0141d0892fad4930c48ea9 : SUCCESS in 11m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c12c15e9122d4e6387f47cd30c2c23ea : SUCCESS in 22m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ad59a94243ed4da18c5cc065d4405cc4 : SUCCESS in 7m 35s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fb9dfbe1885f4ddeab4c067df1036ca4 : SUCCESS in 10m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e907d53c6c5141f8af76ce0faa4d7893 : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d730f243fe214a6781865cf62ecf7745 : SUCCESS in 2m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/af9355ce0392467a9731b45ee5b29a28 : SUCCESS in 24m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4760e3722d274a3b8368398c758b6191 : SUCCESS in 6m 30s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a9a6e5bd4c584598a39e9653f5ec572a : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a4c9044b2e424a57b19512964cdbd27d : SUCCESS in 4m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/038bc7454ce746af8754bbc5cffa0756 : SUCCESS in 14m 04s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/04f835e0721149b48b42cf806556a7ec : SUCCESS in 6m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ebac7ade3df46c9b80974835d325cce : SUCCESS in 57m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/845ce300b6254868a2f79cc2ada8e143 : SUCCESS in 1h 25m 14s","accounts_in_message":[],"_revision_number":1},{"id":"11778b18fca963a7e642342d71d94f85293bbf61","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-28 12:13:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"502513fec38e722615f9bbf028e13fe5dc835729","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 12:14:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d59f95fc13cde525703a787e2d07e30a19ac158","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 17:53:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a92b3582103d478b81d7e68cc5bdd747\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54865a244ab34eee875b94b290f69417 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41e3f4feb2bc429b9d867dbf6539ac4e : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a9c9fd16d6648868c841bfdc794486e : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b7ed3ce24a34bc0a66f379e778660e3 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b7f0bbaa6234d84bee73bd161103c94 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/601ecd90eae84a0f962bde344e1f5e35 : SUCCESS in 3m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/acac8dfbe8a542cab25613b6e5971d1c : SUCCESS in 10m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/881504cbe2cb4b1d887d2566c0dab40d : SUCCESS in 18m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/780242adf83e4433b7a3544941659220 : SUCCESS in 15m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/dcba5002550546e2ae013c10019afceb : SUCCESS in 8m 21s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/22ac7cd061664a3393099976b75395a5 : SUCCESS in 7m 21s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/50f1fd38d70845328c2a4ee27dc9257e : SUCCESS in 3m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fc9a74bb463644f68d00f1b2cce1e97e : SUCCESS in 7m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6b3864399e1a464b81905b77a224ee47 : SUCCESS in 17m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c89516cf86ac4951b75f7ba28f0cba1a : SUCCESS in 7m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5b3773b4adbc4865b8c2ac971561c7c1 : SUCCESS in 10m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/12278cd4e1494398844c29f22d3786a3 : SUCCESS in 15m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9a52468978d04ff399b0902b00da44d6 : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c531dc5186084433b1cd5765bff3a042 : SUCCESS in 11m 20s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/42951a5d387244eea9d099ea1dbad0eb : SUCCESS in 23m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fb0fbe29a85a4f5080af99a1bbc9f4e6 : SUCCESS in 8m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fa3f61debfd24786834bb38867e94044 : SUCCESS in 12m 29s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8bdbdc987a484c9fb730453b6e75e88e : SUCCESS in 5m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cdfb217446304ea3b0780e86d4b5c18d : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86c94745302e42a6a9f49fcc4afcce51 : SUCCESS in 27m 17s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d0fbe2bcbcff4e938d928ec8ba82cab0 : SUCCESS in 7m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/968b41d0c1da4e1a87894d25ee740b27 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ed5c1dcd23a47e7a3e3b2ede195b90d : SUCCESS in 8m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/61864a550fdb4bd69cf6a6c701026a72 : SUCCESS in 14m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/19902f80dbc348dfa49b87e60472fe1e : SUCCESS in 6m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/894349f294514894833303ba98dca6e3 : SUCCESS in 1h 00m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/307fd73b27b449be89c599c1020dbf3e : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f82e5101dbe778a840c0e67f24e934ecaba9aa72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 17:53:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"581030c6a5a0d72aa473d93bfcaad7585a009e99","revisions":{"581030c6a5a0d72aa473d93bfcaad7585a009e99":{"kind":"REWORK","_number":1,"created":"2025-08-28 10:04:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/958743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/958743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/958743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/958743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/958743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/958743/1"}}},"commit":{"parents":[{"commit":"91d4d40cda062c036d4e42390e5d9d032153a6f0","subject":"Merge \"Bump pyroute2 to 0.9.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91d4d40cda062c036d4e42390e5d9d032153a6f0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-28 09:54:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-28 09:54:25.000000000","tz":0},"subject":"update constraint for python-observabilityclient to new release 1.2.0","message":"update constraint for python-observabilityclient to new release 1.2.0\n\nmeta: version: 1.2.0\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I923ee0c8ae1776cb929efb4ea3ab1f8ce71e832c\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Juan Larriba \u003cjlarriba@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8a2e2af7a0e527ff52f118396d601c7b7f79ba79\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/581030c6a5a0d72aa473d93bfcaad7585a009e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/581030c6a5a0d72aa473d93bfcaad7585a009e99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
