)]}'
{"id":"openstack%2Frequirements~958752","triplet_id":"openstack%2Frequirements~master~I6f07c198cb707f41ba88cb6c91d9f4eec86bae06","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6f07c198cb707f41ba88cb6c91d9f4eec86bae06","subject":"update constraint for python-octaviaclient to new release 3.12.0","status":"MERGED","created":"2025-08-28 10:14:25.000000000","updated":"2025-08-28 17:53:24.000000000","submitted":"2025-08-28 17:53:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958752-new-release","meta_rev_id":"1c9b4b28c0104d428152fbca62864ad2f32178d4","_number":958752,"virtual_id_number":958752,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-28 17:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-08-28 12:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-08-28 12:13:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-28 11:45:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-28 12:13:31.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"c26fa8d8984e07241ca50551fd54251ab8f88c56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-28 10:14:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b9f90a0bc54c2668664e65edd621a01bfbfe266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 11:45:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cab9e5e33226402dadd45240d46703c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ced75e8bca3c4915b725a298400f0036 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52885f04515841f99b6db393cd252b56 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4822b10c4b7f40869ab29dedc09a89ab : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff1c79a1ba564cf9b04a6f9cf6a2e407 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3055615a35b419fb22f67ddf0fff776 : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae5e169ac1204406b9e08321f9772e20 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/84efa2a69b734477b5233f54ff73172d : SUCCESS in 2m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b57d4be427494e288a48559ee06e3cae : SUCCESS in 11m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/54e84ca9fb6f4bcb96d37378ddb7c457 : SUCCESS in 12m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3c5e9e74c34441208fa222ee5bb26300 : SUCCESS in 20m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a1eb9f48465d4af28f096062cd17a5f5 : SUCCESS in 25m 48s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/41107d4d0d0444de862ae8266f46ae83 : SUCCESS in 5m 15s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/cb6ee45addea4491ae224cd3a38b1d93 : SUCCESS in 4m 15s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/827ba156df9a413992180436b5319b8b : SUCCESS in 4m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/402b8a9dc3bf47cc85e50a1a5db9b351 : SUCCESS in 11m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f89f004f1a6a480fb2fa7477ea1861d0 : SUCCESS in 5m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9ed74dade7fb4c64b33b5791aa725b0c : SUCCESS in 5m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c9674c3ec0e84e818d6f3321cc500ffd : SUCCESS in 5m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/136f2bfd3bf64b51be9c1a822605bc22 : SUCCESS in 18m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5922f627d0db40eaba0241adeac90a87 : SUCCESS in 6m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5b08fa3401f341de8a219ba38bdd561c : SUCCESS in 11m 25s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/11878a4edda24ce5a2695d94d00518b3 : SUCCESS in 12m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1fe261d9685743a0bbf07be02f173f79 : SUCCESS in 2m 31s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3e33df0126bc4228b101e6d89aa578e1 : SUCCESS in 10m 05s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ba75cb41e7a64cc2803eac22e6c6af70 : SUCCESS in 3m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/966e73f573244aa59c14e818a4cbbcea : SUCCESS in 5m 21s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/bb6ed1a9c8f74ffa9a36d9316cbbc1bb : SUCCESS in 13m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b1be8ec1db5149298f457fd7dd69d189 : SUCCESS in 20m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1ff38fab65ec4eff9a2c0424ee8e53e7 : SUCCESS in 8m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/724c6a74199443169e5340e1996c4774 : SUCCESS in 12m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fee5ece5884c463f8e1786e9ba88da73 : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e270dc651166487986df36e7d174b208 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/32a71d48ac934161b3f7d0c741a79fef : SUCCESS in 24m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2488d2fe2e77422db10e07bd26494cce : SUCCESS in 8m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d342d699accf4c489cb4025c8d7cf9b9 : SUCCESS in 7m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc2ad7d0925f47dfa8b0d0153e9fc1fb : SUCCESS in 7m 53s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dcbbecfa8067491c967d9ecc06ba536e : SUCCESS in 17m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/08cf45a3481044438dced91e12a1f2f4 : SUCCESS in 6m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7985289d6174733b169fd97d8302a7d : SUCCESS in 1h 04m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b381f6da1e94cbcaacaa62bfad504ee : SUCCESS in 1h 24m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2acddf1fd395febd47ed5573afd0c365fcad1a81","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-08-28 12:13:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f1eecbebee71b5fcf07d3e71d658d07980ab3a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 12:14:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e62b519c235c249d2f589b675211406c4ca4ebba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 17:53:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0b5576af2ec4a14b1409db5d08edfc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c775ef6a0f44f2849224c213eb4dc7 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fc3ac06999049ec901a5ed13424711e : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af95d3606a884025bffa1ab98c55c7c4 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f525e1b88caf413eadd6bc61f43f396c : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38e2c288409d495b9f686cb75f258f85 : SUCCESS in 19m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08990b45cf4a485f82f4d8a194f3b38d : SUCCESS in 3m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/23d56e92c76a4af39978f0e24e3a3341 : SUCCESS in 10m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/685d2455db114d9c866272a64e822de4 : SUCCESS in 18m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e23b35f270f64ab78581b7e1a96ec75b : SUCCESS in 10m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/91ae2c94f90d494a9f4711743d598768 : SUCCESS in 10m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5a92469f5052429780621242d548ba8c : SUCCESS in 5m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ac6ac103f52e48419fe0563675ab049b : SUCCESS in 6m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1ecc56c7a84d4638aec73947c18120be : SUCCESS in 7m 05s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7afe9839163046589e3b1fb730c1e715 : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27ed55d98b2442689050ef91d7fab890 : SUCCESS in 8m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1de464b110824d4785031cd9dda862bb : SUCCESS in 23m 23s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/00e5f4df70654738b379f28bc320b618 : SUCCESS in 8m 45s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1cc533c409894060bd651d83c8dfb3f9 : SUCCESS in 3m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5a5e5701a62444c7919821fd0969161a : SUCCESS in 7m 25s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fc71a46a494548f287eb226318025325 : SUCCESS in 15m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/682fe39fb8224119864447e2171e949d : SUCCESS in 4m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3bc90d4b7401435b8758783e70964989 : SUCCESS in 15m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c73c37b073074884a22b5842611c2b77 : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8dbbf887621a4c389d179020f7f6aa93 : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df11e87be8024158934dab85c9551631 : SUCCESS in 26m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/57067b6f659447a695cc353bd9302645 : SUCCESS in 4m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6b3024d2ff1e441fa73a45dc405fd256 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe294a119ae34d41a236e91e462f9b0a : SUCCESS in 5m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1e0563681244487ea9543cd2ff815306 : SUCCESS in 16m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/34ae9cf0b4314e06b54b1c2237fc3b20 : SUCCESS in 6m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99644da763fd4cb78e5fdc7a10b1cf44 : SUCCESS in 52m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9615bfa16e084b8a801e384e12de5b4e : SUCCESS in 57m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1c9b4b28c0104d428152fbca62864ad2f32178d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 17:53:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5316a5a5f7acd58b07b4adf2c41c15dde7478c0","revisions":{"d5316a5a5f7acd58b07b4adf2c41c15dde7478c0":{"kind":"REWORK","_number":1,"created":"2025-08-28 10:14:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/958752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/958752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/958752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/958752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/958752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/958752/1"}}},"commit":{"parents":[{"commit":"91d4d40cda062c036d4e42390e5d9d032153a6f0","subject":"Merge \"Bump pyroute2 to 0.9.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91d4d40cda062c036d4e42390e5d9d032153a6f0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-28 10:04:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-28 10:04:23.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 3.12.0","message":"update constraint for python-octaviaclient to new release 3.12.0\n\nmeta: version: 3.12.0\nmeta: team: octavia\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Change-Id: Ifbea85cc84457819a422727c0ee9a46d48aacef9\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I6f07c198cb707f41ba88cb6c91d9f4eec86bae06\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5316a5a5f7acd58b07b4adf2c41c15dde7478c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5316a5a5f7acd58b07b4adf2c41c15dde7478c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
