)]}'
{"id":"openstack%2Frequirements~958754","triplet_id":"openstack%2Frequirements~stable%2F2024.1~I28be78d0a68bc507190aeaf4b94aa1772ef8fedb","project":"openstack/requirements","branch":"stable/2024.1","topic":"new-release","hashtags":[],"change_id":"I28be78d0a68bc507190aeaf4b94aa1772ef8fedb","subject":"update constraint for oslo.cache to new release 3.7.1","status":"ABANDONED","created":"2025-08-28 10:22:58.000000000","updated":"2025-10-16 22:43:27.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7bb8c99dcdb017912702283a8fcda56bdc54334d","_number":958754,"virtual_id_number":958754,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-24 09:32:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-09-24 07:29:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-28 11:44:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-11 13:18:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-09-25 02:30:58.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"}],"messages":[{"id":"1e6678e7ebaa2fb44a8ec1b5077c30e09749ca2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-08-28 10:22:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d110247bc558775fec0dff820a3371e9982c78b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 11:44:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59dfdbe9618440ebafb088c499e874e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceaa52f16cfb4da087b543ef70918715 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c697ed2d138d4dcb88fde3ce6c312fd7 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08be7804ba9746a6b5225e022a88a7ca : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1366950f19b74d6ab9b0306968d2866a : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18515e533d7648c58e7ca09362588398 : SUCCESS in 2m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/178e40dc097c481191b2e3304b882ef1 : FAILURE in 4m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5a670369fca54c659fa75a177bc1040c : SUCCESS in 9m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6ed32995cc0240a993e708838c01ac10 : SUCCESS in 9m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/94232e24db0c4445bdf0d9cb72178e50 : SUCCESS in 8m 15s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/16a35bec38084c20968c2699ac1e4499 : SUCCESS in 5m 58s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/fc950a9be70546c49ead9e8baa63297e : SUCCESS in 5m 44s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/0d916fd6164542959441115d049f3384 : SUCCESS in 2m 59s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c9e1b9af09754194aa347d881892583b : SUCCESS in 11m 50s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/7ee7dbae56a848b38cd877664d5f0163 : SUCCESS in 5m 17s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6feaf8929e7144b6a136bf93316f92cc : SUCCESS in 4m 26s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/38c03a4e8d6c4652a4a3637489f92468 : SUCCESS in 5m 06s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/56b71ff3cdaa4032a18e5cc3f9c745a2 : SUCCESS in 14m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a654f1b055e42979a43bc6e1ff5f27c : SUCCESS in 4m 58s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/76639eb65b414304bf2de08660fe99b2 : SUCCESS in 9m 28s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e8338e761d1440b381ed4cbbd87fdcf9 : SUCCESS in 12m 04s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/41c017e9cd494b26b2d801f07085da68 : SUCCESS in 3m 10s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1b26cf95b94348658449b4ee7fba8ecc : SUCCESS in 6m 48s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/a88cdf72172740f0a83daa7d66f31121 : SUCCESS in 33m 31s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/991978773c1f4b948b9c3eaea386502a : SUCCESS in 3m 15s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/bf3f2cc2ef8a41b1a3211434c9d70851 : SUCCESS in 18m 38s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/5e9b497a5b384ba2857c3319888482f1 : SUCCESS in 25m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d98f71aee9c44389bcdcff84a019fb56 : SUCCESS in 10m 49s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/af29c70ee05b4ad28d2f4e6b2904c9f6 : SUCCESS in 12m 09s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/52c1362ac717498c9e84914c876e18cc : SUCCESS in 3m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e2bea89dc69745f4ba14831c572c2d0d : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04938ce81f744096b6b2e98e82d4a64c : SUCCESS in 25m 52s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/cc690ced5e75438195ba2e7c212a9085 : SUCCESS in 7m 42s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/6d4cdbe940604478b31c2567932425c3 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c10dc99ff144507857b67110d8daf74 : SUCCESS in 7m 11s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/609c538bc12a4a589d592f900a6f13f4 : SUCCESS in 14m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c7022e361dd435fbc7b8c336d52f23b : SUCCESS in 42m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ecfd5d6c1f14256a36499956094298b : SUCCESS in 1h 19m 09s","accounts_in_message":[],"_revision_number":1},{"id":"28957f95eb7789bdc6edabfc1c7c956dd199bf40","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-09-11 13:18:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa383dc029cb67d922472a5d7ed1c1afafaadaea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 14:47:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7aef57fe03094bc29e8cfa2a771df61b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a826f0d5ac1e429fb67e32bfada3d785 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5db2f2a3e3254c18b4e056796cbdc72d : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8445615e597e4b8dba9bf2bf2dd57017 : SUCCESS in 2m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/444991b50ba2429f98651f8431d960dc : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8dc05e74bb3a4740acd55fc426853a29 : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee55c34002704019b6929fd706e2cf2a : FAILURE in 7m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7e8b48a87605442db01b3942598a17db : SUCCESS in 15m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8dbb710252da4ba384f11eeb11ba1572 : SUCCESS in 4m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dd6bad51dac3435b93cdd13919128d69 : SUCCESS in 18m 54s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/0b91cf79175646e9b917ee8d798a569b : SUCCESS in 5m 18s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/1e8be9ca959e4b6aae4feef145152706 : SUCCESS in 4m 05s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/ed987070bd084b0e8243269c7feda45f : SUCCESS in 4m 23s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4ec10937db024bb8af0ff6cce0023d4e : SUCCESS in 9m 14s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8b041db002e1410fb92263e33c6dcd7e : SUCCESS in 5m 19s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/68a4af172aa64de196065e03a37c772e : SUCCESS in 7m 05s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1dec2c7a76d6407bbb25ecf5158385bf : SUCCESS in 4m 27s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/cd53133f03e449cfa72b098bab0d1c0b : SUCCESS in 13m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d149042c9c1844eb8d6fddf161a50978 : RETRY_LIMIT in 1m 54s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a74c781edb1f4b648a7e320bc66bdebd : SUCCESS in 8m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/6431f308aed34164a71f7fe963843d6a : SUCCESS in 13m 40s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a46a4126733b4dc5b05f0956234fb1f6 : SUCCESS in 3m 51s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ef89655e1ab947af83401e0c620d4a73 : SUCCESS in 8m 41s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/2a27900e8283432a9552e77475bc4351 : SUCCESS in 4m 46s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/bf33dd9d374743ed908d6404ee4371fc : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/2ecb3bf0d9894f96b8684fd5a0fa57b3 : SUCCESS in 12m 22s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e14f6c3865d243779ac11e5c81c7c321 : SUCCESS in 20m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7be2346711894143b899ce2ddef1670d : SUCCESS in 6m 26s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/88fee7e14e214c5db688ee188f1bd29b : SUCCESS in 14m 34s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6911fcbf62a147bc99ce7f4e784bc20c : SUCCESS in 6m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/90d1f3c2f584478b9e4659781758ebcf : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0c0a06e9805442a8b8d0c4c9adb78e4 : SUCCESS in 21m 29s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/f6136fe80f1a400180116d538899a13b : SUCCESS in 6m 26s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f8ec5091844d4ce290ce10ec569ffd60 : SUCCESS in 3m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c4b631557f94911885610bff7448c59 : SUCCESS in 6m 48s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/503e4f9439e4435bab47300919a4f58f : SUCCESS in 14m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7e7231e342e420eb9a0ec017531ba02 : SUCCESS in 1h 01m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a98f4598016482ea88aec32c38e0f4f : SUCCESS in 1h 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6ad7f0761ec0d116cc9f3083a18bed2419733016","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-09-24 07:29:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"77f8347ef2a0eaec5664a6643146957b621a9286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 09:32:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a70e6cbcd5c4ec8b4ff5430a182a26c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0e9c33f4a194201bbe87adc6b087474 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdb77b0aa59446d288175c60e81eb8c5 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd3e87d7464640c3aaaa5e8d47b3090c : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/effe2250f05949d381e70e61b240318b : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b7b626e884e42f58a3b1ba9308144fb : SUCCESS in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d4404d3e4fc641fa8bfd312c12a8b7bf : FAILURE in 8m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e6824ee38bc2499c880d0b21f3637e04 : SUCCESS in 13m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e175d8b646c240d3ad992a4c4e3e82a9 : SUCCESS in 8m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4b37ef6e4172406a95bf99744c97a05c : SUCCESS in 18m 44s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/1cae0ce8e03c4dea85fe4bf33126fc33 : SUCCESS in 5m 07s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/6473e7f3aa0346588fd2b18cf2ace9ff : SUCCESS in 5m 03s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/e962472829a34d0ebd829be64367d054 : SUCCESS in 4m 25s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/0a11c71e47ab404abe5ffdbca047c5fd : SUCCESS in 10m 43s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/111c2a6b7a7d44e792871cd7ab33283f : SUCCESS in 3m 06s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8473ff8d291f4148a78b3139010c6048 : SUCCESS in 6m 56s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/3009a559241646a0b9e1a42c8852d2bd : SUCCESS in 9m 00s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/5e3ffee6586d4b46bc8cb5f7b46929cf : SUCCESS in 5m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c63ad2e94e7e466ba0c6efde52ed8840 : SUCCESS in 8m 20s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ad186d7e0f9e4151a7e87b994a8f3327 : SUCCESS in 5m 58s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/1885cc085f754428970448b327562141 : SUCCESS in 12m 18s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a68367ab3ddb4839abdb87427068ca7b : SUCCESS in 3m 39s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/354e2dd5bf2e479e90b5bc9b633f0d00 : SUCCESS in 8m 58s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/a3f5fa46b9c64f0cbf4a03cd8142a6c5 : SUCCESS in 6m 19s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/64870cf0edd34a1a86fe3ca858088eeb : SUCCESS in 6m 22s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f6abefc1d876467fbfe4ef90e1a11084 : SUCCESS in 11m 03s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a29e1e7dd9f848678d34c66755ca789f : SUCCESS in 16m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7e256ca59e154e80a90020ecb134e8d9 : SUCCESS in 6m 37s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/3c1b361393924470bee3ab451ffb1517 : SUCCESS in 11m 38s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/844e965bd43c425280c62782019726b9 : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d1658f25a23840e9b8b12b519e02dd8b : SUCCESS in 2m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ebca088713a4de5b59ed8677c5d4581 : SUCCESS in 22m 59s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b7a8e0002ba34b6bae8492ebceab786a : SUCCESS in 2m 37s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ad54b5bb2c1944178d07f1216589847e : SUCCESS in 3m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/664255c84a20462596fcc4f45bef95c3 : SUCCESS in 5m 36s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/a47e94540d9f4866a433be4f44a7db71 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c47e2fdfa70645eb885c1d1a37637dfc : SUCCESS in 42m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bc2a7f79ee541f3bd73589290e13a93 : SUCCESS in 1h 53m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a36522b0d52aff7701d8c95961db3c0595828314","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-09-25 02:30:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5212ce8fcb4c7582d925a23ef1271dd9df5a00e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 03:53:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68837e98b6074ce79bd24ba46c427698\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60029f934c44614a952defec3299a64 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab5a176affc241a18afadeaaddb13d89 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ef6e01b20f94c599fb5608262e54e88 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97116f6c464246cfb10271ba80d7ef28 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/178ff90d84f643e8bbe824f6d69a8ab6 : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d42282f29ee043ce9edd914271e87d73 : FAILURE in 8m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d34a3ddbd1a74778947cff75e0a7c862 : SUCCESS in 14m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/53f691e43899477dbb0478139c0861f4 : SUCCESS in 16m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0336c95f9cde41b9b81a275f62383bc5 : SUCCESS in 20m 51s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/f816321907164ad1ba0ff571a2b11fd2 : SUCCESS in 5m 54s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/aee0d7003e564e879f486c80aee14f42 : SUCCESS in 3m 13s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f537fcba8a4146819e42df3ab4d14f83 : SUCCESS in 5m 47s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c0344b49995f49eb9c199d468fcc0d47 : SUCCESS in 14m 54s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b4ad952b84e243858d9cb07f2c02e516 : SUCCESS in 6m 45s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1ba2dce3b03241f1beed5ec7c3beb0f9 : SUCCESS in 7m 38s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/37f674248a8c46ccb65ab142578c37d6 : SUCCESS in 8m 37s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/dacfc992c7e84dd08caa0f265b11a1b8 : SUCCESS in 12m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/28febb78ee5c4e93a5e41baea667a3ec : SUCCESS in 15m 22s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/e6ead1ee340f410bb9b065de2eed8e47 : SUCCESS in 8m 59s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e72abad7b60e4691ad9035c519e60528 : SUCCESS in 10m 21s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/764d3ad97658468c8dff350086d8eb28 : SUCCESS in 3m 42s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/154e391922c9435b8e13ef984a2509ca : SUCCESS in 8m 10s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ee34627620064e7badb005b9416293c7 : SUCCESS in 6m 17s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/d72de3650e904438987c674db0e957b7 : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/c32ab618b9f04c75a72dd87f0deabbb9 : SUCCESS in 10m 37s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e425dfcd1554471795554f533ae7a4c1 : SUCCESS in 24m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e50ad3fb07a94eddb65bfc422e8ca4ed : SUCCESS in 3m 47s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8320b19e958a4e6591e2302465b37121 : SUCCESS in 13m 58s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/a4e7e6664ae747cd8d9f492952981e8d : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c6f9c5c677e9423fb163c49786c9ecc2 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96f9c2eb481e40658930ae93f08536c2 : SUCCESS in 23m 07s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ec614f637475476bb212cd2ce9529677 : SUCCESS in 6m 18s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1f6d07073cc44fc3937b9c3875dfe72f : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84e6b6a8530f48faafda819da58296c1 : SUCCESS in 5m 14s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/19194c8a04174d8399a361b7a8f7bb90 : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/851d38f0b41249f0b6262562e9d3e64e : SUCCESS in 55m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/272036a4d19440938c7a9542dd49cf21 : SUCCESS in 1h 20m 31s","accounts_in_message":[],"_revision_number":2},{"id":"7bb8c99dcdb017912702283a8fcda56bdc54334d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-16 22:43:27.000000000","message":"Abandoned\n\nDue to[1,2] these wheels aren\u0027t compatible with pkg_resources() in python 3.[89].  Changes to out PTI[3] and re-releasing these[4,5] are needed.  As such this change can/should be abandoned\n\n[1] https://github.com/pypa/setuptools/issues/5051\n[2] https://github.com/pypa/setuptools/issues/4853\n[3] https://review.opendev.org/c/openstack/project-config/+/964251\n[4] https://review.opendev.org/c/openstack/releases/+/964252\n[5] https://review.opendev.org/c/openstack/releases/+/964253","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15497d5b5359bcc3ef88dd9971533c8fb42c20dd","revisions":{"d661826b4ec9c5ddcde8c4987cd4ef55ab323fc2":{"kind":"REWORK","_number":1,"created":"2025-08-28 10:22:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/958754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/958754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/958754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/958754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/958754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/958754/1"}}},"commit":{"parents":[{"commit":"a409487a0904de521d1ee034900320a2144356c1","subject":"Merge \"update constraint for python-manilaclient to new release 4.8.1\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a409487a0904de521d1ee034900320a2144356c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-28 10:12:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-28 10:12:55.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 3.7.1","message":"update constraint for oslo.cache to new release 3.7.1\n\nmeta: version: 3.7.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: stable/2024.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Iedbcf0e3e02e1f516364ec535e64e7ac818df378\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I28be78d0a68bc507190aeaf4b94aa1772ef8fedb\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d661826b4ec9c5ddcde8c4987cd4ef55ab323fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d661826b4ec9c5ddcde8c4987cd4ef55ab323fc2"}]},"branch":"refs/heads/stable/2024.1"},"15497d5b5359bcc3ef88dd9971533c8fb42c20dd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-24 07:29:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/54/958754/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/958754/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/958754/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/958754/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/958754/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/958754/2"}}},"commit":{"parents":[{"commit":"2a6f714e459568d7d0bbf3cb9606bea28b338464","subject":"Update horizon cross-npm job to define a node_version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a6f714e459568d7d0bbf3cb9606bea28b338464"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-08-28 10:12:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-24 07:29:47.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 3.7.1","message":"update constraint for oslo.cache to new release 3.7.1\n\nmeta: version: 3.7.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: stable/2024.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Iedbcf0e3e02e1f516364ec535e64e7ac818df378\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I28be78d0a68bc507190aeaf4b94aa1772ef8fedb\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15497d5b5359bcc3ef88dd9971533c8fb42c20dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15497d5b5359bcc3ef88dd9971533c8fb42c20dd"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
