)]}'
{"id":"openstack%2Frequirements~959054","triplet_id":"openstack%2Frequirements~master~Ief1cf8bb83bd2f2a7c61760a8bc1ed65d8b92545","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ief1cf8bb83bd2f2a7c61760a8bc1ed65d8b92545","subject":"update constraint for python-novaclient to new release 18.11.0","status":"MERGED","created":"2025-09-01 13:49:33.000000000","updated":"2025-09-02 13:29:58.000000000","submitted":"2025-09-02 13:27:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959054-new-release","meta_rev_id":"8462e33cf80d264d09dc6389691f404f0674b416","_number":959054,"virtual_id_number":959054,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-02 13:27:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-09-02 11:46:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-09-02 11:46:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-01 14:52:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-02 09:13:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2025-09-02 11:46:26.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"ff34669a4769d516f6027922da00285e347dd3fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-09-01 13:49:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af8cdc628e6a175d0db58cda183b6d39ceaa1115","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 14:52:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b386c62911d416a86b17ef5e1c7d946\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385c2922170d43cfb885e1df72d5be3c : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b71c03315fa8462aafd524c1e0036dab : SUCCESS in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c0f510773c8488fbe39e8ac42ed317a : SUCCESS in 2m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e9661ba7e7749589f6a941617459643 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55c3b85bd20040fd93ab99dc4a0cfba1 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01a8296dfb9242d990c243209d54d58b : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51681a4df56041ce86c5f43259ffebe3 : SUCCESS in 3m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/21c606678c2b47a39f31a2733d9ca65f : SUCCESS in 4m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6cb13290acdd4023806760d80682ee4e : SUCCESS in 7m 22s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/12fd6fe571284159b5beccbfe2071cdd : SUCCESS in 20m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/baac361f34fb48909cc92e8a503b9909 : SUCCESS in 26m 16s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f95d56d2ace6446b9a3d23fd77af67af : SUCCESS in 4m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/cd3aa7b9adf14142b324b72532fafb31 : SUCCESS in 3m 09s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c1d8fd49ac494093ab6adb6579636273 : SUCCESS in 2m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8ea3a53fcd12456da4cf022d84949c44 : SUCCESS in 11m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5ae7654e697c4d9cb67b30a3eb2d96d3 : SUCCESS in 6m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/de52484cdf1a4c0dbc91febd6160362c : SUCCESS in 3m 14s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/92f28392dda14e33961502e0cf3e4838 : SUCCESS in 7m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/420bd3b329a34e2d8ae1470a1a65f8c7 : SUCCESS in 17m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6682f95859d7491aa4dcff8efab07f3d : SUCCESS in 6m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1a109081705f4c81ab7f2c201b20b818 : SUCCESS in 11m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2371703aa14f462c8d8ea53dce665f7e : SUCCESS in 7m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9fdce2d1e4214a18a01d4bf44d181322 : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/54faf3ae35db4543b6cd3fdd3c506f4a : SUCCESS in 9m 13s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/2cdaf11648244d05ad6926458d5d0c9d : SUCCESS in 5m 03s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ee75eeed149c437c9545a67aef7ba6d9 : SUCCESS in 6m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2094822fd7514def909b0ddab8e5d6b0 : SUCCESS in 13m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5f6a10efab254ac4b4e0777b27f067e4 : SUCCESS in 14m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4e303d288ebb4fdd8bb68c565cfd5415 : SUCCESS in 10m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/955f255188984cac90fb6cfd4cf4117d : SUCCESS in 13m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/86b64791a91a4e3586cbad33514420cb : SUCCESS in 3m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de53370138684e838a4d1a03568fe112 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67cdd3a1c5c34fac9cef67c71cc5bce7 : SUCCESS in 25m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2a10e404c2104374b83aaee0d0a32bca : SUCCESS in 3m 59s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0e85aa509c29458ca9b0724b0e52b0c9 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/562e176268404de38ca856dc7f0458ea : SUCCESS in 7m 49s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/eb26c01f0daf4149b17796ddf65a712d : SUCCESS in 17m 24s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f7104af4bd324fd3bf6b20af8eecceae : SUCCESS in 3m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e6312611afe4f2c98bf2a5f5b2d491e : SUCCESS in 57m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a90be42306d84921b370c21cd53e87bb : FAILURE in 47m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dda2857fea5d8bdcd02dc942438c8d8d05a291fa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-02 09:13:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bba3f100315734a6b15e3796cc2a2a1e97d5b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 10:52:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b9693a52d440dcae6ae94deacdb1e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f5dcb27fe1438ba01336dc084db8aa : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae2aee48c4fe4e60a31c538045189cac : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/575d4d52e3e247b3bf3c5798b0572ce0 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b0d48e8ae6543dfbc75ec427e34beb6 : SUCCESS in 3m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bae3346e99b449409d09c4f704418d30 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50d4da872b8e46d99363a1cb63431c09 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92632d0f4ecb419886cf6db256d2e34e : SUCCESS in 4m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/605ebcf934fe4e6d89848d9ca5011a1c : SUCCESS in 9m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ed6dcfdf31d842c3b1a80d2e83abda87 : SUCCESS in 19m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/037cefd6ace34a688adf1823071c48f0 : SUCCESS in 18m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/aeea41fc2c404281be27f2233d19d86f : SUCCESS in 22m 13s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6c6f1bfbc3124d67aeb06f4015b6cf41 : SUCCESS in 5m 28s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/badf82398f5e4982a7af22f7c8aaa670 : SUCCESS in 5m 16s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e56fb949704c4d0f8c9e0151a47c3e1c : SUCCESS in 4m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/885e765981cd4b67a477d9abe96cf5fa : SUCCESS in 10m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/aa781893bea149e991a1b71070821912 : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/55a6d71932714220a39f5efeb3a45ed2 : SUCCESS in 5m 35s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a7a46c490125444f97701c0959b00d06 : SUCCESS in 7m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/914c5af50360438e88f093d45a323134 : SUCCESS in 14m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b4bd10d271d45c7bb327ce5552890d3 : SUCCESS in 8m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/373592aa8aee45a1820480cadb0ca986 : SUCCESS in 8m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/00d8ac096583470b8c89044a98971ae2 : SUCCESS in 13m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f888a99beb2c4b73ab9220910d9755d6 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0bd0af8248eb47d3a27176a2f9a0a926 : SUCCESS in 10m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0dc09259df104c23a43cc543b84e20a3 : SUCCESS in 4m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/863763aeb3fb42c4a9fed56966cb67ef : SUCCESS in 4m 56s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8b424f91ab1d4be99a4b3aede553eca0 : SUCCESS in 15m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/bbf521e6c3634dc0858db0f533394ba0 : SUCCESS in 17m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/74890e3e795d44828b0d214eb057740c : SUCCESS in 7m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/81041c3437f14aec8a5490a47863907a : SUCCESS in 14m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/77e6bab42b7c4661bd601d67888584b3 : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e16b3f04899f4d8392e689bdef498c39 : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04df8e5a97d44fa196ac1b46849b68f5 : SUCCESS in 26m 09s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4a83dd6d5a0c4c4f95fed75be4677fdf : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7cf531f228a443789a58fad72172f905 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2b777964c4f4b1794ff2fe7b56f969e : SUCCESS in 6m 30s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/44657c0e77cb4045acf734c7a29b3277 : SUCCESS in 13m 52s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9550982e91f8494f8efe0dbb17fc04bd : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd32ec5ace7a4aa1afa598119917a97f : SUCCESS in 37m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49eae4da9aa74024a801ffbaf8c92fc5 : SUCCESS in 1h 33m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6644b62bc0f24dccda2ce05df5067a4e4b053c96","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-02 11:46:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5eb99110bf982e2e45cc0f935a212b3746579e72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 11:47:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbbb16a892fcc4079b8b6ad6bf1149cc05e1e4c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 13:27:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/140d72461a594424a87b18d09823e3ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046e5521f1c34422ae87200701d62ca6 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38fecc0f282b4a66acc4c68a4b85fe67 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1be9a3401de944018ab9a89dc1bc7383 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc62242112ec4049b7e00fd43573b663 : SUCCESS in 2m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d70cc789361543c9a81cd9c7870f6371 : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4728842d39bf4e6a9a474f421982e5f5 : SUCCESS in 3m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/839d7820b72949758246a8e485bb5fb4 : SUCCESS in 5m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b02fc6d1be3f4c09834a7317ce076f91 : SUCCESS in 7m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cf136f7d8e76401ea05cef206166c114 : SUCCESS in 8m 15s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/d947a8c1684f4561a30331052d5bb1ee : SUCCESS in 11m 17s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c9061dd9675948ee82fcf8287be0b4aa : SUCCESS in 7m 11s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/dcff3eec20fc49549588e76c4f345c07 : SUCCESS in 3m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7360211a03a64e669dc85ee6bf5f0bcc : SUCCESS in 7m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/097a08d02f0a4e6188093d7335495f3c : SUCCESS in 16m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7193d7b8813d412b9f02d51ea75c73b7 : SUCCESS in 7m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/61bcefcb03cf42168dc2a2ec70cd2d60 : SUCCESS in 8m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bc68e33212164f5fa2170e1bbc6f3487 : SUCCESS in 11m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/23cd36d67c634dd88afc7a1c006536f3 : SUCCESS in 3m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/28b43e1665584eb5a54026815a9ad9e5 : SUCCESS in 10m 48s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5f569404360e49349045d9f7c917435c : SUCCESS in 31m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/45a67ba1845c43a4b9d4d34f04e2371f : SUCCESS in 8m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/217a714f12784b8e8f79af08e929ca0a : SUCCESS in 15m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d04a7fd595044687a5c48ab8706ca880 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fad2f3436a6f4876abb0205b764f19c7 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/afb366222f9545fa8f9bf3bde1e09ebf : SUCCESS in 23m 45s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4d0754afc6434563b1caadde26464f56 : SUCCESS in 6m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b8b3ecc2917e4845b9e5ed0254787242 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0c752f896d2d407d8d0cddcb677be6cd : SUCCESS in 7m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b2913916737d4accba0ef6e7f20f84e7 : SUCCESS in 13m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/790173d596374035bccfdc1a669266c9 : SUCCESS in 6m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c78252f8cbc41db90a4231e9f149e96 : SUCCESS in 56m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3aae8bd0463462799d759c1bb60fffb : SUCCESS in 1h 20m 56s","accounts_in_message":[],"_revision_number":1},{"id":"404adf68301f7e6970fd414706b09206829887f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 13:27:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8462e33cf80d264d09dc6389691f404f0674b416","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 13:29:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d3f228e768c4ba496870fb60607ee4d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20ec260c4024617a23864b0de8594af : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78c75cc2759fdde0fbc594b44576c91567d5f9ee","revisions":{"78c75cc2759fdde0fbc594b44576c91567d5f9ee":{"kind":"REWORK","_number":1,"created":"2025-09-01 13:49:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/959054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/959054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/959054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/959054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/959054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/959054/1"}}},"commit":{"parents":[{"commit":"56798df8bac5995e09a8bf3ce8820e0d4f8710fe","subject":"update constraint for futurist to new release 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56798df8bac5995e09a8bf3ce8820e0d4f8710fe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-01 13:39:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-01 13:39:32.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 18.11.0","message":"update constraint for python-novaclient to new release 18.11.0\n\nmeta: version: 18.11.0\nmeta: team: nova\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie3c350c309e1eb98017a380be3e9a18c1e67d5ce\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ief1cf8bb83bd2f2a7c61760a8bc1ed65d8b92545\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78c75cc2759fdde0fbc594b44576c91567d5f9ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78c75cc2759fdde0fbc594b44576c91567d5f9ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
