)]}'
{"id":"openstack%2Frequirements~959062","triplet_id":"openstack%2Frequirements~master~Ib69509cb3b4f0ee9d8c425956e2511fb0452e8b9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib69509cb3b4f0ee9d8c425956e2511fb0452e8b9","subject":"update constraint for python-ironicclient to new release 5.13.0","status":"MERGED","created":"2025-09-01 13:57:53.000000000","updated":"2025-09-02 20:22:47.000000000","submitted":"2025-09-02 20:21:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959062-new-release","meta_rev_id":"f30dbad9e6791d87983bd6f751a5b1c83d1ac102","_number":959062,"virtual_id_number":959062,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-02 20:21:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-09-02 09:12:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-09-02 09:12:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-01 14:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-02 09:12:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"6d6b923135f5d86b6a95c73e03b46f51b99b73d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-09-01 13:57:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"999be6deac66d759711c746a9d99710af629cd36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 14:59:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff6e82a6c3894deeabcfb2dc85079a1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2800a733f7f40d7aa03b56840e4f7af : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2885e6a7d67d4e20819d0e9a7031398f : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62ccbeaad35c414abb30f42eb8526d9e : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f179911363e43e5a9ff2c12a1d16d02 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3ee68bd4b59841e09b034033d9ca1e00 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69dd10de52d94e288b00f27605d290b8 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2f306729307441ed9e863dae73cf6f72 : SUCCESS in 4m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/81385aa1dc5c43b7961cb9f7a6ab9fcc : SUCCESS in 10m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/826ca3c1caf74cc4b7a68bb442fd1293 : SUCCESS in 16m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f5d68676e39c494db37c1b86ab12a0ca : SUCCESS in 21m 17s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b7c33b49f1584b89b88a0c038721a1c1 : SUCCESS in 9m 59s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e8ed762fde9d49ffa7eab7df5806fcf7 : SUCCESS in 4m 44s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/302911e508124795bbb9b547930c43b0 : SUCCESS in 4m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2041232e01d94ac5bb3092c14d766fa9 : SUCCESS in 2m 46s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0b4cdab4124b45e29f5e920243b1343c : SUCCESS in 12m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0c71441541a54878a48b695b104b8b3e : SUCCESS in 6m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/632c574f5bc04cd5aba9fa38e6c57b27 : SUCCESS in 5m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1073b294c51545b78ea42d9d18ab4c98 : SUCCESS in 7m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6c916043a1c846bdb86042ef147b2fa4 : SUCCESS in 8m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0cd93bfc2e84a9592c64b4cdb261394 : SUCCESS in 4m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d3c973d2b68b4274b75f578d312edbc6 : SUCCESS in 5m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/36a1668af91e4f33b16334f5782d8d74 : SUCCESS in 7m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ab33566e36d54d8faee553f3677b66d1 : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ff6013effde5458aa2dcb487fac250a1 : SUCCESS in 10m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fae908ddcca146cdb9e45857065d85cf : SUCCESS in 4m 40s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/74322b5cbf5b4c18aca403f6e1b84de5 : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c21d3a0c81594dffbfc29cb1bcfb4789 : SUCCESS in 14m 31s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/bab42e020e0940deaa1baa903aee4560 : SUCCESS in 26m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/79555c6576964c58b6d72bb60ffebeca : SUCCESS in 5m 36s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/72dea87f4b844ad5b6f5a1f536573efd : SUCCESS in 13m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e65d301ecb074d1baa544ab3818818ca : SUCCESS in 6m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d3bedbf2f2e48ea847f7db44d69e1e2 : SUCCESS in 2m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f268a96b3f92448b825589bdb8da0ee5 : SUCCESS in 27m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ad5b2140f6ae4e09aaec81cb4a1915f5 : SUCCESS in 6m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2e2c12d1d3d84d9bb3cf503592cf550d : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f35c2b5980534a5e9eac86ad3dc6bb8b : SUCCESS in 6m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6a8c72199c3f4e0592473fc44bac994a : SUCCESS in 19m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a2af9d5e955b41dda53196df0e0bdb68 : SUCCESS in 3m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e30d81df1da54f83b3ddb34a714b2926 : SUCCESS in 46m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5875022a5a174b3cbdf66b8c718d5fb3 : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f8bdc0f699ead56519dd462ebad16a322db84d7e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-02 09:12:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4c6d08a004d4c119bd1505327b601911dcebd37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 09:13:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dca99726001a9f1d3ec4e710b7bae367f762357b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 11:05:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14cc8e9b8ee04105b76ec4ddc398e10c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00cfabf86cf843e7a8442dad8ca44bc5 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66116414d9794a3babe3df9d4a32953c : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8990195fbd7d4d0aa24c5a883ab9db63 : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b19f2882810c4546b7e7b8b710db3d11 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0619a6499f2432bb090c9cbeaeb040c : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3bba0a9dd27f447f8dac3ba05e141c23 : SUCCESS in 3m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/238d279867c64e2ab31ab0257148c96d : SUCCESS in 9m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9d2a178897254769ba657aabcdb51e26 : SUCCESS in 13m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6dd89765e2ae474c83f80b76ef33f3fb : SUCCESS in 19m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9a5b82ca79764b60b84cbf69e8671134 : SUCCESS in 10m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/efd1738b2cbd4a7daa307546804ff7f5 : SUCCESS in 5m 27s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a2985ff366b6404c91e7d111d31ac7c5 : SUCCESS in 5m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a4db6519d0e7424193c77cb6b878a1f6 : SUCCESS in 8m 33s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8712b19ac1b94f49938b86951e613cf3 : SUCCESS in 17m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/370a8957f55b4ce391400b3563ca0c7d : SUCCESS in 7m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f9e623818162448ea73e347520005874 : SUCCESS in 6m 44s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2b342b5dc1b04be9b6da4d64cac4b651 : SUCCESS in 10m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c9d88d947e6b4c6e8717770425095079 : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d06b132b855649df8d52744285a3fefa : SUCCESS in 10m 55s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ed8a7f3fde27410880310afb3141f432 : SUCCESS in 14m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8b7dde62f1e64761b93a9da2001b695e : SUCCESS in 5m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/146bd6d81377425b9e3c9b916bc2d838 : SUCCESS in 14m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/45e6fa75ba36453ba91d1cbb9b7ff574 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/25f759e056134c73abf34742bfab51f5 : SUCCESS in 2m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8bfe2f57099a4840a7a62b988341a926 : SUCCESS in 24m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3a117890919b4fe28a91699927083982 : SUCCESS in 5m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/66a02322fcbf438aaa13db923ad07692 : SUCCESS in 3m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eca1e72155a64214b041f4533df7e186 : SUCCESS in 6m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c46b4cfc4f0648659b548fffe86cbf6a : SUCCESS in 18m 50s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/62419bdbfef145f8a9f90ba9a0fb185a : SUCCESS in 6m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab5eddeca1af4a1fa67c185a97c8b065 : SUCCESS in 47m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4be9457350694068b7b8b0cfcdced0e7 : FAILURE in 53m 05s","accounts_in_message":[],"_revision_number":1},{"id":"67001ed4fe21c5d49244423750c00dd8d2482c47","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-02 11:45:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"528a3ec0e73e1f1b0d3199ef2e14a7541452de6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 13:28:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bbfffcd5ee642a49ff3f8fbf567d0ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8acb116e4f4f4fbf89651f3144b1e4d1 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbefde9d49b74b47904afa29b0e66af0 : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23e93ba54844452ba316296178c843a6 : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c12b91cb31074410aa6ce9c921f9aedd : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c1d5b8e84004f9d94d977eaf1970bd6 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09f7ceffcea74d439f45eb36201499c0 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c9caecc51674e94a49f8286c0a44983 : SUCCESS in 4m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4799be5b3a3247bc832e483cc0c77749 : SUCCESS in 9m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7266760c2d0d4472b3a774574410a89f : SUCCESS in 17m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/87417fb1c303474ba82b97b3d0e2f7fd : SUCCESS in 18m 47s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/baf6c1c8c4b14f118e40804d6db14477 : SUCCESS in 24m 26s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f7cfce623e2044748d540d363581f664 : SUCCESS in 5m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/98b1213f252c47d4974721c4230865c8 : SUCCESS in 5m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ec72eec76024466ba274cd22a1a5638b : SUCCESS in 4m 32s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a11faed5723d4fb38eadca17687a31c4 : SUCCESS in 8m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/77693ac3919641348f0d4f878948319b : SUCCESS in 3m 48s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/74829a047da4415680465a918906ba94 : SUCCESS in 5m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/34709218127d4c81bf78fce065d1f0ab : SUCCESS in 4m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/98a78b00d93748d980703ecd2ec66c5e : SUCCESS in 15m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99fea682b33f435a90c8f73d6d385124 : SUCCESS in 5m 18s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3a17c3aa6e34483c8d58f57c528cb628 : SUCCESS in 6m 00s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3bfa851efec44d8dbece60dd10b808f3 : SUCCESS in 9m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8a05fcfe2d97446dbc56cca43cdd1787 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cebde81ce4024d03abf5e8d57597b9bd : SUCCESS in 6m 08s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/54a5f6e13f734337bd72faa751ae7641 : SUCCESS in 4m 27s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4983180d69ff45caa985cceedc4595da : SUCCESS in 3m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/cac7e2c5f5a24a7a9ebdf47cd42702a9 : SUCCESS in 12m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/177f2ec7b1ef41c6856a106e5bb71ec3 : SUCCESS in 13m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/70f486e532e140e997b8665150a87f8d : SUCCESS in 6m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/89bfae0dbd8248f08711a8e762d7108b : SUCCESS in 13m 51s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7cc22378a3584dc5b3986e9d85e8f18e : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/53cede3ba4b2442380cbdfca030c3af4 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b2913e12fb049c087ec12f3c5e6300e : SUCCESS in 22m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7ee9d9f3da91487185644771891f7aa4 : SUCCESS in 6m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/beb273c61ab949e3a3a62dd401163856 : SUCCESS in 3m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15ced087d16842eb9f41477b1b88b0ab : SUCCESS in 6m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/097e2c4264294063b8b84d3f33ed77d6 : SUCCESS in 6m 41s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e6816dd653904063a6cc15820cbc2484 : SUCCESS in 7m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3aa042e3b9a64fafb994dfe6b84f86ff : SUCCESS in 53m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a526dcf30df4ac9a7e22033be295398 : FAILURE in 1h 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"00d7b7af5889efa4f54ea4805481f7b75a6e1393","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-02 15:54:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"877d7fcfd70487b88ae0aa218c453a56b4216c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 17:28:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70180ed1343448a7be9f582df0a3254c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f6f1c3cdf054d10989c17f77741c704 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f68652b25cd941518ee5c34791aefa2e : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d8af7214d084a46aeb79b53b4dc5e6c : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce091e6fdb4644ebbc0b726e65e7588a : SUCCESS in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cacb77823b7a40fd96e4507841106ffd : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/034c7cb7be434f22851bdb7dda0283ae : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/279e2e1a0e024a848dfb8961e5a126cd : SUCCESS in 4m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/347086aa588f463db594d298eed807f2 : SUCCESS in 6m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5bc6589e7b5541c3bd48239362c1e253 : SUCCESS in 18m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7640f3d7c189443896dfbf38b5027b22 : SUCCESS in 9m 57s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5b5ae55a6a324c98bb0cd976fe3cec08 : SUCCESS in 23m 45s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5394ba4ebc9749debb0741c1b29d79f8 : SUCCESS in 5m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8ffc727c295847039062ed43f9b5a1f5 : SUCCESS in 4m 44s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/462c6a869c0940ea8aff29eab3246373 : SUCCESS in 3m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/053a8cb3a0e647029632963f584e7387 : SUCCESS in 15m 11s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/27b8908ecfc44cc98577d0e4bebdb1fc : SUCCESS in 3m 53s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f49b775cf95540b1803c9101b574171a : SUCCESS in 3m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b337e580024741679b8515cbec69911f : SUCCESS in 4m 00s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/28c0f888acaa45239ae658ccdecff16c : SUCCESS in 14m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7728bfb2ad242f5995096af8a8e21cb : SUCCESS in 7m 10s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a51e809ca73c406095c7af6cffa95544 : SUCCESS in 7m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3e5a11382218447e891fd034ef2a61b5 : SUCCESS in 13m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/980ea48a8b1140cab898db41d9609d3d : SUCCESS in 2m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/48146c1753404f29aa1bcbb5d1e9a4ef : SUCCESS in 11m 29s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d5b3940349414dc2a4658e5d94407102 : SUCCESS in 5m 08s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/845a070415f246c68e5ec7d4a09f27ae : SUCCESS in 3m 02s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/980a38b11d2c4873ba7c2ff359350cd7 : SUCCESS in 10m 47s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b9f4155279594569a0f9d6095eeb3580 : SUCCESS in 16m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/59909e0eb90f46948d00f4e356e0f8d7 : SUCCESS in 6m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/97bd9607788542ceba8f4015909165f1 : SUCCESS in 18m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/62d328ae782a4e328a85884317524a29 : SUCCESS in 3m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f79e77b2f6244a4d97ddefb4aee567d5 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e16795de09d2476e934529bf84e855b5 : SUCCESS in 38m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/89f504d4f8b6474590b0b347dd954ab2 : SUCCESS in 3m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ddb474f3765e478994b708a6e3e75567 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b80c249549174942b7768040bb09af30 : SUCCESS in 3m 26s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5f4fbab6bde04a1d89eb97d70976af9a : SUCCESS in 17m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/545b6fbdcad445d1a5373071d83e1cf2 : SUCCESS in 6m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71133ed27c44475cb590a038d044f6fd : SUCCESS in 1h 03m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c883b86be17f42d5b692a15760e38222 : SUCCESS in 1h 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6605b852ac746ea68db05bdf3a29f045798d7ec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 17:29:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"745e7bd3537c4f9a9003d9ae1c51e7d611572fe4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 20:21:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c345de26040e491bb0e1bf57e36915c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39930d581d4b477a9cdb48511d505f77 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383e5e693d534bd2b2b878c6d5b15587 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f114a90c420490bae7f581c43f4d9ab : SUCCESS in 2m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/918519bfb39d49b38c6ad4efc3ca6b95 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/541d951d3db54cc5ad711dbde938ba0b : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/10d90d184a9345bebdc7c4dbd28612a0 : SUCCESS in 2m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ec8be8e2a5ce4033919e905d7be72d94 : SUCCESS in 11m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9641474c2a4a41bd8673eaec6b470991 : SUCCESS in 17m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bc90b26e38834bba88b2417cf2634c38 : SUCCESS in 8m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5cc9efd43eaa4b42afc9df9b472b9f76 : SUCCESS in 10m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9ef37ad2e11f4710a1f0e277a3d1ce76 : SUCCESS in 5m 25s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8ec91737389e4fd9b43416d668a5ba44 : SUCCESS in 6m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b93f247e79b948aa8709053f9239a755 : SUCCESS in 7m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9e891e47d7e849e68bb9a579bad545a9 : SUCCESS in 14m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c16436b25d644d0abada474a4b35de61 : SUCCESS in 10m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ec745b24e4e94481b692f697f883cbef : SUCCESS in 12m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3384d44e5d434dd5a03c79420db72212 : SUCCESS in 12m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8f80854c21464e36b5c1932a771926c3 : SUCCESS in 1m 40s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d8b69a13867b45e69e8c145aea9803bf : SUCCESS in 13m 16s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cfd76adaf19d4b738abdbeeba420274d : SUCCESS in 22m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e8156dcdd92948b69e1c4b5655e563ab : SUCCESS in 8m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a8b49e6c83c24650a8279e01ae1d07e0 : SUCCESS in 12m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/622009d2951a41c2bef79270530aabe5 : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dc6313f8f4b54a4a82322a637c13d3ac : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67c9f58b8a1547deb06f96e4e2be58b0 : SUCCESS in 26m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0a11e07b2cd14c958cbbb41e5a753327 : SUCCESS in 6m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c03babf671cf4f37991b1cba9454fa64 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38695475efb847f1a0fdf4e1472cd51b : SUCCESS in 5m 34s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3f9ed037fabe489f9f93dac7cad92942 : SUCCESS in 18m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/16727f3b90114821ac35fbdb7881e8c7 : SUCCESS in 6m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6956ed7ebf734b69be0bf284890c2e25 : SUCCESS in 57m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/167afa18b70f4de0850de8921719ef5c : SUCCESS in 1h 53m 24s","accounts_in_message":[],"_revision_number":1},{"id":"cbb695280b8eaba06ee26cb46587dcc629718afa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 20:21:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f30dbad9e6791d87983bd6f751a5b1c83d1ac102","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 20:22:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05624d85d27f4aceb0fe1431ee44c003\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f5c9e7e8e164705ab48456c9fdfcbf7 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14b6339b0da8e8632a161d9dc1fda07a54805a51","revisions":{"14b6339b0da8e8632a161d9dc1fda07a54805a51":{"kind":"REWORK","_number":1,"created":"2025-09-01 13:57:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/959062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/959062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/959062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/959062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/959062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/959062/1"}}},"commit":{"parents":[{"commit":"56798df8bac5995e09a8bf3ce8820e0d4f8710fe","subject":"update constraint for futurist to new release 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56798df8bac5995e09a8bf3ce8820e0d4f8710fe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-01 13:47:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-01 13:47:51.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 5.13.0","message":"update constraint for python-ironicclient to new release 5.13.0\n\nmeta: version: 5.13.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Idc07a5cbe14ccf517c27264485a06731cc96832b\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib69509cb3b4f0ee9d8c425956e2511fb0452e8b9\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14b6339b0da8e8632a161d9dc1fda07a54805a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14b6339b0da8e8632a161d9dc1fda07a54805a51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
