)]}'
{"id":"openstack%2Frequirements~959533","triplet_id":"openstack%2Frequirements~stable%2F2024.2~I217a9ba0db6bfe8835b768704dadb027b904fba0","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"I217a9ba0db6bfe8835b768704dadb027b904fba0","subject":"update constraint for ceilometer to new release 23.0.1","status":"MERGED","created":"2025-09-04 14:23:04.000000000","updated":"2025-09-24 09:03:29.000000000","submitted":"2025-09-24 09:03:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959533-new-release","meta_rev_id":"677e849ade10b018cc3cdddc11986438bb571cfb","_number":959533,"virtual_id_number":959533,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-24 09:03:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-09-11 13:08:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-09-24 07:26:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-09-24 07:26:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-04 15:12:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-11 13:08:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-09-24 07:26:15.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"0cf08ea970a0452ac808d922d4bb08bdc97ea373","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-09-04 14:23:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f11d709e7795fe0a0c2d4311d7546bdc03a8f26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 15:12:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/508c689e3aaf45588b335fdc653bfc09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ce807c13c9f484ea3569ee3fb010bcd : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f9bfb6b4f774c128237ec7e4b24c811 : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3737a29681424c1783903af4fd509a0f : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f680de93331e4445a7d3a121ae3cbc70 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/caee868d9b32402382bc7ba49ef39d0e : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/473c82e2e16c4d3fa15e5a7c324efd07 : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/86c0cf289c4c4f9f8e7bf4b7fa630a58 : SUCCESS in 9m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/166655a36a32491284525a3c2933fec4 : SUCCESS in 10m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f298231c3b6f40bfb972b50bc7ef00d8 : SUCCESS in 10m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/12694249586f4ec7857495c0dda8cdcc : SUCCESS in 13m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8bb32870ec61448c987bf58a9f263526 : SUCCESS in 11m 34s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/547385c15c6145d4a9e2c199c3d76a69 : SUCCESS in 5m 06s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/6f9410332c234e92aaa0dfb858af3b37 : SUCCESS in 5m 30s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/e10a0d386296439b82d7463491c371e5 : SUCCESS in 4m 12s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4df611625e034b5191c3dbb2149e0836 : SUCCESS in 10m 37s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f5d4b0d0c6994402ae8d77ea7ec2a934 : SUCCESS in 5m 22s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/9e10aa9a31184b118ac8f8690fecb2e4 : SUCCESS in 6m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/fe909f286ea043d482226e0ed1e33f78 : SUCCESS in 8m 00s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/d6f86a1c033b42828cca109b1875335e : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cca98575ccf04c5b92044bc8ea3cdd40 : SUCCESS in 7m 48s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/560be1ceaec34f2098e16cbbf5ce17b8 : SUCCESS in 8m 22s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/832812ce39024153bba1ba50099cad8d : SUCCESS in 6m 50s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ff649d3ccad24bfbbddf5df62b1fc2d8 : SUCCESS in 2m 19s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/9b2b43113b0043cd8d4eb8bd95befe73 : SUCCESS in 9m 04s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/0c238b24b333479f93c784de59ae1c58 : SUCCESS in 4m 39s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/40dcbf65b98f48de948505204b2887bf : SUCCESS in 4m 46s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/88235051c92b4f4481f188abfa1fb741 : SUCCESS in 11m 45s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/9f6288205aa84a7387ea9f4cdb906d03 : SUCCESS in 17m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ccee682169b442fc8046e131e881067e : SUCCESS in 5m 47s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5ff9606c9ba14e78a7554b4780cbb4dd : SUCCESS in 10m 51s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/1cc6efa1e7ee4ea19bd34cb2d97f2a72 : SUCCESS in 2m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/16defb264aa14418af1112837cdd3108 : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/163ce24742914ec9bfffa70912a79182 : SUCCESS in 21m 53s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/695071b422c641cc952bd7ff65cf3eb8 : SUCCESS in 4m 36s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/acdfc84595fc41cf9eef38653b8b28f5 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87191ee6118749e5bc3ac47537b24639 : SUCCESS in 3m 36s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/dc061c5748f44a6b886fed57c3396679 : SUCCESS in 6m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52d40cce5e044848a59588579466592e : SUCCESS in 31m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0bb5f7d5ef7f4497bed864761aaa1da0 : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b2db6f16ccc00f6e55ed81e12e9b4aab51e591a8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-09-11 13:08:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7d0f8a3c1971077a3e8c3b310cede223f786fb08","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-09-24 07:26:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bae0461db6274c13d39aeaf8556eb42d762d9eb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 07:26:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3336f17b312045f2b0fdb12b9db47632ce34e1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 09:03:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31b2d1dff7d64efebfb9dfd1949ceee7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6253ca51778a4ad1bfb731afeb7686ed : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d80aae227f8145929cef31ed45960335 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d60ffe6e5344533af9160e3d1aaa6cf : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9002216d8612446998b430c829f11f9c : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b81a3c85ec0e45c4af7009d254936e80 : SUCCESS in 3m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f003ae9417b84f8c875c005e866b8a59 : SUCCESS in 9m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/38252de2ddb84bad9dec31aa8f77ab02 : SUCCESS in 14m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0bb4866ddb8a4b0dbd0eac6b6ec41f10 : SUCCESS in 9m 26s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a29609e7c7a24154b8389cd26e96ba97 : SUCCESS in 7m 29s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/2906b5ad39474233b9c2223249cdf6bd : SUCCESS in 10m 57s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/58d51b95ccca40c586752d61febe8088 : SUCCESS in 4m 53s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e8040d56b4b24171a6a80035359f8c69 : SUCCESS in 6m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/dc558b0b0fef4c43a3956afc63ade6d2 : SUCCESS in 7m 22s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b508c6897a8e4d9db3f58e147c041f3e : SUCCESS in 14m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3a6b3d630da94f0f91d1f1abaacaadd3 : SUCCESS in 7m 36s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0213b56ebe694d4795ee49d958d14390 : SUCCESS in 8m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0e6f4d0cf5b54ce89cce06c989a3e04b : SUCCESS in 12m 09s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/0ca48cc236cc4da4b01deb7e401b355e : SUCCESS in 2m 17s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8cfbbb7e9c7d49258b3146434d328923 : SUCCESS in 8m 49s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6e5fd7c9f9db4a7297265a6a0d6ee86a : SUCCESS in 15m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/69df24cedfba4f4cb250e4833f8fd9bc : SUCCESS in 3m 41s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f327419131e2435ca9439b4660aa7fdf : SUCCESS in 12m 59s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/81ffab0a8f2c453d8d99861c177fdfa0 : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43e809d75f6d46e28617527584052cc9 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a647f517f4de43d1b564296597d3b438 : SUCCESS in 19m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8f3ce36097b44a23a7e028f9abe8dada : SUCCESS in 2m 39s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/418acdd416eb479090159f24d429111c : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a8424fd74f64fb09bae2855b2f9dcb3 : SUCCESS in 5m 58s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1ea57fb8f64c40acb3f884c2c7484a37 : SUCCESS in 17m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fbe5c9dad6d40c8928f36a7d2c4b551 : SUCCESS in 1h 04m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/daf8fba3ba7e47f0b005d7dd04925e65 : SUCCESS in 1h 31m 22s","accounts_in_message":[],"_revision_number":1},{"id":"677e849ade10b018cc3cdddc11986438bb571cfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 09:03:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c31269cffe9d6e5414822882d6158934df536ee","revisions":{"2c31269cffe9d6e5414822882d6158934df536ee":{"kind":"REWORK","_number":1,"created":"2025-09-04 14:23:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/959533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/959533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/959533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/959533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/959533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/959533/1"}}},"commit":{"parents":[{"commit":"d0226ef2de72687eed37f4f029e49582d625e284","subject":"Drop build-wheel-cache-ubuntu-bionic-arm64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0226ef2de72687eed37f4f029e49582d625e284"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-04 14:13:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-04 14:13:03.000000000","tz":0},"subject":"update constraint for ceilometer to new release 23.0.1","message":"update constraint for ceilometer to new release 23.0.1\n\nmeta: version: 23.0.1\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\nmeta: release:Commit: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\nmeta: release:Change-Id: I89d46d81f16c3ec9cafd24b097210530e54e78ee\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Juan Larriba \u003cjlarriba@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I217a9ba0db6bfe8835b768704dadb027b904fba0\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c31269cffe9d6e5414822882d6158934df536ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c31269cffe9d6e5414822882d6158934df536ee"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
