)]}'
{"id":"openstack%2Frequirements~959535","triplet_id":"openstack%2Frequirements~stable%2F2024.1~Ie9801dde002a19116b9719d712f5471620aeb5c2","project":"openstack/requirements","branch":"stable/2024.1","topic":"new-release","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-09-24 07:33:18.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_12898\u003e by clicking the attention icon","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}},"hashtags":[],"change_id":"Ie9801dde002a19116b9719d712f5471620aeb5c2","subject":"update constraint for taskflow to new release 5.6.1","status":"ABANDONED","created":"2025-09-04 14:25:58.000000000","updated":"2025-10-16 22:43:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"55bfcbd9508d1551e6e6353d2b0f0005746568c9","_number":959535,"virtual_id_number":959535,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-24 08:39:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-09-24 07:29:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-09-24 07:29:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-04 16:01:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-11 13:09:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-09-24 00:54:20.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"8321ec88a3098bcd3b8c0f3ff989310d42bca04b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-09-04 14:25:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f60b92980db490fd6031d0ed7e9913f292ea735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 16:01:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfb52f845d8a42d2b2ddf892e9dafc49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/047c7a5920f147cea46c044d52229adb : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e97b7848d49349809661b1153655a90b : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c65e4dc5cf447f9a0c15cb2eb5ce7e2 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d01d2c094594991a38d38ac7984ac2c : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/52ca0fb6589a41808cba09550b9968e2 : SUCCESS in 2m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c6cb135fd4d5417ca3757f3a5d8b0ae1 : SUCCESS in 11m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/db8939868abf475ebfd4e943061d61b5 : SUCCESS in 14m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0568da8a09e74b3c8e5287f7809a2e8e : SUCCESS in 15m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5eefb7e5710c431da3694f8f3f5009d0 : SUCCESS in 16m 05s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ca82c3ab2c834ffb888e577b22833d4e : SUCCESS in 5m 15s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d08d5299084f4036b4a76caed8fdb026 : SUCCESS in 3m 15s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/9aa211a68d4e433f883397aac0511da8 : SUCCESS in 4m 19s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4d2d658fbe8a4ca6a1f833bc68545211 : SUCCESS in 9m 39s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2cbe5f7bd9e749eabf2f72d4ec888bcf : SUCCESS in 4m 52s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/93ae3aafeb864fa29f3b325bf0c368b9 : SUCCESS in 6m 14s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/76a775aafb294cd6af6f22f18be8a465 : SUCCESS in 10m 14s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/0017c4d4cf21490787f3dcc8d65ede30 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6180312706448ec9de17f83a54c694f : SUCCESS in 7m 03s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b925494d4d8a4efe9d602e979e5ef270 : SUCCESS in 8m 27s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a6697b0545464b089767bdd85196e589 : SUCCESS in 10m 25s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d64ff7c84e94496f96f42cd7e0d0b5cf : SUCCESS in 3m 35s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/49832b69d47c4fa1995f1bce92d4e06a : SUCCESS in 7m 14s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/891c63d4ae1a493398bde14394311173 : SUCCESS in 4m 30s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/f2cd25054366426f819ca2e17dd565fb : SUCCESS in 6m 20s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/1259db80413e4ffd811cb358c6177298 : SUCCESS in 11m 08s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f138b5c69d2445e4b3b167748c88b08d : SUCCESS in 21m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8f4ff3bf7f6949bbad04cdfca692b8c1 : SUCCESS in 5m 23s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/146246961977400983968c56dcdba981 : SUCCESS in 13m 21s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7da5045185164b7fb7173626a76942de : SUCCESS in 2m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8857ca8800e14081a8b9063166555110 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cecf34a19fe045fb98792819ad9ac5d2 : SUCCESS in 24m 57s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/fdf9b34ff71e4d81905efb6944d26e30 : SUCCESS in 4m 50s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/a6744ffb88254c5a88b31a28e0d6eece : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/630027ab4b4e4a95b8481b02e9799f57 : SUCCESS in 5m 58s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/a600d67388074d519a30dd38aed48d37 : SUCCESS in 14m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c1ec6aefc554cc8952695eea716c03c : SUCCESS in 58m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1878f631914d4c4d87aedcbdb8556222 : SUCCESS in 1h 30m 34s","accounts_in_message":[],"_revision_number":1},{"id":"85f6c33756676fdd2518b645f3bdb78f5c081839","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-09-11 13:09:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6585b64cd0a9f3f6ce5a0f9d43d4bf1ff53b5278","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-09-24 00:54:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70fe9bc039982b01186cfc2322439c108a440455","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 00:55:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ee5edb946d31e686b83e38ef1810a42f3032e86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 02:38:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c786d97b27cc44bd825c92dcf118455b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da00c477ca94c9887907db70c565d2d : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8686f511f8a487fa2f83001404572da : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e560a2db7a94a87ae70ddfd6225ab5e : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b71e0d96fa44a50a70edd6db364c990 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5f50d2b6b954d9c923d36249df90f17 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6783674400244883908595b5afb34691 : SUCCESS in 10m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/48953f61cb464d95ae4bdef720c5dede : SUCCESS in 14m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d6d43403bf9046fe856969ca905478fd : SUCCESS in 10m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bd707ddf195c4d15870bef118020d626 : SUCCESS in 16m 01s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/48254ccfff37450aa0a8311e272b72f2 : SUCCESS in 11m 15s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/48e427f8e4bf418198742da71c3cb34e : SUCCESS in 4m 17s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/f7f788545df146e08af5fe18221975af : SUCCESS in 6m 09s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/18f0356553f54879810cc4203d798282 : SUCCESS in 7m 35s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a8c2ce14db2244c3a3e036cfcce2c1c4 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01d6d4b841cd421284076de918fbbfad : RETRY_LIMIT in 1m 51s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/e171dc647afc4c90a0662c16472a5779 : SUCCESS in 5m 03s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/be7ce070b5334d4da04d3dd080af49c5 : SUCCESS in 10m 49s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/24393d647e65408f9d27f994fb9b7936 : SUCCESS in 4m 01s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/0e3be36111b14b0ab386b163ab9a313d : SUCCESS in 8m 29s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/7e792b9a66cf4d88bdf3aa541932ee0b : SUCCESS in 24m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5c92974afa494f89b271bc210043801c : SUCCESS in 5m 05s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/365c7dfa416f4cf2bbc6d8a99dee76fb : SUCCESS in 15m 22s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f3d50e76828b43e2808622e0d6099bcb : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/daa29e6ecc25498d8a43fdb21c48ce13 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33cc857d633843e5b1038213d0cefba3 : SUCCESS in 25m 06s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5e6b3a6c7aea4e3c9eeb2e4b4bffac4c : SUCCESS in 6m 14s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/71a00546c82246119587bb925d8d90bc : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/787d532fe179415998e249b33a0ae177 : SUCCESS in 6m 09s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/a45ee3d2ff084cce85abc5a1b0f03d58 : SUCCESS in 6m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71ecbb2fa309451696636f84b272ff5b : SUCCESS in 1h 02m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a0ec731554b46b09b5f222bbf9d0b12 : SUCCESS in 1h 38m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5ca6fad3cb7e3b4cfd24c03e8a18f21caecc5eec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-09-24 07:29:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"75faa49096c88961c63e7892f28379663a9e0b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 08:39:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48762a28c3164f9b9a1c5c5b8e6ba8b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d56416e8abc41789289645c767e53cc : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30fe2f085f56462fa2585cdabfa83e56 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b9f0aaa001348169005820baceea141 : SUCCESS in 2m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6dba9a36f1ea48c79fa5f96bff128b17 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/04f68d889d2f4869b17adbd6db7afe46 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f29f5edfb3da45e49db434d87842fae2 : SUCCESS in 3m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/784e897810274347b2fd8b9b637a1567 : SUCCESS in 13m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/70027af8a0df45b3a5a1db84217af92b : SUCCESS in 6m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1d695afcd47b45b1871eba1c427ef59f : SUCCESS in 19m 29s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/40167a4c92f3425f9b3beeaa6266e305 : SUCCESS in 2m 58s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/906ce301db3b43ef9d12ccfee3d7d762 : SUCCESS in 6m 19s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/ee7af8f9b6204751ad6acc4aae33dd1e : SUCCESS in 5m 55s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/704855ad57864a01b6ff9774c0afaf97 : SUCCESS in 10m 46s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/62fa020206d6421db7581be3373c3bba : SUCCESS in 5m 16s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/33df63e7a7354797900dacbac650890d : SUCCESS in 7m 13s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/752a80ee76d54cbe9424731c7c245f32 : SUCCESS in 7m 30s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/98650ec8915649f2814add475f11a954 : SUCCESS in 5m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9769d0e540cb418198ad12bf38c2363c : SUCCESS in 8m 22s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f321ea7047ae4c2fab6f9b2f2c83f2ea : SUCCESS in 7m 41s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a553dec5f2b04ac2ab39adc055e6e37b : SUCCESS in 10m 53s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/00e6c820b8b74655b36d7e3adeb07c0f : SUCCESS in 3m 22s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/99d7fcee2bca402b980db8a4ee1978d1 : SUCCESS in 5m 40s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ae2714285d1c4763a72b683b7edb672d : SUCCESS in 6m 20s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/dfb129df26f748f793fcb6deddbd362e : SUCCESS in 4m 23s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/3002c8317c7f4402993459fae2e519ad : SUCCESS in 11m 38s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/60ffc3d46c4a4c21a3ef3ee3a3cb8309 : SUCCESS in 16m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/675e59872e5d42c4be2be4f1d2fd35ed : SUCCESS in 6m 34s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/bd4e6c6843414bcda87a38a33f5ac63a : SUCCESS in 11m 58s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/faf53d07ae584bcda8bedc772558f08c : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/63759ac745f444d987e980e520ef599c : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4e7d9353bcb41efaff41ed356663d5d : SUCCESS in 19m 20s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/deda2b5a61cb49c58d4f8db8bdd2e1f3 : SUCCESS in 4m 51s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1840822304674a96bc45ff54e348eacf : SUCCESS in 3m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4df45e5641bf434da1a30d9086bbb459 : SUCCESS in 6m 57s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/af485ec1061948c2ba5c73429f2a65f9 : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a52daf0457384c0bae10bcb3cc3da704 : SUCCESS in 58m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ee4b2390cc748929da3abfbae5e3343 : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":2},{"id":"55bfcbd9508d1551e6e6353d2b0f0005746568c9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-16 22:43:10.000000000","message":"Abandoned\n\nDue to[1,2] these wheels aren\u0027t compatible with pkg_resources() in python 3.[89].  Changes to out PTI[3] and re-releasing these[4,5] are needed.  As such this change can/should be abandoned\n\n[1] https://github.com/pypa/setuptools/issues/5051\n[2] https://github.com/pypa/setuptools/issues/4853\n[3] https://review.opendev.org/c/openstack/project-config/+/964251\n[4] https://review.opendev.org/c/openstack/releases/+/964252\n[5] https://review.opendev.org/c/openstack/releases/+/964253","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6bef822c079b7426b57e98d9915351a2105bf7bf","revisions":{"2a93dbb612600feec22d459476474615c2884bb2":{"kind":"REWORK","_number":1,"created":"2025-09-04 14:25:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/959535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/959535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/959535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/959535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/959535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/959535/1"}}},"commit":{"parents":[{"commit":"a409487a0904de521d1ee034900320a2144356c1","subject":"Merge \"update constraint for python-manilaclient to new release 4.8.1\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a409487a0904de521d1ee034900320a2144356c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-04 14:15:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-04 14:15:57.000000000","tz":0},"subject":"update constraint for taskflow to new release 5.6.1","message":"update constraint for taskflow to new release 5.6.1\n\nmeta: version: 5.6.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: stable/2024.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Change-Id: I44d3ab16bd8c3d65f7878b9e462c0a5a6550f2ea\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie9801dde002a19116b9719d712f5471620aeb5c2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a93dbb612600feec22d459476474615c2884bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a93dbb612600feec22d459476474615c2884bb2"}]},"branch":"refs/heads/stable/2024.1"},"6bef822c079b7426b57e98d9915351a2105bf7bf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-24 07:29:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/959535/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/959535/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/959535/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/959535/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/959535/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/959535/2"}}},"commit":{"parents":[{"commit":"2a6f714e459568d7d0bbf3cb9606bea28b338464","subject":"Update horizon cross-npm job to define a node_version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a6f714e459568d7d0bbf3cb9606bea28b338464"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-04 14:15:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-09-24 07:29:26.000000000","tz":0},"subject":"update constraint for taskflow to new release 5.6.1","message":"update constraint for taskflow to new release 5.6.1\n\nmeta: version: 5.6.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: stable/2024.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Change-Id: I44d3ab16bd8c3d65f7878b9e462c0a5a6550f2ea\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie9801dde002a19116b9719d712f5471620aeb5c2\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bef822c079b7426b57e98d9915351a2105bf7bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bef822c079b7426b57e98d9915351a2105bf7bf"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
