)]}'
{"id":"openstack%2Frequirements~960630","triplet_id":"openstack%2Frequirements~master~I091a0fc213f9e615c60755a010e3012ad4f8277e","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-09-12 17:37:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I091a0fc213f9e615c60755a010e3012ad4f8277e","subject":"update constraint for aodhclient to new release 3.9.1","status":"MERGED","created":"2025-09-11 15:48:40.000000000","updated":"2025-09-12 17:39:31.000000000","submitted":"2025-09-12 17:37:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960630-new-release","meta_rev_id":"043d0d868f71bff5e99623020035cffda774c8bb","_number":960630,"virtual_id_number":960630,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-12 17:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-09-12 15:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-09-12 09:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-09-12 15:35:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-11 17:46:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-12 09:24:14.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2025-09-12 15:35:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"325e4a1fba71de5c4f484e14a7ba0c2836759d60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-11 15:48:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"827e1b232f93bd2fe344947f70de87bb46d1bcad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 17:46:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/436d331a5bf34b1bb33152369e98d5a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088f94c808134e65808a7eb25e49cf82 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec822f8a30e94f60af7bf47ccc2af0cd : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/299748efadff49ed90d3705c556a288c : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02cb7c2180ed4e39a5ba55740956df9b : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4f07921d25f4d37abad803361519455 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb5d2f043f5f4cc7b75372071a495643 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7566214c8a0a4a0aa4a4caf8a8fc0352 : SUCCESS in 3m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f7e5631f498f4525bea14c4209da218d : SUCCESS in 4m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3b9bddae55b8415f81923b11c4aa0367 : SUCCESS in 17m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dcfdf5caefbd49a9851154857dfc0516 : SUCCESS in 8m 17s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0f7539c0f3fe40c8bff573b5a8ca88e6 : SUCCESS in 10m 02s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b6a2916895084872ae43b995ac511e88 : SUCCESS in 5m 27s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2f4d12f2f19d40119bf1229f2fba17ed : SUCCESS in 5m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e4a8604ddbbd4351a05dc82bf990283b : SUCCESS in 2m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/98b86eb37bdd4b60a1c1e396592ff2a1 : SUCCESS in 14m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a1c1514ddd8c491680dcf8ecbadfff26 : SUCCESS in 3m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/93c8252a694346de9d284b778d9bc460 : SUCCESS in 3m 09s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/101487d5b9d44bd9818b6a43f740377d : SUCCESS in 7m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7170c783d9f04441be776761182ac096 : SUCCESS in 8m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09e4cd28e567454e9570de4546d92c5d : SUCCESS in 7m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d39e5d3e9c1d49638077e968d45fc520 : SUCCESS in 7m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e1b13dd8bf3c4739b009d10036e799e9 : SUCCESS in 12m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a9e1f6e7cf20472b8b77f3d7a8e14269 : SUCCESS in 1m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/90dc2328397b4b9fb5f590b92455d70d : SUCCESS in 8m 34s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9d383409867b41e489c34d9eb8593b0a : SUCCESS in 2m 46s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b2bf2613fb5049e2b0c506fb65062313 : SUCCESS in 5m 11s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/047e66df6c8548b1be199cacd06e6249 : SUCCESS in 9m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c994c68dc32f4bb7a5525dadb172d92b : SUCCESS in 14m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1b549631c33e4c47857b6b516b76e6d3 : SUCCESS in 6m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1f22dafd2a1a4e0697403ad504cc842f : SUCCESS in 14m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e2b19cfdd9d94bfebd4b43db7a6116a4 : SUCCESS in 5m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d3d6249087844084adf0ff2f1f122581 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b383768f71c4122a6aad5b8238529d2 : SUCCESS in 31m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/410077e704fa474597d6e0abcc144089 : SUCCESS in 6m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dbc6cfa8601c43f6851baa7dac8bbbdc : SUCCESS in 3m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd311e8519bd4963b825b264b554fce5 : SUCCESS in 6m 01s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/99d5bdb5c6324c618fcb9a47f933d4cd : SUCCESS in 20m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/13c09e8225714368bf75f29d82ae8f57 : SUCCESS in 7m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/758f6cdceb65430796d538c5519d3156 : SUCCESS in 1h 08m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f6e7fe38f42441bbbb17e10e6693fa8 : SUCCESS in 1h 55m 52s","accounts_in_message":[],"_revision_number":1},{"id":"093647abe97829314a5b9ef1366916cc67cbe9f7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-09-12 09:24:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ca0879ea4a311fcd27a0c761cceee8ea0a22e8ff","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-09-12 15:35:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d93d3063a633b3aa8598a606671105e8bbe3a1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 15:36:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6580cbb95e5b2cf4b68310c13257bb79c2aa4eb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 17:37:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/587714e426564bdfa19c5246eb20fbc2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336e13cad1b84ec391ffc17d54161958 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02155559a83e415ba2fef5b118e65804 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3dcdaa10fbb46c19770adf436929435 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be1fa4b8ee3f42ef966e77f2ea3e0938 : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/25b848667e0c4f4995a6741dfc100db9 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1aeb80c8b8fd43ee99a6e2da8babd18c : SUCCESS in 5m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9c625b69fd6d4b988291408f5f9db5b8 : SUCCESS in 9m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/37ec5e6fce544929994fe65e5dda03b0 : SUCCESS in 17m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c446f3e30fed4a4f9879e44d075172af : SUCCESS in 20m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8743d66720cd4f6ea8ea64c846325508 : SUCCESS in 9m 28s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b9de676dc1ea4dcfae69d7765ba339c8 : SUCCESS in 5m 27s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6302113b54a4465a8b84c2e118137de6 : SUCCESS in 5m 24s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/91820934302d44c8b8780217b6ddfb93 : SUCCESS in 7m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1288d3f0b3c640e083da1df59a20b9bd : SUCCESS in 17m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7cadf0b5f6034eb695035f72eb913303 : SUCCESS in 8m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/839d20955b154f1c8ea0a21c22a8f412 : SUCCESS in 6m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dcfbb6b62df445cca230bd9d78e33f22 : SUCCESS in 14m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1544c802ca8c4becb403dd5b859045f1 : SUCCESS in 2m 42s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5c060ed779264060812f348afcd71401 : SUCCESS in 9m 22s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c53d2f5fe6fc4648953995ff0808a36c : SUCCESS in 22m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ae0ac6e0326e48b4b2952ff97b95c42c : SUCCESS in 7m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8b69b3f21aa04033aeae602142dcd996 : SUCCESS in 12m 38s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/34323b200e07422bb6966903ded3ef66 : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a26f197342a54caf8d96f1e1f15d7815 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1dc1c6ccd5064425af1aac1c7c1d8ed0 : SUCCESS in 21m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/852acf381c724ab48a96033e91cf6207 : SUCCESS in 6m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e776f2d899ad494b8d220fc8471069d6 : SUCCESS in 4m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd31c0be1c3c4425b9b4349f877f3159 : SUCCESS in 6m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6ff8bdc4c03c42ef9f4c93e940b3d446 : SUCCESS in 16m 04s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/09b81bab8a464ee69794c467fdafe569 : SUCCESS in 7m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/801a5009e15f4ada81d7aae525708a5f : SUCCESS in 1h 03m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a840a81ee1843148fa0b77abfa36d2a : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c45846ba21c1703a3c09d575f521141adc44bd33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 17:37:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"043d0d868f71bff5e99623020035cffda774c8bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 17:39:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7fa4b3a8b5948b1bc02a053d4ee3e2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa38780f36ff4a93bd9b0de594c832e4 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d783344ddeb4e101d1a03c8eb8abee39926e68c","revisions":{"9d783344ddeb4e101d1a03c8eb8abee39926e68c":{"kind":"REWORK","_number":1,"created":"2025-09-11 15:48:40.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/30/960630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/960630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/960630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/960630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/960630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/960630/1"}}},"commit":{"parents":[{"commit":"46a7e3927152884c05f1e257c528af593d67b84d","subject":"Merge \"update constraint for python-ironicclient to new release 5.13.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46a7e3927152884c05f1e257c528af593d67b84d"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-11 15:48:35.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-11 15:48:35.000000000","tz":120},"subject":"update constraint for aodhclient to new release 3.9.1","message":"update constraint for aodhclient to new release 3.9.1\n\nmeta: version: 3.9.1\nmeta: team: Telemetry\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I295f4aa848c0c206e5e4e3d47378b83c24a7d9bc\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Juan Larriba \u003cjlarriba@redhat.com\u003e\nChange-Id: I091a0fc213f9e615c60755a010e3012ad4f8277e\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d783344ddeb4e101d1a03c8eb8abee39926e68c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d783344ddeb4e101d1a03c8eb8abee39926e68c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
