)]}'
{"id":"openstack%2Frequirements~961677","triplet_id":"openstack%2Frequirements~master~I4805d1ac0fa0209c59f3815352ff9ea54108eca2","project":"openstack/requirements","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-14 04:00:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4805d1ac0fa0209c59f3815352ff9ea54108eca2","subject":"tox: Assorted changes","status":"MERGED","created":"2025-09-18 15:00:07.000000000","updated":"2025-10-14 04:01:54.000000000","submitted":"2025-10-14 04:00:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961677-pyproject","meta_rev_id":"f814995863043117200c339ff5b8cbbb514bdc7d","_number":961677,"virtual_id_number":961677,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 04:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2025-10-14 01:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2025-10-13 23:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2025-10-14 01:59:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-18 17:00:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-13 23:44:13.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2025-10-14 01:59:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"979f572135d6ca270bd0b5e7ef91b6c1bd458cbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-18 15:00:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e88259f474566527ab235632e6faf28d329f83d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 17:00:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/655825049f29413a82545ab7a59b54ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3ceab6af744604b39c4bb20a416531 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f8328f8816146ec8b835414e5d466fa : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cb6131f209f480f9855beb3fb9b12e6 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e0949a1389c440e91f953b98d49d071 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0a19e6c729d44bf8a33463523c40a1d : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7ac123da1e848f5ab1dc21c40233cca : SUCCESS in 2m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1893f401741a4d2e9eec45bdd22c7369 : SUCCESS in 3m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f433f041aec494b9c46adae34eefacc : SUCCESS in 56m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c7e5f43e1554343ba28f3b170a162f1 : SUCCESS in 1h 39m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7616badaec311638a4c8dad7c2d5afbaccfd8018","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-13 23:44:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f38eba464fb498025a46c3dac7dc4ff0231d1f73","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2025-10-14 01:59:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c3bac6281968185cea65cadcdc2424f9d1367e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 02:00:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b1da963d2365ca6be9b917b2f594fb0003c18e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 04:00:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c12b9984e7d4462aa73d3d25c3bb199e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ecc844627da4476a8f30f9990b6a7eb : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd780313a3b41afa6c9f3ca62a3d68c : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d6a82658eb54f5881e0ac1490d2e333 : SUCCESS in 2m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96868ce6dfcc4908ae0230c7ae5cd5dc : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cee2163211fb44678f8936a80f872562 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/197b14864f414e159f7d6a09aad90be0 : SUCCESS in 3m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/814615c8f33044f3a0f09ac790566372 : SUCCESS in 48m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4902d4af7cd48d392ec0781ba42c51a : SUCCESS in 1h 55m 22s","accounts_in_message":[],"_revision_number":1},{"id":"4d458ef9c8c7332f950e614db958298a3b0dc51b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 04:00:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f814995863043117200c339ff5b8cbbb514bdc7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 04:01:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/849cf34be123415786aa66afe6b5cd28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91bc08187005477685637ea0f10d311f : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5bc5127cc99be8b86dce266c1060bf32651bbc8b","revisions":{"5bc5127cc99be8b86dce266c1060bf32651bbc8b":{"kind":"REWORK","_number":1,"created":"2025-09-18 15:00:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/77/961677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/961677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/961677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/961677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/961677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/961677/1"}}},"commit":{"parents":[{"commit":"aa77316a1153af165e18e1685f76afee3035590a","subject":"Remove check-python2-support tool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa77316a1153af165e18e1685f76afee3035590a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 11:04:06.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 13:37:36.000000000","tz":60},"subject":"tox: Assorted changes","message":"tox: Assorted changes\n\n- Remove the \u0027linters\u0027 testenv: this battle has been lost\n- Don\u0027t install package for the \u0027pep8\u0027 testenv: it\u0027s unnecessary\n- Remove \u0027safety\u0027 from \u0027pep8\u0027 testenv: the command we were using is\n  deprecated and the new one requires authentication\n- Fix the \u0027bindep\u0027 testenv so it (once again) skips install of the\n  package and do the same for the \u0027pep8\u0027 and \u0027babel\u0027 testenvs\n- Stop setting \u0027basepython\u0027: everything is Python 3 nowadays (tox 4\n  *only* supports that)\n- Remove a linter from the \u0027test-requirements.txt\u0027 file: these\n  dependencies are managed by tox\n- Fix some indentation\n\nChange-Id: I4805d1ac0fa0209c59f3815352ff9ea54108eca2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bc5127cc99be8b86dce266c1060bf32651bbc8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bc5127cc99be8b86dce266c1060bf32651bbc8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
