)]}'
{"id":"openstack%2Frequirements~961682","triplet_id":"openstack%2Frequirements~master~Icf2ae0a15aa1f35ed3954f3ca2fb0b3f3ceee414","project":"openstack/requirements","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-10-22 22:22:42.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-27 17:32:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icf2ae0a15aa1f35ed3954f3ca2fb0b3f3ceee414","subject":"Rework extras tracking","status":"MERGED","created":"2025-09-18 15:00:07.000000000","updated":"2025-10-27 17:32:07.000000000","submitted":"2025-10-27 17:32:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961682-pyproject","meta_rev_id":"7f0d2d9ee4e8b3a277f093bd7d1aca23636b3b26","_number":961682,"virtual_id_number":961682,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 17:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-27 15:04:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-23 09:26:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-10-27 15:04:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-18 17:10:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-21 03:47:01.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2025-10-27 15:04:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"4a7baf6552477522e25a25825acbe3762ed86d41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-18 15:00:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07ed42bd45cb3a1b36ea3900a8ca31e8eda451df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 17:10:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f44df50f72a84918b1c3db8a90d896c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1505a85f5b454a4ca27d0a6321baae9c : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5980bd853141407483c0a85df6700fa3 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76b5f477d0f64405ac8527b3e0fa6d89 : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/61d98704e13f481faa53589f1fafd3c0 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c815ccb8c23411facb9e96060588081 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e19122ece76e4aa49db1bc7f36494dec : SUCCESS in 3m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddabcb322d6c459fbd463e6d1dd11c33 : SUCCESS in 59m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b410a5b50f2e43d6bd44c696d317d373 : SUCCESS in 1h 50m 17s","accounts_in_message":[],"_revision_number":1},{"id":"244f8dd63dd3b6dde04eef8c6711766e42162638","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-14 09:38:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"da8653ef924cb0f26447bec6f764a10dd5328dee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 11:30:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8ee0e5a32fa439ca40df6e21a103647\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99f77114358d4145b1396c4e81f3142a : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef495090758499e86132944f241b2ee : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8fa62c8cd024a65b207f5990b08d0e1 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84384e120ade4bdd9c75542513621956 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c9ab1921f8d417789c502aab3d7b1dd : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ca637d4a37f40e69bdeb521b27717f2 : SUCCESS in 3m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e146810575634fb392da7eadadc0aded : SUCCESS in 1h 00m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6791f077af1b4f5b8877e8c341d4e9c6 : SUCCESS in 1h 36m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5f6cbf5ca6d8137a5c5e02cde406c192cec47c66","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-21 03:47:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"13fb1798c0de9fac26ab5610b3c6835d2740e968","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-22 12:49:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f6ecfc228420e909c0933a7a67a046a8e183fc8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 13:58:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/015c998d873b447dad2aa4ac92d266fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84315a2c818841ce8229737185af2eca : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/531178616379465b95608d541fcacffe : SUCCESS in 2m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e3c91ff9d96446f91624e0d630ed3dd : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcadfd083f664406a472852e86fcf612 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6bfed2c51b824559b36f8358dc85c48e : SUCCESS in 3m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7908070d02a4796833a1889898583be : SUCCESS in 1h 03m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea3f59b22b794fc197f65c9bf95ea346 : SUCCESS in 53m 51s","accounts_in_message":[],"_revision_number":3},{"id":"a99a322f0c40aa1e4032ce3ee28b0d7f75cf5e3c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-22 22:22:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5192f97981c2413e8cac1883ea82be6b239036ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-23 09:26:08.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"1828b431ce50245a0e1791bac4540d88bdff8914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 11:16:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da53c5f2401e42d8bf68d7869b5086d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba778f82a9114c55addf11898b05bd95 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/851714a29a9e499ca8ff26976d92d914 : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2933e011a2f2433c8e295a976c1c25f6 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a1e709b687a401d97af73790fb6f6d3 : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65fdfd71d59f48de909b200a59eb8b8b : SUCCESS in 5m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e45146961b564c5a9382fb1eb628398a : SUCCESS in 1h 04m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54ffec9f80fd4c858b13d8a08f3ed186 : SUCCESS in 1h 42m 40s","accounts_in_message":[],"_revision_number":4},{"id":"a144fcb149e9d45fbb5bf83f3cc0f6caaa7b310a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-27 15:04:50.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2bbce64a17b914ba2e7830fc0a00ec838a0b06cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:14:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"24b3f59d43f9b76b09b491ec2fc198cdebb74429","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:32:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/597316ed6e874e259b8baebb39f3fb59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24f10c21084426484b5515385486445 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a7ce8c3a2fa4d66bbde11d96c76c079 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad2070694f9a421fa176cc989e047c77 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86fd13efc247478fa3064e9548a4ee2d : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c71a18ef0ff243f9a531dd931cf12efd : SUCCESS in 3m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3088a4e6a4f340cfab1281672f7d3481 : SUCCESS in 32m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15ae0e80428a4e00b48b1a08cbbfc5d6 : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":4},{"id":"7f0d2d9ee4e8b3a277f093bd7d1aca23636b3b26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:32:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f7ff019cfa5d68feaa8946a24af97c5b5dbc3ba8","revisions":{"23c497bcdb4748a7442d5249e94b715567f5e7dd":{"kind":"REWORK","_number":1,"created":"2025-09-18 15:00:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/961682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/961682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/961682/1"}}},"commit":{"parents":[{"commit":"ae87b5531173a73e239a4ac064d62b9afd914efb","subject":"Deprecate support for legacy requirement files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae87b5531173a73e239a4ac064d62b9afd914efb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 12:18:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 14:54:19.000000000","tz":60},"subject":"Rework extras tracking","message":"Rework extras tracking\n\nStore this in the same dict that we currently store requirements in.\nThis will make it easier to tack on pyproject.toml handling in an\nupcoming change.\n\nChange-Id: Icf2ae0a15aa1f35ed3954f3ca2fb0b3f3ceee414\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23c497bcdb4748a7442d5249e94b715567f5e7dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23c497bcdb4748a7442d5249e94b715567f5e7dd"}]},"branch":"refs/heads/master"},"90a6f2d284a7cbf5b653728dd77c1e4279f1ecfc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-14 09:38:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/961682/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/961682/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/961682/2"}}},"commit":{"parents":[{"commit":"c8b9b26b0ae9b6d8b648b6f9edc4cebbeeca22b4","subject":"Deprecate support for legacy requirement files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8b9b26b0ae9b6d8b648b6f9edc4cebbeeca22b4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 12:18:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-14 09:38:40.000000000","tz":60},"subject":"Rework extras tracking","message":"Rework extras tracking\n\nStore this in the same dict that we currently store requirements in.\nThis will make it easier to tack on pyproject.toml handling in an\nupcoming change.\n\nChange-Id: Icf2ae0a15aa1f35ed3954f3ca2fb0b3f3ceee414\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90a6f2d284a7cbf5b653728dd77c1e4279f1ecfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90a6f2d284a7cbf5b653728dd77c1e4279f1ecfc"}]},"branch":"refs/heads/master"},"5f4793468cceaa906f5ed2b7df4b0458aa775f68":{"kind":"REWORK","_number":3,"created":"2025-10-22 12:49:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/961682/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/961682/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/961682/3"}}},"commit":{"parents":[{"commit":"3f22604003fbafb3dfcd1650c97a3a4810644673","subject":"Deprecate support for legacy requirement files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f22604003fbafb3dfcd1650c97a3a4810644673"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 12:18:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-22 12:45:50.000000000","tz":60},"subject":"Rework extras tracking","message":"Rework extras tracking\n\nStore this in the same dict that we currently store requirements in.\nThis will make it easier to tack on pyproject.toml handling in an\nupcoming change.\n\nChange-Id: Icf2ae0a15aa1f35ed3954f3ca2fb0b3f3ceee414\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f4793468cceaa906f5ed2b7df4b0458aa775f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f4793468cceaa906f5ed2b7df4b0458aa775f68"}]},"branch":"refs/heads/master"},"f7ff019cfa5d68feaa8946a24af97c5b5dbc3ba8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-23 09:26:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/961682/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/961682/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/961682/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/961682/4"}}},"commit":{"parents":[{"commit":"1691dcb547658a5182e713a0896f2efe2e898fab","subject":"Deprecate support for legacy requirement files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1691dcb547658a5182e713a0896f2efe2e898fab"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 12:18:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 09:26:03.000000000","tz":60},"subject":"Rework extras tracking","message":"Rework extras tracking\n\nStore this in the same dict that we currently store requirements in.\nThis will make it easier to tack on pyproject.toml handling in an\nupcoming change.\n\nChange-Id: Icf2ae0a15aa1f35ed3954f3ca2fb0b3f3ceee414\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7ff019cfa5d68feaa8946a24af97c5b5dbc3ba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7ff019cfa5d68feaa8946a24af97c5b5dbc3ba8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
