)]}'
{"id":"openstack%2Frequirements~961683","triplet_id":"openstack%2Frequirements~master~I2c98b0792228d96f0ce027dc5476b146365cfc58","project":"openstack/requirements","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-10-23 09:41:05.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-27 17:32:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2c98b0792228d96f0ce027dc5476b146365cfc58","subject":"Stop storing setup.cfg, setup.py","status":"MERGED","created":"2025-09-18 15:00:07.000000000","updated":"2025-10-27 17:32:11.000000000","submitted":"2025-10-27 17:32:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961683-pyproject","meta_rev_id":"7572436dfaedbd53b0e4b32d0ab6cb796d6b9288","_number":961683,"virtual_id_number":961683,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 17:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-27 15:08:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-23 09:41:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-10-27 15:08:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-18 17:00:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-21 04:07:08.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2025-10-23 09:41:05.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2025-10-27 15:08:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"ca47fa6eb7278ed1ab051519a7f9a1c8839fbea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-18 15:00:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"beda608c58cb3aefb78c1d0ede0ad82d9513e18a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 17:00:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e0aea0837fb4d5cbb722e6b7f370c59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b2f5c2b85ca4ffd951f30af170c41c9 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f10fce7263344660be65111f8c8ab1de : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f149a3c400ce4385a4c757453b9443fd : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d37d94a0330142bc939716da1ea274bf : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a6edc149da414c8d90e6ace54c4d59dd : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b14cf22439404e968e02cc93d58dbae3 : SUCCESS in 3m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d218d521d04d4328b4212a1068fb9e41 : SUCCESS in 32m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38572fb59998400b8851902da8d2c736 : SUCCESS in 1h 50m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ebb3198ad9ad3e9db559c4d109e70ac0df6b790f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-14 09:38:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bd82b72bcbd8d0116f266b1ab5d2279b842eefc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 10:38:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd5f99c1649440fdbec1a6174ee9d953\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2eb6c0abf548d1b43f4d5b24e4db6a : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3724b2b30c3403e916f2bffb7b4c24e : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a99224fed94a466e9bc6a93a96d9b604 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2bce0ec04958490a9cd9a470a63e783f : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2db5fe352334bb4a7d4505d4423e0a6 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73bb042aafa74ad4a6ca0a1cc5bb1dba : SUCCESS in 3m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3447022958ed48f4bf46a0fa6f540721 : SUCCESS in 33m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/909dc1c76f8448aaa43d5f78390c6769 : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c880258e1d8f8cfdd0fcad9c267c24d78f004360","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-21 04:07:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b947e4cff618482898bdb5161b8787292cbd1f0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-22 12:49:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8cdee218ffc46411117e14bc92b92d395872a1bc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-22 12:50:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb5230724f5da83d858c5a1d26bde96ac4b347dc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-22 12:55:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3d42f81c94d37f4c71d831b4351d5fdff6b843ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 14:39:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe4e0e137624c2a8d3c76b34b8bdb67\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b066c20526df4de498bec438e99b6c6d : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b9492f048714820b38d606851ca8747 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d5266e7341b444980232ec9ceb4f412 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df2ad77aecef48039f35387a9caf35b1 : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cdbdcfa23f6543d2a1d584745c15dab9 : SUCCESS in 3m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/741981eacab74ebbba8b066cc554e4db : SUCCESS in 52m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8007dfc54dec459bbe498789724802a4 : SUCCESS in 1h 46m 02s","accounts_in_message":[],"_revision_number":3},{"id":"26e8f02cc065e292434055e201e6ce1c33200524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-23 09:26:08.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e6d97263789e18ae6b8ea9c26586dad3c7855d1f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-23 09:41:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bc277046a7f05fd6adb0219970adfa172fa1c624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 11:08:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28b0d8c5f8094da19c5fd600add9eaf9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c95ae12a704a4c0fa9ca83a8718618ff : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76db8d45b6ed4d9fa81cbcfb8a1944ec : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62120d813fc446a996afcdf3098f18bb : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad1e45a35aca42ae8c7339e2a98c880a : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a05a0c49b2d74866a55003283b8fb3db : SUCCESS in 3m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f280e84a2264155a32a7a6edd12eba6 : SUCCESS in 45m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/002edb9a50ee4c41b10b7ec663e79464 : SUCCESS in 1h 36m 34s","accounts_in_message":[],"_revision_number":4},{"id":"54b516aa9ec7689958fdfaa5f45451d80d568d81","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-27 15:08:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c636a8947b4c7d0012defc1e194965051ab41727","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:14:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"249b9c1c71769d566cf986ab5a36227474a1d6ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:32:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b847120b686847a281a404790e76eb0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48ffd37cf6084e0d822375aae00d9ca5 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa885d47b8c8401ba18a47798f961fd2 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6a75ad5b0004bb78e632dbee9407876 : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53706a4c40564f4d8b17e548382b8dda : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27880daf256346cba2de1d6a791e39df : SUCCESS in 3m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcaa7d1492624532ad730a24e8a0b5b0 : SUCCESS in 1h 02m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b89406f09eac42a9ae95814e130f5fdc : SUCCESS in 1h 44m 54s","accounts_in_message":[],"_revision_number":4},{"id":"7572436dfaedbd53b0e4b32d0ab6cb796d6b9288","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:32:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"32096476187f47cf05d0f42a832e63858ed1351a","revisions":{"7239a5aee70af2a228d9c0acdc252051860d0631":{"kind":"REWORK","_number":1,"created":"2025-09-18 15:00:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/961683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/961683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/961683/1"}}},"commit":{"parents":[{"commit":"23c497bcdb4748a7442d5249e94b715567f5e7dd","subject":"Rework extras tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23c497bcdb4748a7442d5249e94b715567f5e7dd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 14:41:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 14:54:45.000000000","tz":60},"subject":"Stop storing setup.cfg, setup.py","message":"Stop storing setup.cfg, setup.py\n\nWe no longer need to parse these anywhere.\n\nWe also remove the unnecessary use of matchers from tests in favour of\n(more common) assertions.\n\nChange-Id: I2c98b0792228d96f0ce027dc5476b146365cfc58\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7239a5aee70af2a228d9c0acdc252051860d0631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7239a5aee70af2a228d9c0acdc252051860d0631"}]},"branch":"refs/heads/master"},"c83da5532a51257513b49fa71169cc60fab5d033":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-14 09:38:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/961683/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/961683/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/961683/2"}}},"commit":{"parents":[{"commit":"90a6f2d284a7cbf5b653728dd77c1e4279f1ecfc","subject":"Rework extras tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90a6f2d284a7cbf5b653728dd77c1e4279f1ecfc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 14:41:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-14 09:38:40.000000000","tz":60},"subject":"Stop storing setup.cfg, setup.py","message":"Stop storing setup.cfg, setup.py\n\nWe no longer need to parse these anywhere.\n\nWe also remove the unnecessary use of matchers from tests in favour of\n(more common) assertions.\n\nChange-Id: I2c98b0792228d96f0ce027dc5476b146365cfc58\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c83da5532a51257513b49fa71169cc60fab5d033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c83da5532a51257513b49fa71169cc60fab5d033"}]},"branch":"refs/heads/master"},"27106a683748678f5c3a66c0a509ca504360cf42":{"kind":"REWORK","_number":3,"created":"2025-10-22 12:49:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/961683/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/961683/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/961683/3"}}},"commit":{"parents":[{"commit":"5f4793468cceaa906f5ed2b7df4b0458aa775f68","subject":"Rework extras tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f4793468cceaa906f5ed2b7df4b0458aa775f68"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 14:41:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-22 12:45:50.000000000","tz":60},"subject":"Stop storing setup.cfg, setup.py","message":"Stop storing setup.cfg, setup.py\n\nWe no longer need to parse these anywhere.\n\nWe also remove the unnecessary use of matchers from tests in favour of\n(more common) assertions.\n\nChange-Id: I2c98b0792228d96f0ce027dc5476b146365cfc58\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27106a683748678f5c3a66c0a509ca504360cf42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27106a683748678f5c3a66c0a509ca504360cf42"}]},"branch":"refs/heads/master"},"32096476187f47cf05d0f42a832e63858ed1351a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-23 09:26:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/961683/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/961683/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/961683/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/961683/4"}}},"commit":{"parents":[{"commit":"f7ff019cfa5d68feaa8946a24af97c5b5dbc3ba8","subject":"Rework extras tracking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7ff019cfa5d68feaa8946a24af97c5b5dbc3ba8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 14:41:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 09:26:03.000000000","tz":60},"subject":"Stop storing setup.cfg, setup.py","message":"Stop storing setup.cfg, setup.py\n\nWe no longer need to parse these anywhere.\n\nWe also remove the unnecessary use of matchers from tests in favour of\n(more common) assertions.\n\nChange-Id: I2c98b0792228d96f0ce027dc5476b146365cfc58\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32096476187f47cf05d0f42a832e63858ed1351a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32096476187f47cf05d0f42a832e63858ed1351a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
