)]}'
{"id":"openstack%2Frequirements~961684","triplet_id":"openstack%2Frequirements~master~I3335b5faac72e2e6962d0930eef0e3b704820bbe","project":"openstack/requirements","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-27 17:32:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3335b5faac72e2e6962d0930eef0e3b704820bbe","subject":"Add pyproject.toml support","status":"MERGED","created":"2025-09-18 15:00:07.000000000","updated":"2025-10-27 17:35:26.000000000","submitted":"2025-10-27 17:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961684-pyproject","meta_rev_id":"68214fdce0374ee7484e38f9e68d381ab84be9dd","_number":961684,"virtual_id_number":961684,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 17:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-27 15:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-23 09:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-10-27 15:12:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-18 16:22:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-23 09:37:10.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2025-10-27 15:12:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"0323ccf98d0f26b6bee26217b771ee33cbc85cf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-18 15:00:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dddce9d1b92355a1997c1d69ca5aff8a445a4ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 16:22:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/408cd0d2105346838bea6d7006b304e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5e91715c454c2e9c5c0c83b1590ff7 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532c1f2c4eaf4af98e6531b6a486671a : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efbe6c721ab24c4fa36f542191d141e7 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2079394f1904359aab670cece2922a6 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/817f437a692046a8b89f6ddabf3b9095 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44f6d87baae0417fa1396acdc2ef8937 : SUCCESS in 3m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65a2f55ec2dd489b97cdaa31159e5a0d : SUCCESS in 1h 01m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8964189877b64a6781433f037f333851 : SUCCESS in 1h 05m 56s","accounts_in_message":[],"_revision_number":1},{"id":"38cd39d815ce8944df10a7a9e238616e775f8679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-14 09:38:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"93ad2554dbb27426f1eb884602f83f97ace39791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 11:50:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5269fb108724d99bafd4cebf12299b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea03f7c668c462481c698a4a176f123 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3ad8102d7d47beb43ca4e1c7a0be65 : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c3ec5e745824cab8b62927002179367 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e69596b87057430f902805459608b213 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/427f6ad8b433496da8618c27701103aa : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e88ec9edce64e3bae39924ab1694ef9 : SUCCESS in 4m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1b82285caa644b0bae7990db330cb82 : SUCCESS in 33m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c109449bb404ad4bedb3c4284671047 : SUCCESS in 1h 49m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a14a4f39a2743f8002aa3d59668a208b01138e45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-22 12:49:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"277d87b2c619ee7e966de33b01944b6e3c14653c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 14:39:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9934f60bbb34444ab08cfdaae3ad52b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb7a6c68677e4386988650144771bc48 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a7938f2019645b19a11d64114bd6852 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39b8c9d1c00d4aaead9fb07b28f3751d : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81613550dcc042a1a24db7b9d67eaa20 : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a3a0a597b1745f1a1d13121b0aa4507 : SUCCESS in 3m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a61be33176941748c663f2003e209ba : SUCCESS in 57m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a85abf5ca68f4d1f8c2a18727f0f5bb7 : SUCCESS in 1h 46m 06s","accounts_in_message":[],"_revision_number":3},{"id":"276f54e2d28c65c118314ab0bc697d1a5d28b960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-23 09:26:08.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2cf454edb1de51fc63bd6871ffbcafe8828045d1","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-23 09:37:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a77e604c252f497583fa1155433d2cfe1e0d0750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 10:49:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d62468c06d2844208b3d5c485771394d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5150b6a022ef48e9a9868db99144c075 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5619ae5e80a4b1da71b3c8744a54fd8 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/564ec26cec574b9abf013778112f2bb7 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fe2811583914dafb01eefb3c78e535a : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d590237cef4c49d4bf8e68fe49036cbe : SUCCESS in 3m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/393d93c313e84dc58bd7b16e8ed7be9f : SUCCESS in 39m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1291a2c2e1204be3aec6433d87cd6efb : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":4},{"id":"51aa32ea0b1204f2797c16ce0fd459e064b313e7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-27 15:12:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6e315418ae39c8c8633aa90a1a195d1d9997485d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:14:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"facb1dff72523eee506653a95fc65d40cf9b297e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:32:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b4066ec6bc040e989faf79811d94076\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02b458f05814915bd66ffd953a10e71 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9abb61775b8245099846ffd963e97154 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31c46f9da0bf41e9acf905365d6c21ea : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f84ebea75ed42e5bcce24252bd932eb : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec234e00f72845acac378824bd5c6328 : SUCCESS in 3m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/926d756fc8714a5a898b84e53fb594b7 : SUCCESS in 49m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2749e58ff8845569b3986dff0e3314e : SUCCESS in 1h 42m 37s","accounts_in_message":[],"_revision_number":4},{"id":"0756318c9b7d3878e57b40763fc3df82df5239cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:32:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"68214fdce0374ee7484e38f9e68d381ab84be9dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:35:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/225f2181af4642b095d4aa771595d858\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6c2f1405b94130b99c9d2cb89da33c : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9b79bc474cf7884d04fa9d239abca1c5f6b0899e","revisions":{"006f728f1f4f211d9f464365fe7a6693083073b4":{"kind":"REWORK","_number":1,"created":"2025-09-18 15:00:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/961684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/961684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/961684/1"}}},"commit":{"parents":[{"commit":"7239a5aee70af2a228d9c0acdc252051860d0631","subject":"Stop storing setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7239a5aee70af2a228d9c0acdc252051860d0631"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 12:43:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 14:55:14.000000000","tz":60},"subject":"Add pyproject.toml support","message":"Add pyproject.toml support\n\nAdd the ability to read requirements and extras from pyproject.toml\nfiles, eventually allowing us to move away from requirements.txt files\nif we so choose.\n\nTests are reworked to test this new functionality, with some minor\ncleanup to remove unused fixtures.\n\nChange-Id: I3335b5faac72e2e6962d0930eef0e3b704820bbe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/006f728f1f4f211d9f464365fe7a6693083073b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/006f728f1f4f211d9f464365fe7a6693083073b4"}]},"branch":"refs/heads/master"},"71b112dd8b790ec7a87ffac4c8d89c248fed7369":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-14 09:38:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/961684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/961684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/961684/2"}}},"commit":{"parents":[{"commit":"c83da5532a51257513b49fa71169cc60fab5d033","subject":"Stop storing setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c83da5532a51257513b49fa71169cc60fab5d033"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 12:43:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-14 09:38:40.000000000","tz":60},"subject":"Add pyproject.toml support","message":"Add pyproject.toml support\n\nAdd the ability to read requirements and extras from pyproject.toml\nfiles, eventually allowing us to move away from requirements.txt files\nif we so choose.\n\nTests are reworked to test this new functionality, with some minor\ncleanup to remove unused fixtures.\n\nChange-Id: I3335b5faac72e2e6962d0930eef0e3b704820bbe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/71b112dd8b790ec7a87ffac4c8d89c248fed7369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/71b112dd8b790ec7a87ffac4c8d89c248fed7369"}]},"branch":"refs/heads/master"},"982289a424b997ac6da7e381f2b6cba324adc2a4":{"kind":"REWORK","_number":3,"created":"2025-10-22 12:49:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/961684/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/961684/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/961684/3"}}},"commit":{"parents":[{"commit":"27106a683748678f5c3a66c0a509ca504360cf42","subject":"Stop storing setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27106a683748678f5c3a66c0a509ca504360cf42"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 12:43:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-22 12:48:53.000000000","tz":60},"subject":"Add pyproject.toml support","message":"Add pyproject.toml support\n\nAdd the ability to read requirements and extras from pyproject.toml\nfiles, eventually allowing us to move away from requirements.txt files\nif we so choose.\n\nTests are reworked to test this new functionality, with some minor\ncleanup to remove unused fixtures.\n\nChange-Id: I3335b5faac72e2e6962d0930eef0e3b704820bbe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/982289a424b997ac6da7e381f2b6cba324adc2a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/982289a424b997ac6da7e381f2b6cba324adc2a4"}]},"branch":"refs/heads/master"},"9b79bc474cf7884d04fa9d239abca1c5f6b0899e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-23 09:26:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/961684/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/961684/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/961684/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/961684/4"}}},"commit":{"parents":[{"commit":"32096476187f47cf05d0f42a832e63858ed1351a","subject":"Stop storing setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32096476187f47cf05d0f42a832e63858ed1351a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-18 12:43:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 09:26:03.000000000","tz":60},"subject":"Add pyproject.toml support","message":"Add pyproject.toml support\n\nAdd the ability to read requirements and extras from pyproject.toml\nfiles, eventually allowing us to move away from requirements.txt files\nif we so choose.\n\nTests are reworked to test this new functionality, with some minor\ncleanup to remove unused fixtures.\n\nChange-Id: I3335b5faac72e2e6962d0930eef0e3b704820bbe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b79bc474cf7884d04fa9d239abca1c5f6b0899e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b79bc474cf7884d04fa9d239abca1c5f6b0899e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
