)]}'
{"id":"openstack%2Frequirements~963356","triplet_id":"openstack%2Frequirements~stable%2F2025.2~If480238036e8fd3be9543d4b5a0d89b594a8088a","project":"openstack/requirements","branch":"stable/2025.2","topic":"new-release","hashtags":[],"change_id":"If480238036e8fd3be9543d4b5a0d89b594a8088a","subject":"update constraint for kuryr-lib to new release 3.3.1","status":"MERGED","created":"2025-10-08 09:12:32.000000000","updated":"2025-10-14 00:33:46.000000000","submitted":"2025-10-14 00:33:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963356-new-release","meta_rev_id":"772a98ee38e7ecebc8fafde75a7e50ef8af911ef","_number":963356,"virtual_id_number":963356,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 00:33:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-08 10:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-13 22:44:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-10-13 22:44:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-08 10:48:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-10-08 11:04:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-13 22:44:24.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"ac89de42fc2479cf65e5652e102ef4aa7d7334df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-08 09:12:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e7455ccdb1322954251624c4c0767535cbc0dc2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-08 10:48:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b6e1cc5e1c4747c27b1a64371baef42591a6dc47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 11:04:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35605e8b5baf4caf8a8d50d4ee6be59d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec79b53c25494316aade80eb12569ee3 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47e6c9ad367c46bc87a31349c0ccee08 : SUCCESS in 7m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e444f60df86a4baaacef1f561ad7ac39 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ade65ff68c9d4d86b21a50abd76ab110 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4beb8dc7adb34936906fee8e9d895e8b : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae218fb7f74c44029ca3aba431a10baf : SUCCESS in 4m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b020c66342ac44b8ab86a729bccfb9de : SUCCESS in 11m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/090f2404125f4fd7a211870f06cf6e0c : SUCCESS in 17m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e634aa4e34294d9688a0daa69cc9e0ee : SUCCESS in 19m 18s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e24207c1106d4e0bacb70921362fab23 : SUCCESS in 26m 59s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/223b3ea239a34b97b3eb7b6c8cf7c3c2 : SUCCESS in 6m 56s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/74317362fbde42369e24cd44544ed278 : SUCCESS in 5m 27s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/78fc1c1d07b3499587b7e3eda448f278 : SUCCESS in 4m 36s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c67f8f95751942358c4fcbd9d1b7eecd : SUCCESS in 9m 22s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bb36794bb89243a1aab6c8e8a537a682 : SUCCESS in 5m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9ed56590e9e84617b6692fd250391d0c : SUCCESS in 6m 57s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/648a58ecb7d04206a53db3507f08b26d : SUCCESS in 3m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0538b5731f5744078d6a3675049ac3db : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24a76b2daa164615977b02ae032244f2 : SUCCESS in 4m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d84a3c9df6984d709478c70e0dbe0eb1 : SUCCESS in 8m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/12f6d0ed5300420794b3597380d8f40b : SUCCESS in 12m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/71a1059f2c3741099ade9ceb78f33836 : SUCCESS in 2m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fd5816e7890841c299a0c9f76b66b65c : SUCCESS in 10m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e4fc45a65de4458f8e2c5a4d7257a396 : SUCCESS in 4m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e528f29ad54d4a39a7cf3a0b254e19a5 : SUCCESS in 5m 38s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/62bc300ad765471384948db832ce397f : SUCCESS in 11m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2ed9e4d318734791802755252d1992d1 : SUCCESS in 25m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0e344cc194c44c3c94e907d7190a2848 : SUCCESS in 9m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2ceb0d3f8add4efaadcdaf108a8b0a12 : SUCCESS in 13m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9a97c98f54fa4e309a8658369187b017 : SUCCESS in 3m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f3ee2b26e9424aaeb91b95a431167897 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e34bbbcc233477cb4443c835d17322a : SUCCESS in 27m 44s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/170a01862d7d4d66ac178e6bc7fc9550 : SUCCESS in 4m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c3645fdb85fb424dba2cb2e1a9a981cd : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1556dcf2154b4facab5bcab0f39e8b38 : SUCCESS in 6m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fa5cb3304e7b4d4eb42573506b6319d9 : SUCCESS in 18m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bc96a38dc8c040afb457ec5137e5cf77 : SUCCESS in 7m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/381fee2c776d41949aba3929ca051c1e : SUCCESS in 1h 06m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a070440e0b9498795c28c58bd5e02d9 : SUCCESS in 1h 34m 09s","accounts_in_message":[],"_revision_number":1},{"id":"50cfedf72a9378fbe457cb93d4a1fc366c611491","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-13 22:44:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fce95c9576e93c00e4feb04c2c6d90e3898e5cf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 22:44:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ad9e4dc114119d4b818a9e38c98b0202ecee360","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:33:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/681b599ab11d4ba3be9d7f39f0529275\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99bbd3c66a44430fa1eb8f069b9f58ef : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b8ff777125240a0b884e9c1f3374521 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5981fd5749d442fe93ebd76c3ab6735a : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9ec532e12c04d60b4d5c51d0fa926e0 : SUCCESS in 2m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db630bbee5e84fed93b346e449eec0d5 : SUCCESS in 3m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f90a6856f57646a7a4c0a7736d7f61ab : SUCCESS in 13m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/84d8ebb1cb29451fab976787c2c7fe4b : SUCCESS in 19m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0a818c435b53421fafd336eb45586863 : SUCCESS in 21m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/26f6d9dcb7c0474bbf1e606e2af1c528 : SUCCESS in 9m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/50d1c642b0f049b6a7aa90c78139b6eb : SUCCESS in 3m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6e90e43cdd3c4c0eab20176d4f11d6c9 : SUCCESS in 5m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1e86e9594a6748a9a1eccaade5b6fb49 : SUCCESS in 4m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c5552a3f10494e99a6fd27115ea47c46 : SUCCESS in 15m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e148b401f3e741cf9164332b8eeeb0d0 : SUCCESS in 6m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e5f18433547f4024bcafdff813b7e249 : SUCCESS in 6m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/49fd5128432447db96d1fae96039d4d7 : SUCCESS in 12m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/74a6704da8774eb8b2b8bf6bf6fb8af9 : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9eb0fa8332164b2a879c0af3c837bb19 : SUCCESS in 10m 05s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7a90cd38692d4889832fbc9d4e8b70ae : SUCCESS in 20m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cc68bd3660dd42d689244a7d96d56251 : SUCCESS in 6m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2644d547e46148f5bb40868f0a676bfb : SUCCESS in 10m 37s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fd6925fa78fb44a595c517c7ac9ff9f0 : SUCCESS in 4m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c0aed91101a942afa19fb63996758f14 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86334e80b43443adb0dfda2ad88957b2 : SUCCESS in 26m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/46646fdb0ce1412397889968e1a8196b : SUCCESS in 6m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ad66ea9fb627432294d04034ea8f62a4 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/293bfa17302645608a3fa8dc72b6758a : SUCCESS in 6m 41s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2e175bdfd36643409a8ba3a28fad20bc : SUCCESS in 11m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2aa1383c375540df96199341d5382245 : SUCCESS in 3m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f545aa542c3448ff96f2b4e8300cb971 : SUCCESS in 1h 03m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6988dd20c4f44bbb99ca25b94f9001b8 : SUCCESS in 1h 44m 27s","accounts_in_message":[],"_revision_number":1},{"id":"772a98ee38e7ecebc8fafde75a7e50ef8af911ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:33:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"328b8efec3cde506e93082d101a80f2777f633c0","revisions":{"328b8efec3cde506e93082d101a80f2777f633c0":{"kind":"REWORK","_number":1,"created":"2025-10-08 09:12:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/963356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/963356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/963356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/963356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/963356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/963356/1"}}},"commit":{"parents":[{"commit":"b76491a016dd12987f51a01b75625b405063f669","subject":"Update .gitreview for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b76491a016dd12987f51a01b75625b405063f669"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-08 09:02:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-08 09:02:31.000000000","tz":0},"subject":"update constraint for kuryr-lib to new release 3.3.1","message":"update constraint for kuryr-lib to new release 3.3.1\n\nmeta: version: 3.3.1\nmeta: team: zun\nmeta: diff-start: -\nmeta: series: flamingo\nmeta: branch: stable/2025.2\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta: release:Commit: Hongbin Lu \u003chongbin034@gmail.com\u003e\nmeta: release:Change-Id: I11cf6ff101961b7f74376ec4045986adc062ff29\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If480238036e8fd3be9543d4b5a0d89b594a8088a\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/328b8efec3cde506e93082d101a80f2777f633c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/328b8efec3cde506e93082d101a80f2777f633c0"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
