)]}'
{"id":"openstack%2Frequirements~963511","triplet_id":"openstack%2Frequirements~master~I86b65de76fb46f40d2b91437e866773357b1d96c","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-10-24 00:55:51.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}},"hashtags":[],"change_id":"I86b65de76fb46f40d2b91437e866773357b1d96c","subject":"update constraint for yaql to new release 3.2.0","status":"MERGED","created":"2025-10-09 10:30:55.000000000","updated":"2025-10-27 05:25:35.000000000","submitted":"2025-10-27 05:24:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963511-new-release","meta_rev_id":"2cb4b0c74efa108a1c1c4da9e683b94c43488c57","_number":963511,"virtual_id_number":963511,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 05:24:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-24 00:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-24 00:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-10-24 00:55:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-09 11:39:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 11:45:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-10-23 09:25:46.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"dbb34b02a8bfc2236ccad47380937baf3f0dd8d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-09 10:30:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e9b0554f23568a3462c9a4abba330719055f017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 11:39:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40a87f20c131449b8c3178a882fa24ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512ce29a373949a382e502f032bb2b78 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc2da9c59df49f487e7f7b398ceb2a4 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63641f7631074eadaa89ec93c270c041 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f4f8546ecf247b28dee11c0872065d8 : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98e8300eaa394cdfb67481120c4483b7 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ac3e7e72e4e4618955aa6482ba8fca9 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/396d20c7a81e4e19b5e9007b7fd9feb0 : SUCCESS in 2m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a3109d7906b24e18b7df4f1f0d30b4f4 : SUCCESS in 10m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6fed73f3557d4c9fbb5b9f3523c5a74a : SUCCESS in 18m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2837f571a8cc472ab378f6e90bbb2a3b : SUCCESS in 18m 57s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cbf74bebfe7d4b6e8f5d5611ef0fa981 : SUCCESS in 24m 28s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c675c73c22af46c2b276bcd4befcf0af : SUCCESS in 4m 48s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d9c39550abad477ebb4e64edca45045e : SUCCESS in 3m 13s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/fb016be618e640518b17b4a67e15bdd2 : SUCCESS in 3m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/28fcccf8048f40618400471972b9cdff : SUCCESS in 9m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b99b15c5c03d4431be89df613c13469d : SUCCESS in 3m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/827723c27f4d4689bfb3dbb5cc36b735 : SUCCESS in 5m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/76848e7a3d504855a9ec5dbbdb3b193c : SUCCESS in 7m 37s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/836918ce8a7d403f875e2204a5fa8fdc : SUCCESS in 8m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f41bd708703044ce9ef59de07675adfa : SUCCESS in 9m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b2c04c69a4ea421384c26661eed087b7 : SUCCESS in 9m 56s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e7d275aa657947a9bf75a54dfecda168 : SUCCESS in 11m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/72d603546304440688e1f46bf94a4ecf : SUCCESS in 1m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/394846099b964960b7537b066079b51f : SUCCESS in 10m 06s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d057109a944442799c0ca7973183fda6 : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/974c799be3e44db6b4fa59fee08943ee : SUCCESS in 3m 56s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/529e1208852a4e488fe8d32b8525b7f3 : SUCCESS in 11m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ac38a067b48b4e6e92543a22df50cd41 : SUCCESS in 20m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6b88f8a8681545c1aa0adcbf7b7e4f4c : SUCCESS in 7m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/474826e4a5bc4d8380fa6882a9a7bfaf : SUCCESS in 10m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1f11f8ce3eb54e8ab42483d596c60ca1 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/17fd30dba3664b3ca6427fcf3ea97bf7 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d30198f14a604b778650e894430b4f3c : SUCCESS in 26m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9f323cd965274be5b95a102a3b5dc2b0 : SUCCESS in 8m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/36cb68dc7a7b4bc5969e86db0d00de43 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ee09dc174804b879b67364b1480f9c0 : SUCCESS in 6m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/df7ddbe26a3542f7a10643078a448aef : SUCCESS in 7m 42s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3bb9c103c4314640bc95433f6b924c1e : SUCCESS in 5m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8013cb18ba543559d27913eeeb3d631 : SUCCESS in 1h 03m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5143c0989d24f53beea63d7a8964ea3 : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4b3299df4aa0b2b1f767654f1c2d786bd1ef1098","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-09 11:45:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0250eed84cf401f426b5766a7479fd40299f8c7f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-23 09:25:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89ff6a7b4615ece1147872d99f3715192c5aa0ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 09:26:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae38fab0cbcc140755b2f7603cef92bca296648f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 11:04:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c93ef011e404dd8a0229004473c6381\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a856bb462d3c45d78a7a170cebbf7212 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1645b4bcb0f043b2b5a8559225650867 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a1fb2fac47e45a49d2b5c9ec6aebd2f : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b24407269d34a6aa88e8876d80c875b : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae95cbf8165a498086893c93af5dc325 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c1cf6d97117a4a0ebb29588001f862c5 : SUCCESS in 3m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c5180da07c89405a928e50c33f36c89a : FAILURE in 4m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/65703b36b8d14ba093f2c896a1af9303 : FAILURE in 5m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ed09066e46f94146abafe6790b7e5711 : FAILURE in 4m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b247ad26970e41bfb901bdad8590a920 : SUCCESS in 7m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d189eb8465a841d4bbbe815f063d4552 : SUCCESS in 5m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6e974185e1df4aac97c65c1453d39027 : SUCCESS in 4m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b00b5673d8844a7695c42713ff3a6247 : SUCCESS in 5m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/67c7c14ca98a4e7bb741e37b98512507 : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bdd3e91316ce4f44a3175206f7515b85 : SUCCESS in 7m 08s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3d6382e3fd4b4c78b289dd95f5d8fb51 : SUCCESS in 9m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dc68b7378e4e4d998bcfff9b7fd0dd83 : SUCCESS in 10m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b3980407694f4f6b8bbef2ad058278f0 : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5a4bd1aa927f424bb7dfe45c757f8474 : SUCCESS in 11m 10s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fc6d3fdb476d48d9ab8d0af436b55abe : SUCCESS in 28m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/88b0eee342d54522b6279b09294aeedd : SUCCESS in 6m 37s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1cd27166e3f140b883c84e36d8fae1aa : SUCCESS in 13m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0b2833b628e3405086da244e954ed178 : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/94fd1c9eaac8411f89f6fd609246bc18 : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1950d46c9724c8694362bb13104cf5d : SUCCESS in 20m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cc5777e594454d49b2f52ade527f7656 : SUCCESS in 6m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f742cc856919491cba6570d31fb8cb10 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67dd478f6f1548de94c1683eb2b83b56 : SUCCESS in 5m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/56ce7eb069544b6fb41fbbb7d7d64c20 : SUCCESS in 11m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0bba5606797f465189b09cb5322d87ee : SUCCESS in 7m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/628f4f73a87f4c23b284f9cfe3403766 : SUCCESS in 1h 00m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/024b996016304fd59751f725425d4b89 : SUCCESS in 1h 35m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b2afa289b094c5673341e12a0243b54c1b442bef","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-23 12:05:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"603944355678779eee8f813658cdc23b60d21349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 12:59:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c40325a9af354c95a448224f1ec6d8e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/762e1aff7d2b4c338aa517a6762032fa : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2f254f2ef39477389edc1441bd271f2 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bb1298c755145a7b567bea1770b280b : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/262eccb46c134df6815ecb9be678d0b8 : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/605de23997324e9fbcd259ac2183916c : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3bd3e6e6dcc747cea9cd4b20b73f6397 : SUCCESS in 3m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f96a6b423bd54059b514d3e799c26f37 : FAILURE in 4m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/adefcf1d96654e058855c6413aaf4b10 : FAILURE in 3m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/81fe2ef5eee54ca9aff4f985cb573483 : FAILURE in 3m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3ab7cc30748a4e829f5fba15823074d4 : FAILURE in 4m 49s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/798041221b6646adba13c8881f88134e : SUCCESS in 4m 58s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3312e840aae740558399b3456fa52646 : SUCCESS in 5m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/09d83f33c9f344d9bfcc171db54bebf7 : SUCCESS in 4m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0c49ef9bfe9242b2819a3abd0bd572a6 : SUCCESS in 11m 59s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/710cfefda6834fed87534119b4d7efae : SUCCESS in 5m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7f211aaea05846cd84353556cc88f41b : SUCCESS in 7m 11s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/71b885749b4d4790aad62dc8e6f32a6c : SUCCESS in 4m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6576f0730f5f4d12898b3e4bc72e04fd : SUCCESS in 19m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec82cf5d467a4b38ac235c943f868acf : SUCCESS in 6m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8f0c333035714cb8bb2bfedf0542de92 : SUCCESS in 9m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b27be7714f8643c0a601cfc86e4fff88 : SUCCESS in 10m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/bec2f21e9b6d407486f156219754e8b9 : SUCCESS in 2m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/cff15e3aeb3c41a780582c87e801c972 : SUCCESS in 6m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/05123be02f03444d81992cccc1570dec : SUCCESS in 4m 40s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4e6c0017fe6a439ba643fd09b056423b : SUCCESS in 4m 31s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/70f99b0a38694ba492656a19249930e8 : SUCCESS in 12m 56s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0c6649399c6940e0a7ac5a820c3d077e : SUCCESS in 27m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5d968bc1ecde41bd962d87a76df5d03c : SUCCESS in 7m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b5217535e28e423cb98640af1464dadc : SUCCESS in 15m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/df759f87a29f4482851b4e14fa8accd2 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/191ed9c725d94ea28babfd787393d4ea : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0be2cb6e216e4497a76036cb85b73a43 : SUCCESS in 21m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e1a3f056bb5b4972ae259ad70265504b : SUCCESS in 7m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/171adc1279b443739e46fdef8d6e7b2d : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5067c37b42fd41b8a75a8921fd7bdad5 : SUCCESS in 6m 40s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8e33bba31da84dffb327d5596b637389 : SUCCESS in 13m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dfc95c9cb88d459f97ec83144a271d8b : SUCCESS in 5m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/147ce934349548b89b007ca0f3912508 : SUCCESS in 43m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76726f13301f4966bb2b83e96ee2d36a : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ce5b42fe8c23a0f6d7b5daa9fa8a33f441b6055d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-24 00:50:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"101d0e32bb833c4e56dc7025fcb888cead36115d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-24 00:55:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81b3882523ba566b7868b9a5877549ff03795d89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 02:38:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/883d400fd1e2448386cd2e59fb5ecb53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cacc94b729b947cf9718e83e65311613 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b54118d7a5f3485b8ffb7c927487816f : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5efcaa1697e64ea0a2b33b9ea8d9b979 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f836114c48b4acd98c1896874b4ecd1 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48481bf3361a4e45a6e84dcadb0a0adf : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9fe763503349416b92047822917f3bd7 : SUCCESS in 2m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0c0773eaf0304af8ae6f42ec3d90c64b : SUCCESS in 5m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/998b0fb946c54f11b25fdada3265e623 : SUCCESS in 7m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6d729b0169c54c30bbbb840c8c5f4113 : SUCCESS in 7m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ec2051efb1e84dcab0cecda68df50018 : SUCCESS in 10m 12s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/25020a2c8f62482680a2e82290b88342 : SUCCESS in 4m 46s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/79aa521c820d4ba09ed1cc2c638f4695 : SUCCESS in 4m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/f6b1d8501f6a4dea80c9ed5cca2e0263 : SUCCESS in 4m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ddec17dfae2b443890edf0f6ab95ff5a : SUCCESS in 13m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/45f53940300b42d0aa9367fe6dadb7fe : SUCCESS in 5m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/13b0f0c5fe984ff09e7050cda762ac40 : SUCCESS in 5m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4c4c776b88ca4ab2aeab3a49c0f0431c : SUCCESS in 7m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/35a8ab14c8774e9d8de038285f26b26f : SUCCESS in 18m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f1a9a6413264c54849f95dd011e5808 : SUCCESS in 7m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4b3af38bc7da48b48a9b6eb3a8437cb0 : SUCCESS in 10m 23s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9041a043c901484496cdc97ac887b422 : SUCCESS in 14m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/074fe33e2f434da8b5264d4810c94669 : SUCCESS in 2m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e3499bc6c3e144efb9debb596d1f5e97 : SUCCESS in 11m 27s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/26ce5fcfb8694c0ca53f3136c83dff3a : SUCCESS in 4m 40s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9383fc5670864ce0b1af690f59a5653d : SUCCESS in 3m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5b5dbc3bda9044a398cdf95fede9edd5 : SUCCESS in 9m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/862b89105c204b7296e03bd57a2d76ed : SUCCESS in 25m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/606265944d994131a5daf8ee78803c44 : SUCCESS in 7m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4a8337463ce04e97afbe63b1a1d44f09 : SUCCESS in 10m 43s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1d89ca33a430465fae90c4b232774b38 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0dd03c363ce94d61944fc063544d6664 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/879ad58c8e3a4c569c068345a847588f : SUCCESS in 21m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/30612ec2b08e4cd0b0f8872c58904fb0 : SUCCESS in 8m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/424a6ea1758d475eb6b5a603b3c92d18 : SUCCESS in 3m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12574d4bbaea4aa2a0c51f08f8c869e2 : SUCCESS in 3m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d3d616c94d094c49a62457b104ffebbf : SUCCESS in 8m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/126b5e17b6b9463487153efae883da28 : SUCCESS in 6m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/331a611266fb4bc3b8675ddca6edf522 : SUCCESS in 58m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab0372dc58d343f3924fff67d0063c8a : SUCCESS in 1h 42m 17s","accounts_in_message":[],"_revision_number":2},{"id":"100c55d9c4eaa45e8641274e28aa444de0fb3648","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 02:47:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"59ce32e5572cf832216ba4bc773b7280d772fce8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 05:24:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc530de0e5cf4d2285fbb5f002d8db3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e612a2fcf7714a459ad9e296018710b1 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa29d8fd1057440783c8ebced217049c : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69895d504ee14898bdceb85cbe2d08e1 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f438e81741884c2db796ac53777d6442 : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24d18a6111eb4cc1bf6b797e5f83b702 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e8902db828f24ed98468f341b74397d0 : SUCCESS in 4m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/93b5de70fc4e432fbbd21ed31caa45e0 : SUCCESS in 5m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/23759511752143a39ab56cddbdd9bf01 : SUCCESS in 7m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/12f58ce0a0cf4e7290cfa4e16683ffe0 : SUCCESS in 13m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f78b552df2284b0ca343eaf824179095 : SUCCESS in 11m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4f7fa4f750b64f04a0c4c465efab4bfa : SUCCESS in 5m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/272165e5b0554deba02a529233e99d41 : SUCCESS in 7m 15s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5c29ea13ad804611a0b1a1272bdfc57f : SUCCESS in 9m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7269844b44fd414dab54f9e70abcb74a : SUCCESS in 17m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a4b92663a4249858b29241067211464 : SUCCESS in 9m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/25cfc55a50bf4325aa6e30aba186f022 : SUCCESS in 8m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/fbf0fe26da784264b08c94c0ff2f3eef : SUCCESS in 13m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1a8c528de3e74fc7829a18924df0c849 : SUCCESS in 2m 03s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/418c406cfd67410089d002d711beeac0 : SUCCESS in 11m 44s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e0c42643af8d4b5ab25e02e38bdf40d3 : SUCCESS in 23m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3cea88d9d00c471aaeb0f92e2134ba62 : SUCCESS in 7m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e809d32c4f7e4876910224899f044d7d : SUCCESS in 11m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/78e33d65ee824d93ad4efe2e2ffd6b96 : SUCCESS in 7m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2ceb2310407341d7951373a5e37cded3 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/41d657ee9675463998a16f029006e87f : SUCCESS in 29m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fe2c30e79e6348bd8411aced4d5ed726 : SUCCESS in 8m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d3e5649bda7b496991e318dc819786ba : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12d822e39e824ccd9efaf1bcafe5486b : SUCCESS in 7m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/09e3866e12dc458aa350b72b9c85d55e : SUCCESS in 18m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3e562ad6d5974931b6883ff743d1611d : SUCCESS in 5m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28bcf88c7c9e42eebfc88fb87766db0f : SUCCESS in 46m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ddca6638a1b44d09497e21ae0fbabd3 : SUCCESS in 1h 39m 51s","accounts_in_message":[],"_revision_number":2},{"id":"c79a7555a69cb7a814b673a8c307215cfe0745a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 05:24:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2cb4b0c74efa108a1c1c4da9e683b94c43488c57","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 05:25:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f13dc3cbc95a4dcb80011ef7edbd2fe9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954ef4e233024119834c5ec46b3a3723 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"856df5bd49cd2af983fcda85001ebb79e812f228","revisions":{"2bd7d8fb4b66a9e7ae0403b5dc736e29734a3a58":{"kind":"REWORK","_number":1,"created":"2025-10-09 10:30:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/963511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/963511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/963511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/963511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/963511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/963511/1"}}},"commit":{"parents":[{"commit":"af5fa8a98f1f093d01d87f8e57579e14e7ca5f61","subject":"Merge \"Deprecate ironic-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af5fa8a98f1f093d01d87f8e57579e14e7ca5f61"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-09 10:20:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-09 10:20:54.000000000","tz":0},"subject":"update constraint for yaql to new release 3.2.0","message":"update constraint for yaql to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: team: heat\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Ia38dd74d7e5725a387e9fdcf08cd66fa7fec19f3\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I86b65de76fb46f40d2b91437e866773357b1d96c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bd7d8fb4b66a9e7ae0403b5dc736e29734a3a58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bd7d8fb4b66a9e7ae0403b5dc736e29734a3a58"}]},"branch":"refs/heads/master"},"856df5bd49cd2af983fcda85001ebb79e812f228":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-24 00:50:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/11/963511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/963511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/963511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/963511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/963511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/963511/2"}}},"commit":{"parents":[{"commit":"6b3dda33d305df109b9cb50e7c0a9de6bd6dd567","subject":"[check-uc*] Create essentially a \"no-op\" dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b3dda33d305df109b9cb50e7c0a9de6bd6dd567"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-09 10:20:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-24 00:50:23.000000000","tz":0},"subject":"update constraint for yaql to new release 3.2.0","message":"update constraint for yaql to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: team: heat\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Ia38dd74d7e5725a387e9fdcf08cd66fa7fec19f3\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I86b65de76fb46f40d2b91437e866773357b1d96c\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/856df5bd49cd2af983fcda85001ebb79e812f228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/856df5bd49cd2af983fcda85001ebb79e812f228"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
