)]}'
{"id":"openstack%2Frequirements~964059","triplet_id":"openstack%2Frequirements~master~I6633770c3533c6510f8331831ac3b900ed3421b4","project":"openstack/requirements","branch":"master","topic":"feature/standalone-networking","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-10-28 01:52:01.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-10-21 20:17:24.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_12898\u003e using the hovercard menu","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-10-21 20:17:21.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_12898\u003e using the hovercard menu","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"25437":{"account":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"last_update":"2025-10-28 03:32:59.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I6633770c3533c6510f8331831ac3b900ed3421b4","subject":"Add networking generic switch to requirements","status":"MERGED","created":"2025-10-14 18:37:22.000000000","updated":"2025-10-28 03:33:49.000000000","submitted":"2025-10-28 03:32:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964059-feature/standalone-networking","meta_rev_id":"edc491abd027aa6f7ad76a1b90927fc2069e0989","_number":964059,"virtual_id_number":964059,"owner":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 03:32:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-27 17:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-28 01:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-10-28 01:52:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-14 20:09:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-14 21:50:22.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2025-10-16 23:03:03.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2025-10-17 11:12:02.000000000","updated_by":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2025-10-19 23:11:02.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2025-10-27 14:59:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2025-10-27 17:36:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"62670f7807aff2791c6db0763ab7271ac9b49417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-14 18:37:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"815aa8ac60264b361d164d0889c2a825df21c74c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 20:09:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdb4b821cfc14d37ac03354be94e82d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2840cc86db542d2846e6b2b0bdbcfeb : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdfc68c55ad04623a06a3494cb2ce3b9 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/591d4fd9bd6d4e0aa0b5e3c11de4d839 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86f2bf3472bb41a3ab539bb9f9dae91f : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a9639e4b5c1471caa32d88a4e83ccad : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76f9ee3448264ffea195f039665f15a0 : SUCCESS in 3m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b3d3d3f91fde473ca598dc44054410af : SUCCESS in 10m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ca004d28ec1a47a08e0fcecf38a134f8 : SUCCESS in 19m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9122dc4bd21e4e6ca7cc1203de5f67f8 : SUCCESS in 19m 48s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a5fbf7212f0b4118ac6712852c88256b : SUCCESS in 10m 12s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6f521a6808af4d1fba9ba2c9685f1874 : SUCCESS in 4m 52s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d1933de0dde84f3a86dfd6f62dd1df4f : SUCCESS in 4m 08s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1bb6fdccb62447b883960936069d2aa0 : SUCCESS in 4m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/618bb66fbfaf472aac0a214f30d78650 : SUCCESS in 11m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/623e31434e7445359828db084da25774 : SUCCESS in 5m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/795d5d280a534ac3834cb4d252fcd6af : SUCCESS in 5m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/de84323c02f44fa8bae0e5c005559384 : SUCCESS in 8m 13s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/77327d0b2ae54c8aa5113bc45939efea : SUCCESS in 8m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4343834da8ee47fe8ceee15e1437130c : SUCCESS in 7m 10s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fb501be08d0e4c2ab721918995cd3f01 : SUCCESS in 7m 37s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d21e7dba5038438e89f2b11c5b9bda4d : SUCCESS in 9m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/39724d2c02d24dc28bfe1648fdbeba13 : SUCCESS in 3m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e4ddca4ddd224d6bbc4212a2a7b952af : SUCCESS in 9m 12s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/73774d7629ee447ab6db4aa364455a3f : SUCCESS in 4m 06s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d7f67fad26014ea8ae0c56c425893612 : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6b250cbe813d4e0e8192fdf9f72aa890 : SUCCESS in 11m 08s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/512595f8480e4d839afbc52089489b06 : SUCCESS in 22m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7998f7746f904e4aa15507fe580783e5 : SUCCESS in 4m 25s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9d650f2bd09e407bbcb872a88f27502a : SUCCESS in 8m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b3927746246d4bc8ab2818bb3c64aa2c : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c31f2169c91944638de7eecf4089d30c : SUCCESS in 2m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd52afce74ce4ae1a326ae6e8c444e3b : SUCCESS in 21m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/78406963561340eeb4a4a271e09cae3e : SUCCESS in 4m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a69922be2c574ae7baedf63ccfa1525e : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f9835f6d40e148b28c6e9db055027acc : SUCCESS in 8m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/94438326c62a444990ff7814ab69f152 : SUCCESS in 13m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/11bd4393872e4c4790d602b32e8ee739 : SUCCESS in 5m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37653339d6de4de49ac5abaa395861ad : SUCCESS in 49m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1f4928ad71f409e9f973140caea8d29 : SUCCESS in 1h 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"72389958fa3d9e7220ffdb21ee7341196bf7b4c5","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-14 21:50:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8741e905b5a282190e3900a18381f80fb5546b8b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-16 23:03:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f71879f7daab79a46bac98aeeef7980130b42894","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-17 11:12:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2726eab4f36df3a9fe4abc76464ea49302fe367","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-17 11:13:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f51c48d384281930a555ffaadea588a3a7ca3597","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-19 23:11:02.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4bdc4d03efbaec4f8f7bb5e77783fd951a69450d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-21 20:16:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"386c67534c7830d7a026d318e264b855314c173b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-27 12:00:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fa8efb39d7bfabcc12f20c78b2032393fd8fbb61","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-27 12:01:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd6181b945ea5a2c3aa1d2b71802eaa543f5ea8d","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"date":"2025-10-27 12:01:50.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"7c004f5bd2bc1a0038ecf1d3d32460ca858193b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 13:45:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81144ea0295a45b881a009242e0e2d50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a75e32061d5461183383957da50af9b : NODE_FAILURE Node(set) request cd7b8f9e399d4560a19b7536ce3a406e failed in 0s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc5a41e24c5c4126a1d00676cb091c29 : NODE_FAILURE Node(set) request 26c3547d25c64159a43aef9eaea35737 failed in 0s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5b01f1a160f4f25908e6bbe891c651d : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04cd63aa26504266a6a49a91cdc05250 : SUCCESS in 7m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fd32bcc4fe943bfb39bf81d05426293 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e73a7738c3ca4f70bc9520de86e460fa : SUCCESS in 2m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d52bdaa9e00542aba076b6ecabf8959f : SUCCESS in 3m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b1b2e37798db45168793e4a99978ccfa : SUCCESS in 6m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f8723a068b4f4814bf686962bff81c82 : SUCCESS in 13m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e171950944a54c65986db7b17d951b1e : SUCCESS in 10m 17s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/55c3f5b31663470d97b1c23dba877524 : SUCCESS in 4m 58s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5827b0d2d2a9454c943de05a2ec15893 : SUCCESS in 6m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4c0364f36e0f4f14a60780617a2d1eef : SUCCESS in 2m 42s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3e9a606650d64bfebe916bb904a93b2a : SUCCESS in 12m 13s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e12b4518c78f4443a3f4b99263885043 : SUCCESS in 6m 34s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b77fd3db2a1f49ee858d32b90e8a8427 : SUCCESS in 5m 40s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/dcb5598689404befb78ce6a506e1c11c : SUCCESS in 6m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/53bf787d642f412288abcc2115030c4b : SUCCESS in 16m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36f52f29894d405db36db55dd2d294a0 : SUCCESS in 9m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3ef990bfd7044c0e9f5adb8d8a1e7c07 : SUCCESS in 10m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/511d7813374342c7b763caea1bfb61d7 : SUCCESS in 14m 02s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/adf40b064d324b208422cc560bcbf90e : SUCCESS in 1m 33s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6c7c8b91e62f4652b3df076b09f6733e : SUCCESS in 7m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/9a474ad2b05148f0bf181279ed2e19ce : SUCCESS in 6m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ef6c9f84a8c9414eb2f86ae5c453da79 : SUCCESS in 6m 51s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3ed87b15aab845aa8898d90133149ff6 : NODE_FAILURE Node(set) request 1ffd76ce16f54a9889dbabea63a59054 failed in 0s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1a18207b30f740a5b45a8c85173bc707 : SUCCESS in 25m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d82e311834984261b3840fcdf0b969fc : SUCCESS in 8m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a910e7866ac84f958cc2148c05eabb05 : SUCCESS in 15m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/824b8bf980fb4f2c823f4192e6a2230d : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/950df68d509344c4948990faccf52db2 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87e1b4e8a42c4c60a0e2b5ecff8f3f7b : SUCCESS in 27m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7b6b65a62b5b4aecbdf5abf31b302e8a : SUCCESS in 6m 31s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fe01830f7cc741fb825e4bd3d5991952 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e79d3e124684ddbad8cdbc54bdc005c : SUCCESS in 6m 40s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5a872befb7904831af1a471c4c6b6908 : SUCCESS in 18m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/dce8aac089fc4534b0f9243c0e2d135e : SUCCESS in 5m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5905121e09f458a852d21cfac10c8ce : SUCCESS in 1h 00m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42db4b02e7aa40519af205bb9b544ba5 : SUCCESS in 1h 41m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a54d1299dfa48ca839d6a2d7c30ca6e4240178d0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-27 14:59:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4ec28c488ee2f1bfd1437e3daafca86f72dbeec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 16:08:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7adb2d6fdde744259e83a9e8b465cc4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9243d155564b2a99adb4969357d563 : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbc662d798274e85a5eea9c57070b1cd : SUCCESS in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ee8d1fef13a4832bad460e56ece6cc9 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cfda046a8a94f71b168af385afe5667 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49af89d7dc534e359e3ac25138043f28 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4da0f772da54478bb13a5f144945277 : SUCCESS in 2m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4ab922ab88cb4ba4b354c685c1e76434 : SUCCESS in 6m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d73926267caf4fb79e4c1770a21e7c7e : SUCCESS in 3m 42s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a265492c6ab443a497fe22e8e2d0980f : SUCCESS in 6m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/03f90f96ac5d41c697f81f685789bf6d : SUCCESS in 10m 04s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3d7d31d01e334e6b9133bd5180e11ae4 : SUCCESS in 5m 10s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8b8b913752f04380a7d2f5b5a02326cb : SUCCESS in 3m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d2bddafc40e24accb4a902e7c277a818 : SUCCESS in 4m 23s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fc23d0a7fc794f0393a2cbed762f6721 : SUCCESS in 7m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4e2a44acb98c4ff0b6bef9a68d9679f3 : SUCCESS in 3m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/edd1d57412914100b335c90191ec0a21 : SUCCESS in 7m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9553dfcd5b844d87958cf829975027c4 : SUCCESS in 8m 27s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/515426f2714f4ecabb335517c78b694e : SUCCESS in 17m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94336dd395f0439abb6e2ea2c5661e6b : SUCCESS in 9m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9e17cc24a0f54fafaee00c9715efed4e : SUCCESS in 10m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/fcf13386e6154e60b998f1009a44d1c3 : SUCCESS in 13m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/532df996f76c465baebd923de008ec12 : SUCCESS in 2m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a4beda104a8f4f73ba7781bbaf4b61dc : SUCCESS in 7m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8956455fb82845b5ae47151c57ae34b5 : SUCCESS in 4m 45s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1dd78eec9a574f15bfb703d88e9bf87b : SUCCESS in 3m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/1f78d4757ac143fdbdba299e599d2b03 : SUCCESS in 11m 44s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2928fd85958c46d9b62ca791c09db1bf : SUCCESS in 28m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/07d57be384d14fd78c025e961b4cea0a : SUCCESS in 6m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8dc1bc59582943b9a3814bcbde651b29 : SUCCESS in 12m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8a00fc55d38b41fb8e62345807f94c6f : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c1ce5e226d14157ae38357fed73d6c6 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a0548f7834c4bf39d0a7d1cb0447757 : SUCCESS in 32m 48s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/facacbc4b9b149f282bd8e5e7143c0de : SUCCESS in 6m 59s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c404d036bf694ee5938ac06700f016b1 : SUCCESS in 3m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/161791b72793486fa7f9071774a79a8a : SUCCESS in 7m 24s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/44598e09c9f14da3857c6620240fc3a6 : SUCCESS in 21m 04s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9d9d7f71343f4077ae6abe46c6f6dd46 : SUCCESS in 5m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f87fa762e0cb4844a88f9de8137cfa8c : SUCCESS in 1h 06m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2364facbb60b44559723b96cda027c47 : SUCCESS in 56m 39s","accounts_in_message":[],"_revision_number":2},{"id":"59e3aa4d168c28116777dcc7821e02ce003517d9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-27 17:36:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e8c869a78945530a20c9835af44bbe720e005e8c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-28 01:52:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8209a2f2d08033cc594ddeec8ec0773d4202a5b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 01:52:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4cc1724e50d0153d519538e171623dd52f914135","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 03:32:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e312041755944b55b9a3660ed0306c15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb603601aef4464b962d3d3ee3f56b9 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e98584fde5649f68c6b93801b58778a : SUCCESS in 2m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/488164fd58474ed29cf509f714c45465 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/686c903803fe4a2baf6d063c64f60fff : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c92ef708f1a47e9a507d4793a46782d : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d1766ee850234d94900391e3fe2a4833 : SUCCESS in 2m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/53da456872d64cff8abfc48a448ee6d7 : SUCCESS in 6m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/aa99e2aabd624f209a7c481f075cee85 : SUCCESS in 13m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b8ef604ff6e24723a8ddf1a9a9336fb2 : SUCCESS in 3m 46s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3ac36f8735144af585c43f876ddb9ffc : SUCCESS in 12m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1a08ab73c7444051ab9bd7d0a01b0b2e : SUCCESS in 3m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/735bb1df06e945b9a0e266c1d4feb6af : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/824f518186124eaaa8ceab515ff83e78 : SUCCESS in 6m 13s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4bca234d54634a3b992027a35441cb0e : SUCCESS in 17m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ab468f8cf3e49dcb37f45a5d63eedc6 : SUCCESS in 7m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fdddfe593f734c6b86c4ab474122a6d0 : SUCCESS in 9m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bc1abc57e3bf47b48e3ac8de5268a29b : SUCCESS in 9m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/aadde167c14f43229ecab219808cd913 : SUCCESS in 2m 37s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/75fa33fb9235498b8bbbd5ee3a0ee235 : SUCCESS in 8m 26s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/69186830941c4d719d66e1d3df736d13 : SUCCESS in 24m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6da0cd3bcf674f7c9cb32073419dd4d2 : SUCCESS in 7m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fa0b9d0131424b5ea8424400bce3ff2e : SUCCESS in 14m 28s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a833c98ce5df442d84b5e7c655b42acf : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/01d002057c524ac28ea2ced1ee26b0c9 : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/caddc413dc1943c8806d8f73ffcd8fac : SUCCESS in 27m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/659210069bfa4cb7bafb80794ad5a9e7 : SUCCESS in 5m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f251556a0e454ca5ba8966dfddc87f25 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb460a87b6fa4fa18d277dd24b494b2b : SUCCESS in 5m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dacf5b850cc04143aa88f755256f4dcc : SUCCESS in 19m 48s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e850492d8778440e88821eea3a47c5d6 : SUCCESS in 3m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd353c707da9477e81eebba8a0b7d51c : SUCCESS in 50m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71d185dbf3524106a188119472c2c839 : SUCCESS in 1h 39m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a55ffadcf38f3a87aacf6285f1db7f7d18615220","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 03:32:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"edc491abd027aa6f7ad76a1b90927fc2069e0989","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 03:33:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/502ceeb6a03746d994407dfee03c46df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cfc780b5604f8aa4a58aec46caeef9 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ebb2c96aace18b35c2ec566c905c8e6cab2e75bd","revisions":{"65e0dbde54b019c382f7bece755325ebd4ed632d":{"kind":"REWORK","_number":1,"created":"2025-10-14 18:37:22.000000000","uploader":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"ref":"refs/changes/59/964059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/964059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/964059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/964059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/964059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/964059/1"}}},"commit":{"parents":[{"commit":"73fb1fe9fc5f158ec49e33a79be73ba3641800f3","subject":"Merge \"Bump pifpaf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73fb1fe9fc5f158ec49e33a79be73ba3641800f3"}]}],"author":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-10-14 18:33:37.000000000","tz":-240},"committer":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-10-14 18:37:17.000000000","tz":-240},"subject":"Add networking generic switch to requirements","message":"Add networking generic switch to requirements\n\nThe networking-generic-switch is being added as a driver requirement\nneeded in unit tests for Ironic so must be added here.\n\nChange-Id: I6633770c3533c6510f8331831ac3b900ed3421b4\nSigned-off-by: Allain Legacy \u003calegacy@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65e0dbde54b019c382f7bece755325ebd4ed632d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65e0dbde54b019c382f7bece755325ebd4ed632d"}]},"branch":"refs/heads/master"},"ebb2c96aace18b35c2ec566c905c8e6cab2e75bd":{"kind":"REWORK","_number":2,"created":"2025-10-27 12:00:39.000000000","uploader":{"_account_id":25437,"name":"Allain Legacy","email":"alegacy@redhat.com","username":"alegacy"},"ref":"refs/changes/59/964059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/964059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/964059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/964059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/964059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/964059/2"}}},"commit":{"parents":[{"commit":"b440fdb48633591531976d052ed1bfed53c9c9f3","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b440fdb48633591531976d052ed1bfed53c9c9f3"}]}],"author":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-10-14 18:33:37.000000000","tz":-240},"committer":{"name":"Allain Legacy","email":"alegacy@redhat.com","date":"2025-10-27 12:00:27.000000000","tz":-240},"subject":"Add networking generic switch to requirements","message":"Add networking generic switch to requirements\n\nThe networking-generic-switch is being added as a driver requirement\nneeded in unit tests for Ironic so must be added here.\n\nChange-Id: I6633770c3533c6510f8331831ac3b900ed3421b4\nSigned-off-by: Allain Legacy \u003calegacy@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebb2c96aace18b35c2ec566c905c8e6cab2e75bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebb2c96aace18b35c2ec566c905c8e6cab2e75bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
