)]}'
{"id":"openstack%2Frequirements~965151","triplet_id":"openstack%2Frequirements~master~Ic9885cf58cca1ef560c2a25134fe5402fd32bc82","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-28 22:26:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic9885cf58cca1ef560c2a25134fe5402fd32bc82","subject":"Fix keyname used to store extras","status":"MERGED","created":"2025-10-28 16:28:57.000000000","updated":"2025-10-28 22:26:16.000000000","submitted":"2025-10-28 22:26:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965151","meta_rev_id":"a89691c54bd54024bdcfb77848175937bd24c67c","_number":965151,"virtual_id_number":965151,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 22:26:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2025-10-28 16:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2025-10-28 16:43:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-28 16:43:51.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2025-10-28 18:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7ed8c218123e1f118d40d47642e8551673e5985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 16:28:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"434d6544f56c5bac07ca0bce918c804d903386f4","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-28 16:43:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01ed30bf7deca42993a87356da01958c68c85a5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 18:15:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97c0292d67b44c5daa07542adf908ebe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5fa868b0764c12802dd0b2c7b808a8 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81f1b680089a4310a0d74139b32c7a27 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7c7c5dc4c014827903bbdd06a47d2de : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5eba954fda0441eb3762441610e9bc8 : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec12fc6d67ab4377b9736b9bdf831c3d : SUCCESS in 3m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bb2c4859939425793954c80556d3da3 : SUCCESS in 54m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7ffa6374b7644c584d8e5d05f8ac338 : SUCCESS in 1h 44m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2b5837ffce853fad5f31123ff68f3478779a062d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 20:09:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"63ce270836c82efa5eb3bf1b77b5083bf9bd9a81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:26:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3e138a8345245e8813ed9542c9816e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd3d39df3c4483ea6220cc05af58458 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee08f326f6d4fbd81293e5262e4519a : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2ce221d3b4b4e88a118b0a76255585c : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4c3e663e196403f9ec264a59a0155d0 : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2b8ba019e1c4bd0bbac8f53bef38462 : SUCCESS in 3m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bbfe92e0e0943a59edf9ed1deaa1b81 : SUCCESS in 32m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff67d682b5b84a35a0828b049155263c : SUCCESS in 1h 29m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a89691c54bd54024bdcfb77848175937bd24c67c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:26:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4c9e30d26dd6301258e5d218cbf44e02d1f0e14","revisions":{"b4c9e30d26dd6301258e5d218cbf44e02d1f0e14":{"kind":"REWORK","_number":1,"created":"2025-10-28 16:28:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/965151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/965151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/965151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/965151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/965151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/965151/1"}}},"commit":{"parents":[{"commit":"99ad43d08b36a6182e9dea3457d34b8eafb928c5","subject":"Fix broken extras check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99ad43d08b36a6182e9dea3457d34b8eafb928c5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 16:28:41.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 16:28:47.000000000","tz":0},"subject":"Fix keyname used to store extras","message":"Fix keyname used to store extras\n\nChange-Id: Ic9885cf58cca1ef560c2a25134fe5402fd32bc82\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4c9e30d26dd6301258e5d218cbf44e02d1f0e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4c9e30d26dd6301258e5d218cbf44e02d1f0e14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
