)]}'
{"id":"openstack%2Frequirements~965152","triplet_id":"openstack%2Frequirements~master~I915269401436820af463c84240111794bf116611","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2025-10-28 17:50:54.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-28 22:26:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I915269401436820af463c84240111794bf116611","subject":"Add tests for parsing requirements","status":"MERGED","created":"2025-10-28 16:28:57.000000000","updated":"2025-10-28 22:28:37.000000000","submitted":"2025-10-28 22:26:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965152","meta_rev_id":"fa106cfab37f8106965e5e880f0be23e7c3aed59","_number":965152,"virtual_id_number":965152,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 22:26:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2025-10-28 17:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2025-10-28 17:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-28 16:45:44.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2025-10-28 19:08:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79baa7df8ad7e4817cad8edf810af8b3810b48b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 16:28:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02e763c2dd901280221483913a8e834a872e75c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 16:35:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ab382ee69875d7fa3261af9094b4276a5837be4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 16:43:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dbd158508a7e0ae6f2d11b24ed0cdc3baafa6149","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-28 16:45:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"63e73ac6cd6376644b894f6f6e1891dde7aef7a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 17:06:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"56651fe1b47fb70f36e2ade64e66d9b6882acd93","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-28 17:50:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b2064aeedba0c1ec15728615e2fd2d633c5f55bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 19:08:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/190f166d899249678e6159b5db977f24\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c7dc84eac82452181c06b60caf1892a : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef1dcf5effe64f27ba6cba45660c546b : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3002bcf0fd7d4667b1c95beb78a76928 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37e3c509203a41c08fdc96636168657f : SUCCESS in 8m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02158dd1d76a40bf9457141cb4a6b368 : SUCCESS in 2m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cedde59b4d94aa8935de42f37190971 : SUCCESS in 34m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/496a5833f807448da1dfeb4200ea89bb : SUCCESS in 1h 51m 11s","accounts_in_message":[],"_revision_number":4},{"id":"f1e928a11c21471c010bc87e55602d7647f9154e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 20:09:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e9ce7385e582416e6d320859ef4e413975884a6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:26:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b141c29bb80649969839f3c581dfd75d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24670f17dc0c4f18bd259dd2f21bd750 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e5c8b867d514080b559d3b437d1fd5d : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f904548414774764854bef33aaf16e94 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ecffc61ab6e642fe8994ad83ae0019bd : SUCCESS in 7m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/800c132b704d414f9b6cd3e8bcb8c883 : SUCCESS in 3m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea822509380a437a86b31fc4ff89a95a : SUCCESS in 59m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/633b1107d4cc429ba3b19454abfa0275 : SUCCESS in 1h 57m 27s","accounts_in_message":[],"_revision_number":4},{"id":"939bb620b644ec064279337558f4c61931236873","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:26:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fa106cfab37f8106965e5e880f0be23e7c3aed59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:28:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8f3805138534f2f8e7ecf4b1b6b1954\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7750ad622b9b4504bffba38317cd00ed : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"41dd43d9a7ecb57274b816861a0eb8abc0f8e657","revisions":{"3a3ed49eaa3dd9bc353e46cdcf82558ad5f2a02e":{"kind":"REWORK","_number":1,"created":"2025-10-28 16:28:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/965152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/965152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/965152/1"}}},"commit":{"parents":[{"commit":"b4c9e30d26dd6301258e5d218cbf44e02d1f0e14","subject":"Fix keyname used to store extras","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4c9e30d26dd6301258e5d218cbf44e02d1f0e14"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 16:24:54.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 16:28:47.000000000","tz":0},"subject":"Add tests for parsing requirements","message":"Add tests for parsing requirements\n\nChange-Id: I915269401436820af463c84240111794bf116611\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a3ed49eaa3dd9bc353e46cdcf82558ad5f2a02e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a3ed49eaa3dd9bc353e46cdcf82558ad5f2a02e"}]},"branch":"refs/heads/master"},"fe3c1d8fb39562578ac5af74187a4b774fecaf7b":{"kind":"REWORK","_number":2,"created":"2025-10-28 16:35:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/965152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/965152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/965152/2"}}},"commit":{"parents":[{"commit":"810cb5c2e0d9650e7214d502c94c0d66204dc28a","subject":"Store pyproject.toml data as strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/810cb5c2e0d9650e7214d502c94c0d66204dc28a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 16:24:54.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 16:35:13.000000000","tz":0},"subject":"Add tests for parsing requirements","message":"Add tests for parsing requirements\n\nChange-Id: I915269401436820af463c84240111794bf116611\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe3c1d8fb39562578ac5af74187a4b774fecaf7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe3c1d8fb39562578ac5af74187a4b774fecaf7b"}]},"branch":"refs/heads/master"},"fb9dd62a2b397a6b1409902eef2ecb8acb00ac13":{"kind":"REWORK","_number":3,"created":"2025-10-28 16:43:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/965152/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/965152/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/965152/3"}}},"commit":{"parents":[{"commit":"810cb5c2e0d9650e7214d502c94c0d66204dc28a","subject":"Store pyproject.toml data as strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/810cb5c2e0d9650e7214d502c94c0d66204dc28a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 16:24:54.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 16:36:06.000000000","tz":0},"subject":"Add tests for parsing requirements","message":"Add tests for parsing requirements\n\nChange-Id: I915269401436820af463c84240111794bf116611\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb9dd62a2b397a6b1409902eef2ecb8acb00ac13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb9dd62a2b397a6b1409902eef2ecb8acb00ac13"}]},"branch":"refs/heads/master"},"41dd43d9a7ecb57274b816861a0eb8abc0f8e657":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-28 17:06:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/965152/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/965152/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965152/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/965152/4"}}},"commit":{"parents":[{"commit":"1505061d9cbd9efb184fe60968af80d46d33d1e6","subject":"Store pyproject.toml data as strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1505061d9cbd9efb184fe60968af80d46d33d1e6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 16:24:54.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 17:03:19.000000000","tz":0},"subject":"Add tests for parsing requirements","message":"Add tests for parsing requirements\n\nChange-Id: I915269401436820af463c84240111794bf116611\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41dd43d9a7ecb57274b816861a0eb8abc0f8e657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41dd43d9a7ecb57274b816861a0eb8abc0f8e657"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
