)]}'
{"id":"openstack%2Frequirements~965358","triplet_id":"openstack%2Frequirements~master~I5b6f8f561181696432fab11df5aae6e9e1794161","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5b6f8f561181696432fab11df5aae6e9e1794161","subject":"update constraint for pbr to new release 7.0.2","status":"MERGED","created":"2025-10-30 10:18:33.000000000","updated":"2025-10-31 03:35:57.000000000","submitted":"2025-10-31 03:35:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965358-new-release","meta_rev_id":"2429428edd001a5b09fbcc43dd839985a4ca1e60","_number":965358,"virtual_id_number":965358,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-31 03:35:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-30 12:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-10-31 01:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-10-31 01:38:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-30 12:21:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-10-30 12:23:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-31 01:38:14.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"68b92a1ad426945d4b0d7cc0317693fec051ff25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-10-30 10:18:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d507cf6ff3d01edbdaa3f7f9a4dd8ec8c3482d0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-30 12:21:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"111bbf4f5cd6aa92466592aeeacc0e3007aa6da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 12:23:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/216828967de84c749e1a5f2b93cb2190\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/987cfef1f6f14758bff207f3570d46a8 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9bf5f04004f4039b7e83dd87ff6b324 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c87b5d9cc8af4568bce7ae17bc4fdba9 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/672a6a5e246d40789573edd9c9d0742a : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d21188aa37ba4f3aa42099c1bd27cc8a : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7290556990e4626bda4b9de7f8425b3 : SUCCESS in 2m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d29471126d6a4f1fa09a3a97439b781e : SUCCESS in 5m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/64ecbb4d39e74880bce9e1bf9174456c : SUCCESS in 6m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9a2ddf266eea44829c478c2a44549f5c : SUCCESS in 3m 35s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/30b82144dcaa4b968bb584c0f26b2706 : SUCCESS in 4m 55s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6cdf259a507e419e8567e68a8d3ed60e : SUCCESS in 4m 08s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/60185dad4935494a8cd9dd43869e3fb0 : SUCCESS in 5m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/645930efe0284aa6badef65b4d9cfe74 : SUCCESS in 2m 35s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/59850e4301364493af05f09a544cb27a : SUCCESS in 9m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e1eafa433bdd4194ba8ec81d99751d1c : SUCCESS in 5m 00s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/53ce003982b94435b04ff1ed39fd164e : SUCCESS in 5m 41s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5995b6ef50cf47fd81f82b89b2bfb511 : SUCCESS in 4m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5c971b37e414424f8ed932f1f0483c21 : SUCCESS in 14m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0abc30b2877414d9d3ae8e1cbc7d7a2 : SUCCESS in 10m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ec83d5960c2a4ff9aa2e135b05dc9453 : SUCCESS in 10m 22s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3fb6373fd3d648e3b829ea3c266e8e59 : SUCCESS in 11m 52s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f823e061a1fb462e9be771244274f72b : SUCCESS in 2m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/83c88e5b254d449e9b2d3075fbb6551a : SUCCESS in 6m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0fd9621dc933415c9911b65026d4cb7b : SUCCESS in 3m 06s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ea5a31a412814e3ea7ff0bf91eec0bf8 : SUCCESS in 5m 06s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a5f8518f02a14572abf26ff98085e2d9 : SUCCESS in 10m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/57f10955eea14ffd92f69d24e0298b3a : SUCCESS in 26m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/872ac8823bff458fae667b023eedb6ca : SUCCESS in 4m 14s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/645c4efb6cb042d7a377b83778a3e1b7 : SUCCESS in 13m 42s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a707e34e504d4221b726a9b8534e9d1f : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/39a04f5746fe4481b5be6fa21a62e180 : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e74bff178c44b81a327642b6b615c98 : SUCCESS in 24m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/81475e66cff245438b97f5f7dbf924bb : SUCCESS in 8m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/67c2757d7f7a4f7485d907eac9dcccd5 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/107cf6bc784a4e7aaa1d06605a8e4907 : SUCCESS in 6m 05s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3fa588a6c7484297882c740c798827c0 : SUCCESS in 18m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9240678752054d14a97d5c2df512f70b : SUCCESS in 5m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/665b787a8ef544e693b904eaa7c654b9 : SUCCESS in 1h 01m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56f6dae081cc46a7a5f56330c74c1292 : SUCCESS in 1h 59m 52s","accounts_in_message":[],"_revision_number":1},{"id":"37b7a94b01ee194413bb8fc01894102c887b1152","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-10-31 01:38:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5dc151a6a0514b7f5532bbe8d04b4c31538d4a41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 01:39:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9567205a19181e5a00783e3cd18f19cf19592a8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 03:35:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ae51115a2424a21967ef1999585bc78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a5495b19704ee1a2d2d9ec4e5634ea : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0368b20c2baf4d318238f6b25b751d8c : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e71ece46997a49b8ad0bb1529fd6265b : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df463b7c4f2e48a3a308c0df94ef1184 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97c0852524984328aa6ebd476507161c : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0736ccc4f6404a3bbcaaa6e97e3ae4a1 : SUCCESS in 3m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5c3a53f86c934ac285e601b4af3b7060 : SUCCESS in 7m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b82713e3138048728d77974c6376035c : SUCCESS in 6m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d93b66433ede47c4bca895c4a063d2f8 : SUCCESS in 8m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/36c06de9245e4d5c83cd48fa17fab82b : SUCCESS in 7m 23s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2dab8db770db4bc4bcf6c80fa3e861d9 : SUCCESS in 5m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/19b8784bcf0d4b0d88f8194588701ed1 : SUCCESS in 5m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6914a4c51ceb4dbd8c1b6d3e9dcbf16d : SUCCESS in 7m 46s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/89a3bb9a216846abbc52472fd56bf1ca : SUCCESS in 8m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee97a0b0cbda42b1b8d672e2fe930f3a : SUCCESS in 6m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d3b4becb79df4ebca5f3fdacef036251 : SUCCESS in 9m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1e8bb228310541a3845e3d59163eebb1 : SUCCESS in 11m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c0e8263a86a740de9511b88d761e6dfe : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/618d00b3d1f94ad8a2c50be0448e6c8c : SUCCESS in 10m 59s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/322d1d9e620542eb8111de85213d4088 : SUCCESS in 32m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d85e8bfd367f4504ba2c3f16b7f318b7 : SUCCESS in 7m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/da1bf8f3990f4e6db144707dd83785d5 : SUCCESS in 13m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/24721f09b72d472ba92525cf8f1ec0b4 : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3cad0d90ec9a496cb16ff1499257c469 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f8a3a6c141141b083a259deaac0dbfd : SUCCESS in 29m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/296f9ec4a2f6449e92652aee0d1c601d : SUCCESS in 6m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0ec08f9f0e3b4001b8d62ff9dafa5a48 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a7988cfaeeb497da7d21720af64f962 : SUCCESS in 6m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e6b1cd01242a4d4db9e60b7393e8b315 : SUCCESS in 12m 02s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/05bea51faaaa41abbdf84877c1eee0db : SUCCESS in 5m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c197ff23376432a96741e4c719a6dcf : SUCCESS in 58m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/172497304f7e4696bc61c35b0819b3a8 : SUCCESS in 1h 51m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4bee066cc4ef8b47b5ff4590e9699d67190b57cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 03:35:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2429428edd001a5b09fbcc43dd839985a4ca1e60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 03:35:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87dade87545b42418b6474aad2a89504\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a706abe07a6a4776831eabe2c2ff9141 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b542873b8fe527b6280790bd59e5dc146350f14","revisions":{"0b542873b8fe527b6280790bd59e5dc146350f14":{"kind":"REWORK","_number":1,"created":"2025-10-30 10:18:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/58/965358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/58/965358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/965358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/965358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/965358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/58/965358/1"}}},"commit":{"parents":[{"commit":"6098eb19d46b7f44e19054c5ef9c0e8185d1315e","subject":"Merge \"Add tests for parsing requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6098eb19d46b7f44e19054c5ef9c0e8185d1315e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-30 10:08:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-30 10:08:31.000000000","tz":0},"subject":"update constraint for pbr to new release 7.0.2","message":"update constraint for pbr to new release 7.0.2\n\nmeta: version: 7.0.2\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Commit: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Change-Id: I29cac66277deb12ff8bb67ce7c91cbf09e100a7f\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I5b6f8f561181696432fab11df5aae6e9e1794161\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b542873b8fe527b6280790bd59e5dc146350f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b542873b8fe527b6280790bd59e5dc146350f14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
