)]}'
{"id":"openstack%2Frequirements~965873","triplet_id":"openstack%2Frequirements~master~I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2026-02-23 11:47:11.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-02-23 10:18:42.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-02-23 11:47:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568","subject":"Updated from generate-constraints","status":"MERGED","created":"2025-11-01 08:20:17.000000000","updated":"2026-02-23 11:48:00.000000000","submitted":"2026-02-23 11:47:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"965873-openstack/requirements/constraints","meta_rev_id":"323fbc669eb23b9b24da69e0036534bb60672475","_number":965873,"virtual_id_number":965873,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-23 11:47:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-02-23 10:18:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-02-23 09:58:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2026-02-23 10:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-02-23 10:18:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-01 10:37:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-27 09:28:23.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2026-02-22 11:22:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-02-23 09:58:01.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-02-23 10:17:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"49453d27da3fbbccbd78c9933f367013c5cdfb9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-01 08:20:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de9b626511dc24790e597453ca35e7eb43434660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 10:37:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4ede6b9c23945a79927b64d865409ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b9fed0905bd4ef1905e0b1172907d0c : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f53105389d248d5a5adf1c40c403b94 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36e046b9b61a435fa940f9f8465e6c50 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f9131e5bed1420986e38bd28b64bd2c : SUCCESS in 7m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84f348d176114921ace12385334c7f28 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b8a379a26604c5f8943eeee8c08fb04 : SUCCESS in 2m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/54b8a3abbfed4b1f9191a0a69cff021a : SUCCESS in 5m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bdc3f288b4514e65b4a2744c7c093690 : SUCCESS in 11m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1f2a9c849e434cf6853a5eb42de3d8e5 : SUCCESS in 7m 00s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ab92e115e63f45d7b919fbbefb166b87 : SUCCESS in 6m 56s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/c83e863317884aab9b876a4a3ef1330f : SUCCESS in 5m 07s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/269481b5c7ac45b59502a48ad7df2e35 : SUCCESS in 4m 32s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/014486b250db42b397a4ee192875190e : SUCCESS in 4m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0df69b5d32794cc9932181d940338be1 : SUCCESS in 11m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7d9fde07823147ab835339534dff0283 : FAILURE in 5m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4f1d70e450b445b0a1a4e65b90c21e23 : SUCCESS in 5m 45s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/51acb4dccca1486b9d42e08c9098d2ee : SUCCESS in 4m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d0f745d46ff74eb989e3d7ef4efced23 : SUCCESS in 17m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f477310bd34841138d818b6c07641a38 : SUCCESS in 6m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/47a5ee87cca94e3ba09ead41d2a8c122 : FAILURE in 3m 22s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4a393779a1be4a5eb1d69ee80163fabc : SUCCESS in 12m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fa52af1cf11149e6b5fbb6a7f5616a5c : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ccc37737bef84d309f0dabadfd167345 : SUCCESS in 12m 57s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/452c3c035107454eb270b5f5ed94b001 : SUCCESS in 6m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ae2264661c3a408a9319f159c49d3ccb : SUCCESS in 5m 20s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7f3f20311b2d45d192fc2accc711bf86 : SUCCESS in 10m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b922993898e1462a9eb522b0f87ec2d9 : SUCCESS in 23m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c3471ed0b5ab404a9e4af7de344f1cc9 : SUCCESS in 7m 37s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b81e4bfc2e3043169750dd8233f81d42 : SUCCESS in 8m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/19a11770a6a84790953617fc50dbe8a9 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b9ebdcbf838841c780883e9f135d74dc : SUCCESS in 2m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5cd0eb807b4d4c4784f0c9a588ea852c : FAILURE in 28m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6bf459bb73fc4b25a2577b886d1406ad : FAILURE in 6m 35s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/45496199a0bf4e888ec1de563e06fd95 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a616aeddabe34c50b27a8e882d7cdf52 : SUCCESS in 5m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/668b389567ee4c22ae5cce63d807677f : SUCCESS in 18m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0b83dd592bc04d81a560348d486362b7 : SUCCESS in 5m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cde5953a37a401ebd3941e70b35bf95 : SUCCESS in 39m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dab780f5829745d5816e097cc7c399e4 : SUCCESS in 2h 10m 46s","accounts_in_message":[],"_revision_number":1},{"id":"585f8b3bcad73c10863a0b207dd41adda39e23af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-08 08:32:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d2fbc9ce400898cae820f9536de3bb406bf89a3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-08 10:15:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/462a1ae92e8c446890e3ae376a9b07c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e87712468f4beb8535312bffb95f3d : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad15462ca094057a57c5c615a2f834e : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1101af144d4a44a0939c8449afdb4ce8 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/973232729be34626a5b6db4e0583f293 : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/335f4ba8344d40f6b297a1c291462652 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a37232bb817b4790936c9e75b7403cb2 : SUCCESS in 2m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8e2212c36ed040eeb230a87e63c42cf2 : SUCCESS in 4m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/23d137ebd19b42b9b5c098dcd75836d0 : SUCCESS in 6m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1f8af4ef5a5b46c893ee16ab54b21057 : SUCCESS in 7m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/10152223110d403aa09f4f2ad0ae7f40 : SUCCESS in 6m 04s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/67d031caaf534ee29339bb985503a63f : SUCCESS in 4m 10s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3b5578586a7f487da8ac44a1a17cf4e7 : SUCCESS in 5m 30s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/030335cc6b2a42f3b1a48c42f3d04936 : SUCCESS in 5m 34s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/12be47fe599d4b45b75c97d11bc0eb06 : SUCCESS in 8m 46s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4206538f81db45d68c33b6f2ddbba026 : FAILURE in 5m 13s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/23e9ec736fb449cfaac43f2f51f55cd5 : SUCCESS in 5m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b3da6c06dcd1425b8a137e7c5b82b2d4 : SUCCESS in 6m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ad5004ddb7a8403e98087ba5ef281e29 : SUCCESS in 16m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd1762af64564b968a461c6439662029 : SUCCESS in 7m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3c9dd2b3bab141debd07f94dcaac2551 : FAILURE in 4m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/71b4b3241aa64cf8ab5e8e7623acd9ad : SUCCESS in 11m 18s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/464c8afb6f2045769b77fe3dac59817f : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d7f86a26ecfa43f6b4211681a219f451 : SUCCESS in 7m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bb5486fe44e145b49f66fae774f0b9a2 : SUCCESS in 6m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c932cd2f6b7543098847e48b20726ecc : SUCCESS in 4m 27s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d91e7221edd24b0684344b2d7972a111 : SUCCESS in 11m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d18e95a306da43bfb40abb96701defdf : SUCCESS in 21m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c1606022e366490db35026cf05302122 : SUCCESS in 8m 01s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b71e6358e2f840cbb65a48ef9559f6eb : SUCCESS in 13m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/71cfeb6d86b6436bbcff136c5084bcda : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/573c58fcd62b4306ba3381040fc55b46 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ef575ced3c44df08ae33a20e5422a70 : FAILURE in 24m 20s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4644b4405ea741ec825a3f62adacdfbd : FAILURE in 3m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/667c753174fb4de7a60e904e4c7b9810 : FAILURE in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49a7d31fbdd14419a91a0e519900a762 : SUCCESS in 6m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c91614ded13443bebcccbf04bd8e7f63 : SUCCESS in 11m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/64393c35ece84a0cbb0b474e7067b638 : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2955985dbb874985966263b56c82734a : SUCCESS in 35m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ff04dc15b8a4146b4b0c19cbb10c0ca : SUCCESS in 1h 42m 11s","accounts_in_message":[],"_revision_number":2},{"id":"6074509d2b26d7bd374175d3aa5a75eee6a04949","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-15 08:27:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"17a6fc8d72957d3b9ffcb72d4c6710ee5c287a2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-15 09:43:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/157466f1d3bf4b5591fbf5c8f890a290\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3296f45b45064b5c97275306182a8aac : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/049afef9ee404607b5f8956c04183c56 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec956cf072b94e40adc3b6a384bc680c : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d17b7f27e0d84219ad14c782ed624585 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce084e56dfd0487d89784584c2d2f35f : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7392c9cfb08542b1bcf155c1dbca6175 : SUCCESS in 3m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1c5a89d2838047bfb52410a7963819e2 : SUCCESS in 5m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/27cb28ca505d4810a82168fad2a79bf8 : SUCCESS in 7m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a09fc067ba1b4d98a8771959331b7a8a : SUCCESS in 6m 51s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/4632eaf43f1940ba9439601742bd5572 : SUCCESS in 5m 16s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7d6eaa13957a46fdb13f3b064dc34454 : SUCCESS in 5m 07s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/70ebb8089fe246b7b22a4b6ea7e5adb2 : SUCCESS in 5m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/62e99f6a18564b14a39aa3c0c375da42 : SUCCESS in 4m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e5e15470b85d4a389c6d0b3e4adebc07 : SUCCESS in 13m 30s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/38fcbb320d364e22b88f77e71dd922bd : FAILURE in 5m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3f1d7460e6fe43eea81a3bb096543ce7 : SUCCESS in 5m 59s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f55203f259eb44f79e628d5269abfecc : TIMED_OUT in 40m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d7e6de0403064c72a0184e49a2b887c6 : SUCCESS in 9m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea0284f373c3487490bd68880f721aee : SUCCESS in 4m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/90b3c6cc99c9493885ec649de65b03fa : FAILURE in 4m 11s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dfa5cd31afe5487fa135fac2df3d85d3 : SUCCESS in 11m 54s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4d3a51d7d20943e0954b82d2d310f2a8 : SUCCESS in 3m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2efe24c6e9db42909fef6758e79ed36a : SUCCESS in 10m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/83f4e59bb82f4dee9e0f25fb41d4556b : SUCCESS in 4m 48s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4a4d1c7df8ae4177b2cdffdfe6dc32bc : SUCCESS in 5m 06s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/872a0de4fa47479580c27d3152805342 : SUCCESS in 11m 13s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8a48a933702f4b5188e1c8dc255390ab : SUCCESS in 27m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f4fc06bcd84049d79aef199b14e9c973 : SUCCESS in 11m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d3d3e160a97743ad9e7f63b6fd8e77ff : SUCCESS in 8m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4f256d6bfc154513b032749b9f6b742d : SUCCESS in 3m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f50ac5b394464715a8f2983e71d4f170 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b7ec303e776436281d94fa955eaefe8 : SUCCESS in 26m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/360bd313773b41f9b6592157462ffd8e : FAILURE in 5m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/19984c976b7e469187c6ce61f7cd55bc : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0ec36491783047518821d23c8e7618fa : SUCCESS in 7m 03s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6f2b6cce4fc044e0acfd4b8f5dbb6011 : SUCCESS in 19m 37s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4d405afcae064fabb9b9abb3e5e8689f : SUCCESS in 5m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/579dee71cdb5482fa54a26b02e0c128e : FAILURE in 1h 13m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b82411ea9b39413bb9a09f0b6169dbdb : FAILURE in 44m 09s","accounts_in_message":[],"_revision_number":3},{"id":"2ac228e8a0337086ac02087f97e8abe2601c073e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-22 08:19:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9062bf2bb11ac5da4ef73ab317fceac6f273657d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 09:38:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b028552b878b41e081cedd3451e18695\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51fdd5cdd11342aea29e6b15df907838 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed9028a47c4e4cc0989656c9e9633f82 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/792b67cb977c49c2bb779162dec6b0c1 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/737437201f184d2b83816caaee231c5e : SUCCESS in 5m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e7ac4832fa44c089c2d24641269def5 : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e03c1f4311054420a17e3f4325751b64 : SUCCESS in 4m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1c85b5c6c30b4f9b8e8ca938867ffeb1 : SUCCESS in 6m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6e3ed762c83a466c8d889f52cf4aa62b : SUCCESS in 6m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4d38f0f0361a48c6b1b29ec5a5bf9e9c : SUCCESS in 12m 04s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3fe94b7a03f144d9945d033140899c23 : SUCCESS in 10m 06s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/65f8b8acc1bf493c976fac41705ecfd9 : SUCCESS in 4m 00s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5d1debd8a3bf452bba7bc416e046c93f : SUCCESS in 5m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b55aefe55dd74863a89af947eb2c0d1a : SUCCESS in 5m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8609324b62cd4493b1a680823e7cb7e2 : SUCCESS in 10m 10s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f93bc154bec944688316b3d890f20588 : FAILURE in 5m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1652212f682f4251b0c932c00614fde6 : SUCCESS in 7m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6a1767527b964dd9bd99ea670e166e7e : SUCCESS in 5m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f081087a25884a88bb41788f960589b6 : SUCCESS in 16m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/129e9d7ad3384411919db3402c13eee4 : SUCCESS in 5m 26s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/28d69c317ce144eeb2a840f80ed56caf : FAILURE in 4m 14s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a547a53f2a36437e9ccdbbf4d8045e1b : SUCCESS in 10m 06s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/46902190ac3a47a484fec1d7c88f167c : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/29d396b346d04b10ac3c43ecb4ec4127 : SUCCESS in 7m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6a39bc83ee2247e98e11f5382d74b7fa : SUCCESS in 4m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2f2bd308c5ef42c7aa1876dc0485855b : SUCCESS in 3m 56s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/08fce5688b814c63913a38b4b26575db : SUCCESS in 9m 59s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/359d97f2dc0a40c2ae11933cd0117292 : SUCCESS in 22m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/617d4eec0f9a4b42b46ff48a3aa742bd : SUCCESS in 5m 31s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/be935602074240a7934c2c8f76b728cc : SUCCESS in 10m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6a78bf8271d84ffe822442fb59d5132e : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c748b21a93f64b5aa513cda1d3ad5350 : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9e30a0610c547a497dd10b155ab1fa6 : FAILURE in 20m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c1712c272645457bb3f3d660fb84772c : FAILURE in 5m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7575eec26dc94c91aa2a22758bd06ec7 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9027aefe7bfa4b7996ff92529a25c524 : SUCCESS in 7m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bcca8ee15fbd4ebb802cb58f1f33ce1e : SUCCESS in 14m 13s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/74c417aa2db24cc69aed1eab2b1cc304 : SUCCESS in 6m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d82394922b44f46a7f38383cb70b9d1 : SUCCESS in 1h 00m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc050552554b44fc867bf0c163ed9c91 : FAILURE in 1h 18m 21s","accounts_in_message":[],"_revision_number":4},{"id":"b9e7ea5662aba065157e1436349934f8b46215ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-11-27 09:28:23.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d05bc890da0afa895dcf647eb4a7cffc79139c5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 10:49:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38c9a130d7ab40e6bc10434d9c97507c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dcfbc116ae8423f9ab51663a83e292b : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ce97273f53042a7a684cf53a4731e58 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cae2453e2aa34bc5843ee276e9408585 : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be47e81ae3e54ffcb958c1769c601b49 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/63a02c287c3e4fd2a74ff1f98d67a61a : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/15dd4f52a3c64ad99e7fc59ce8bf2489 : SUCCESS in 3m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ebdf6f0a800942cb8f5ab30b02dbf378 : SUCCESS in 3m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4ce644e64d574092af281dfe7a948789 : SUCCESS in 5m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b7e334d792854ed8a8fc38f2dcd2b515 : SUCCESS in 7m 02s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/308dfecda6bb4011ac3621615bd76bb7 : SUCCESS in 9m 41s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/54317b3935774cb88dda479eea2998d8 : SUCCESS in 5m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e288adc1bc2d4a58b9d8414252ff1b14 : SUCCESS in 5m 28s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/cb90689be3984871903802cb17f0e172 : SUCCESS in 4m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6c0803a898f04000b009ea623e9d8c01 : SUCCESS in 10m 29s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b99927181c21438099615cc67ac93cb4 : SUCCESS in 5m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d470cdc31ce4457c9fcd62cf69a1c328 : SUCCESS in 4m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0e27ee151ecd4f9fb1f0905c86472f5c : SUCCESS in 4m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f40964d3a62246fab39f640a19214e6a : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f68cd9e063704305838a318d0e5e65c9 : SUCCESS in 6m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/17003a61f541465288de6fafae2a86c3 : SUCCESS in 9m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b93526709dd54a5795333cf27e6dfe05 : SUCCESS in 12m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c1ed1ef3ceab40d39324347d3d176470 : SUCCESS in 1m 52s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bd1db47d772c49e5a3d5de1506485153 : SUCCESS in 9m 42s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/35f4c2a05235431780c00719e01b2f7d : SUCCESS in 4m 57s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/bf495cb3f6ff4e328721b862893c6a06 : SUCCESS in 4m 17s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3acea9b800884762b1dfd851781a2066 : SUCCESS in 11m 08s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5bf0bd6d79c54b41b7011483a57735b4 : SUCCESS in 19m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e217aed2b5e2450aa8f16bd5465a9041 : SUCCESS in 8m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b35da30e8c0e446aa649bb291ea23c56 : SUCCESS in 13m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2a6af657afea4fdfbe98b37ac4aa0c48 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c7a6c4f5ced4555b3a0c3eb318645f5 : SUCCESS in 2m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07c7534ee06f48a6b2b52a1b75b6b06e : SUCCESS in 24m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/52edc390fca848c89850a5cf4e59e958 : SUCCESS in 6m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/36f56622a77c4b70901a760906fb0068 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45245edacb2f485e8937d9a87bf406cf : SUCCESS in 6m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fc0886e9140746a4969d8d4f9b28f382 : SUCCESS in 7m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/936ab222d81249698a9f9388eb700b2f : SUCCESS in 4m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37f4c07070734a1db7de03b4d22897d3 : SUCCESS in 1h 00m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/767457aedbe14d67956f6a9f9be5a409 : FAILURE in 1h 19m 33s","accounts_in_message":[],"_revision_number":5},{"id":"e6c7039799d6ba96a4572b0bb4cc38a8fec50413","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-29 08:29:24.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cc2cf1973ba347fc188c261f59d6aa2f8d3f4a8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-29 09:56:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e99d34f6f184bf980fc4e0386e23bde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e763326d7ce047f1a8ea8896672380ba : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8969361205d4e1daaef4ed72a76039b : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/876d7ff19d7b45bf817cf930d11780f7 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5778dc95449948d5bfb4c57f8cebc931 : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3d51b18d6994bf2a62f62cd74608065 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/483846d5cc484fb4becf071dc030267f : SUCCESS in 3m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1e9352b857314c15a1287f6ccc7466c2 : SUCCESS in 5m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5f7359d49b1f492190bbb45b35b0fdf0 : SUCCESS in 6m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fe13d179f3c946e4ae3cd5712e363a30 : SUCCESS in 6m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ed2d00e4049243da906e5bbc55395e78 : SUCCESS in 5m 22s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/029c682b18f848889c58d9f41a5cf7e0 : SUCCESS in 6m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c871414927a549b4950ed29265a6d051 : SUCCESS in 5m 11s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7f3e1624d747427caae9704577b2f632 : SUCCESS in 4m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/700f0c37d27b4efea1d8c556ab7e4631 : SUCCESS in 11m 46s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/aba45ac632134ad6921277050e6fb25b : FAILURE in 6m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/69b7b60779f741d992e0458c1d63556f : SUCCESS in 6m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0e5fa61775204619bf65470c6908b01a : SUCCESS in 5m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ac2ef4595f4948bc924f4e22d4d7a4dc : SUCCESS in 18m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e948f12029c8467da444a297a985a8c1 : SUCCESS in 6m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d015bb59155841adb48e93f316b5ba6d : FAILURE in 4m 21s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cecadf476c004dca8a2c4c49f5311d04 : SUCCESS in 7m 18s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f0e15035f072439785cde39d0787a3e1 : SUCCESS in 2m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/819e55759de643e5bfcf30d4f458095c : SUCCESS in 10m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4f75e89ba7da4f61a00128afd0725a0f : SUCCESS in 6m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/faf92c048b3f47d7aaeac3632c4b5aaa : SUCCESS in 3m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d4ee06d777374cd193e2027b160b0f9d : SUCCESS in 10m 11s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/85c86a3f34934fe4bd8cdf4695799971 : SUCCESS in 26m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/215b0ad45ee246ee88f24522ea9667e1 : SUCCESS in 9m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/44d325ca9ea242c7a9a5a12a7d740f6e : SUCCESS in 11m 12s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5cabce4eb92f43b9aad4a755fe0526ef : SUCCESS in 4m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e79d90d196c443f090504f7214ba52e9 : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/244bf26a18c34d2c91ab802a922373ca : SUCCESS in 26m 01s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/72dd8e5a35524c77a70ac5faef3a9c6b : FAILURE in 3m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2a3c48c089344731a44cb37264c08846 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be01909c3c57483db98a450b995a82e9 : SUCCESS in 3m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ce90cf3a83414a9ab191b18accc3b90c : SUCCESS in 7m 06s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/28df084e0c31420db85078215db2f69e : SUCCESS in 5m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2c5d9a13c69450897a16ed0988f8289 : SUCCESS in 35m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee81f9304d724a2dacceb7d808a05787 : FAILURE in 1h 22m 20s","accounts_in_message":[],"_revision_number":6},{"id":"afb81a9b36294d7655a4c946bc2439353920752c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-12-06 08:37:53.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"2ca42e258a376894d9688ba575d7caada2396bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-06 10:17:18.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b99b909f2a4346aeb29ddbc6b8bdb7f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d83196bb41274fd7a8b17592a81e0972 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6b74a0204448c9a198290cdf1b4de0 : SUCCESS in 2m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe7a64011ced426f9a5d19fce79eb5da : SUCCESS in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76cdfa40827647abad6c1927d45d8a3c : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44c52b0f241146adae360b86be5fe67f : SUCCESS in 2m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/304edf7f11fc4a6996773ceba8587603 : SUCCESS in 1m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8112eb2dab79429ea67abd793cb66490 : SUCCESS in 5m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a6e42376830a45a1862ff797f89961b5 : SUCCESS in 3m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/de19236c60e040b081d0333a769329bd : SUCCESS in 8m 51s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/720b4374c97f42db91ddf06b9691e9d7 : SUCCESS in 9m 42s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/56aaf703fd5a4788aa1ebe28b4f036ff : SUCCESS in 2m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1c368c0e74024677a00ab0ff2849568a : SUCCESS in 4m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5337d245f7ff4d6e9cb3e670d8063e3e : SUCCESS in 4m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/99735f33f7df455e863ba04684f67669 : SUCCESS in 12m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2139829f1f6f4ee98803d3d9b57b8283 : FAILURE in 4m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/10ce3c10eeff4c0790bcf9d65339441d : SUCCESS in 2m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2d121352fcbd4fc6b76932047464f8e8 : SUCCESS in 6m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3e74514b53d344a79f93178b38191fc7 : SUCCESS in 17m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9db0b8ef4e084590b752ac48dc9e753d : SUCCESS in 6m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8424a3aa570a489da50958da0e334ba5 : FAILURE in 2m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c4c0d77e7a1a48829ac48b945b649cd9 : SUCCESS in 7m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/85f1eacdf6cd444583bd9bed4aa5e3de : SUCCESS in 1m 44s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ed1e0dab954147a0879d05515e652f8d : SUCCESS in 9m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8ba37420bde949e4811761abed2d0749 : SUCCESS in 4m 15s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/7289b2d922844463bb07da9856ba9e69 : SUCCESS in 6m 24s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/bb622749f8da4a93814840bdf2323c65 : SUCCESS in 10m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e53dd7c1304c448eac8dc07557dd9821 : SUCCESS in 22m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6e5a92ea83a549bba8d58566fb205dd7 : SUCCESS in 7m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1e9b6dcd5efa467cbc796a9d3dfebbb5 : SUCCESS in 13m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5c397fe5dfc549bea7185523f1aca90f : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6118feeede05457698d2983a86a7974f : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c81f20088c634562a2bf16905c9e2d31 : SUCCESS in 27m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/609f574137184eb2a3ba3ab7d827f3f7 : FAILURE in 5m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4e665c1ee48d4f84bd9beeb6f22e9e44 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/911cdbc911164eba837abdf93debd2ec : SUCCESS in 5m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6fcf833e9996494290d984203749f7ee : SUCCESS in 19m 26s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/df79b47079294a6dbc6f87d0e40dd362 : SUCCESS in 3m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1da9df5e76ee476bbb0c99468aebfcd0 : SUCCESS in 32m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f47eb22f7bb44c09ae10146e3f14700 : FAILURE in 1h 29m 57s","accounts_in_message":[],"_revision_number":7},{"id":"2e39aeff4c2c6d7a3e37c86f73b3c6ba7c2ae60f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-12-13 08:18:22.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"cb9af38775c1ae68be02bbf03dfd7a40e70938aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 09:22:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cde6affccfd1426389290826b7fb4f71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeddca32ff5d4b3c974003e1409c4a6a : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a0be4d5aeaa41bc81f443473b0d12dc : SUCCESS in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d081b8a84ec24fc2892f0c714df4a232 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/645e3b19ca6f4c86be6fc82c4eb8e958 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81513873671b43d6856897b7ce270431 : SUCCESS in 2m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/040be6ae51e64926852372c34cfd7240 : SUCCESS in 2m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a241347faab943a594667089331d91f8 : SUCCESS in 5m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e6f177d9734d492ba4067ac4e2364556 : FAILURE in 3m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5ac3cd85ca064580aab6f77f8bcb479e : FAILURE in 4m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/751da7e2dafd4a8ba4896a7373538771 : FAILURE in 5m 49s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b7ac61d19c9145c9aebae6802b26ac49 : SUCCESS in 4m 16s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a31c1654168e429994ba702d62d34ed7 : SUCCESS in 3m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1521d9e6bdac46e894e1b5ed2433f203 : SUCCESS in 4m 04s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8a6b423d99554604857a4b4c0ea1bbb2 : SUCCESS in 8m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f343a7801e594e57aa8ba31f2eb7b377 : FAILURE in 4m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ffccaf5ce26c4430aced4c6330a4b68f : SUCCESS in 5m 24s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1b623347f19d47488a4bfa8c1d85dfc9 : SUCCESS in 7m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bfc48d79bcb040dd91ae861f93bca094 : SUCCESS in 17m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d5fd6ef8ca184bc691a2986d31f1207d : SUCCESS in 6m 28s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8e7e5b41f70e42ff9317ed145be97395 : FAILURE in 4m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ae58696a544449d9ac20583c662371d1 : SUCCESS in 11m 48s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/50920e4585c84f778460034c4d958fb8 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/faa5e401a24b45e3912017e271ecb044 : SUCCESS in 9m 18s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/87fe2dd77a1740ecba3142658917758b : SUCCESS in 4m 18s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9239d912b79441008fb51fabf9c4e096 : SUCCESS in 3m 09s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/bc93b7d4dbc84a9c81b3f89565384bab : SUCCESS in 9m 36s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fb87405b506744dca7aa3992373a48d9 : SUCCESS in 18m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/da077b1f1a77451d9b969163868b7893 : SUCCESS in 6m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5a6e488ad3574985aecc14dfa90d5378 : SUCCESS in 12m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7c2db1a7a7e14d6aa7f514d5ba29a596 : SUCCESS in 2m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a95ab509ab64a60994f0cc1d3149da4 : SUCCESS in 2m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bea7aa24ed054bdfa70538e74ae3e11c : SUCCESS in 26m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5df9724f96db4dd2b6de287d0f468730 : FAILURE in 2m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/97ca84ddcbbc49488deba9e9217f9e74 : SUCCESS in 3m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24c91350db494e27b699ba25fbf9f615 : SUCCESS in 7m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8f74941c372f42caa710e1f944532647 : SUCCESS in 13m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1a725c10ba2f43d88883b6bb43591ce2 : SUCCESS in 5m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f863c513255d4d1fa113ebbd4ad828cf : SUCCESS in 35m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec878a056b8742e1a51c4705be04c2d5 : FAILURE in 1h 02m 28s","accounts_in_message":[],"_revision_number":8},{"id":"636f01fc69202414272dc3723162de349c4c2c36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-12-20 08:21:23.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"a4bc79a47f484f4bf1885b06aaf6c9eb7fb3ffa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-20 09:23:57.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd6204788dc347aabb8eb259a98837a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/392afa22bb61411d9b42359d42760777 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c511e88e6fdd42fa85aab31b71cfe642 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddd28b0d606c440ebb05f3624a9a9f49 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9659d6e685684e3ea427dd447d91df09 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a7ee3715a9943199009ec6c58840297 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c1412dbdba54e7d93ffca2bb707d7f7 : SUCCESS in 2m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a7f72d9035c945108190fe406e97ae7b : SUCCESS in 5m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3d5ef3b184d64f319842e32feaf7aef4 : FAILURE in 5m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6656c4c619174428bd62f2747cffbc2b : FAILURE in 3m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/58e94c6208af4043ab2887669bb76b34 : FAILURE in 5m 03s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7ea6764b4bdd4007b779bb4ffc58bacf : SUCCESS in 4m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4b84968cab65488a9df3c18d4dad11ed : SUCCESS in 6m 21s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7fc48aedae2442eb98b30061bae05719 : SUCCESS in 4m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/99b5e7d584b748638d8f6f6e32bc17c6 : SUCCESS in 10m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/314a546729e94dfca360e441c4a8de4a : FAILURE in 5m 14s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/18036678e5c44718b93f7e767322cad0 : SUCCESS in 7m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/31718d23a7e74ed8b47bd82f0dbeb70c : SUCCESS in 3m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ba9442e0e19548ca97a057a1f0eca29d : SUCCESS in 13m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d624607376b488198a9d397cc8a3017 : SUCCESS in 4m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/01343671c66c49a189a85ac546411205 : FAILURE in 2m 23s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/0e4a511100734e10bbf350810922df7e : SUCCESS in 9m 49s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/44a84fbe027a43799a3c7b827f7d58fe : SUCCESS in 3m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/2243f861f2094a1ab5f25c5d8557fc32 : SUCCESS in 7m 46s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e0f3c07654704ac8b2aa86a2ce5f6fe2 : SUCCESS in 3m 41s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8ae1aec7136a4af3817d5063878a5e0f : SUCCESS in 5m 03s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6ef50c39028e41dbafc89c01fb732a2c : SUCCESS in 10m 40s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/46dabbed8aeb4dc095ebd425e0cefc13 : SUCCESS in 26m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2e63c39896144777a2157d1012bca9ee : SUCCESS in 7m 25s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/903f8d22a68140c996cabb78246bccd6 : FAILURE in 6m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fe016c6ef552488e96353514b2574c42 : SUCCESS in 2m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/72a6260409d049acbab14b6a4831a256 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4530d3930194d73b60c1190f3a8830a : SUCCESS in 26m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fdc47b41ec264134acc148ddbe7aa39c : FAILURE in 4m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/db946dbd7b1f4fc39d6357562d7a9cec : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b080f24372b4ffca91693ce220a03e2 : SUCCESS in 5m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/899112c57023438eb0f73069db67d3e6 : SUCCESS in 11m 16s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cc9b37993ce24815ad89ac2b2141eba2 : SUCCESS in 6m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3c0b411610544778374fb4bbcd21fe2 : SUCCESS in 49m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d22263b1cae941218d578026ffe2eff7 : FAILURE in 59m 38s","accounts_in_message":[],"_revision_number":9},{"id":"5733411897992380143324722d75c5ed9982dc4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-12-27 08:21:49.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"1438b8f5482c06941a83bf740706b2fd5deb0b1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-27 09:44:49.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9e157f5cb1141c4ae245d6617bea129\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1fcc2a51f64172bb609fc007537351 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/442dfc6d304f457eacde8b38634e3772 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d08429df691e4aa39b0404d92530634f : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4eba7f8bd344689b4f2396a0a4d3ac1 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9ea5e8c6f9749e3af6cc99ad7a10edd : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c3d3d1905f2f43be98b696f2d9a9d450 : SUCCESS in 1m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4b6eb324331b4206a571d4cd96fd75f2 : SUCCESS in 5m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cf3bf68385674171a964646fa97cde60 : FAILURE in 3m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cb2cf60db70f487b984b04ee184b997a : FAILURE in 2m 43s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e074943bf6f44d6fa3a1450a2be88a14 : FAILURE in 7m 55s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e430cdc9e444472baef31d0af44d3594 : SUCCESS in 2m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2562de60a79643899b24b398d9ef9bab : SUCCESS in 4m 04s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/181a2cb855c4402d9e97e0eca031ad80 : SUCCESS in 2m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b35559182a684f23adc9d5effe5f2ffb : SUCCESS in 10m 39s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b2c9b3609c4b4ab79247cab7fad0a6a8 : FAILURE in 4m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6a5696cebff24aafb21e549d68e7999b : SUCCESS in 6m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9ae1f17b564e4efe8afba889af67c202 : SUCCESS in 8m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/92f0c9cf923b44f89c3af1df1aa7f7f9 : SUCCESS in 15m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7de3c4eb020e49e0818c893dbea0e899 : SUCCESS in 3m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f5916c2e67ae4712baeca031b79df408 : FAILURE in 2m 41s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/25b36f68c83542dbb6ca0bc528681da9 : SUCCESS in 12m 40s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b1534a405de74c12bd130da37c7bb34b : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/06c6275c964645c3b62c8137882733c2 : SUCCESS in 9m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/017f67529b9f49428d71222f6ec8ec36 : SUCCESS in 4m 41s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1ff717d69a324eee99b3708e70bbdc82 : SUCCESS in 6m 19s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8be137db0e6b4fc3a8e0b1db5754819a : SUCCESS in 10m 32s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a628815dc1dd48d59c323cde7765828f : SUCCESS in 20m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c0420b80ed6c4b179988320357f893ff : SUCCESS in 3m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1bdd7dfd96bf40f6bd717953425c998f : FAILURE in 3m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6e3eef1a7bd14584930647a1bf146e27 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d51e8cec2f5d4f3297735ea866394a4b : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89a253de553d488c83270916b37a3490 : SUCCESS in 27m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5b3b139d67a7456296061d353477c221 : FAILURE in 5m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fa2acbfdc1114f08acbb7bba54dd5153 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82f2ebf5ca7041d0bf47dbbdcf96d4f0 : SUCCESS in 4m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/11ba51954d9843b0ae08d58c4026cf0b : SUCCESS in 12m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0c50ce502d754c509fbb63a18431e9a0 : SUCCESS in 5m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9bb5e922a4f4f82860d5e82cebfea8d : SUCCESS in 1h 01m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8e3bde374714c798f1a5486ff814992 : FAILURE in 1h 17m 27s","accounts_in_message":[],"_revision_number":10},{"id":"8cbfb9e637df1e6e6d9827f791fad412ddb7f828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-03 08:23:06.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"f1ba5e5ea796061ea3ae9a3432786ecb4acc68f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-03 09:52:19.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50085948382b40a6938da82311ced192\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b78a1543704186b344283ecc4ef59b : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/819048ad8a354d37b5b75aaed0a2183d : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53824848fd0e4403ab014440d45dc25d : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de17ae7dce6d45f5871858313773793f : SUCCESS in 6m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/258850407a5e4e118730a88c268a8a1b : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3b2e84b05464370ba9f5c400bbc97de : SUCCESS in 2m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d66771306420490aa4d3a3cfd8603e5f : SUCCESS in 3m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6eb6d398e410431c8090d1a215449200 : FAILURE in 3m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/425794525a764fd892b6dcecd497ad0e : FAILURE in 2m 22s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/aa79a1e5d3d14538a2bfd02753bd64b3 : FAILURE in 4m 01s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/17f212f10945488993b515036171880a : SUCCESS in 2m 47s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4ff70db4ab0546b690bdc0ed0e187c4b : SUCCESS in 3m 00s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dad83b935c5c4fea9c387258caabe543 : SUCCESS in 2m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5d9eddcc86ed450aa34ecef65b91419e : SUCCESS in 10m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/864181f283bf4442ae2a49a84867c8ee : FAILURE in 3m 14s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6f8f569439d0432fadb9f4cd9a153b8f : SUCCESS in 5m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f84f57e6d72b493f99432755b30e7fd8 : SUCCESS in 6m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5316c4bc111f49f5ac01b282f1d4f0b8 : SUCCESS in 14m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95f6874114134472ba04c3eb511cc588 : SUCCESS in 4m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/99c9c331d9d144f0bb93696b1776da78 : FAILURE in 2m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f3b88f0c09324c91a44d4d858e92d59d : SUCCESS in 11m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cd94016f7c7349d7a2ffc706daed1fa4 : SUCCESS in 1m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8361fd0433a041aa9a599814a310fa9c : SUCCESS in 10m 03s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/5eb81c8b720d48d2a6d75117534bd549 : SUCCESS in 3m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4d8cfea078934be2b721bf13e907ebe7 : SUCCESS in 4m 51s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e87464e416c84bc2ab113409d1c43761 : SUCCESS in 11m 04s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d2e870ccbdbd41058e88b93985061278 : SUCCESS in 18m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/700dad2f1cf24963a6ea333cd42decae : SUCCESS in 7m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5566064faf8e47f1bb8bed03c6c8796c : FAILURE in 5m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1944c3d2b47045c2b8016db50933818b : SUCCESS in 2m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/544c224e54eb44cca136bd189d9b79d4 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/871a5402275d4425983c447e6b9cad4d : SUCCESS in 27m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6281595fcbb34b02bab22e4cd721d3ba : FAILURE in 4m 47s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e4edaf87e0254d19a62e3adfe11a01aa : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a7964fbd92a74d36b318bbed615d572a : SUCCESS in 3m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c1f0138c8f9d45908baf63229e4c2321 : SUCCESS in 7m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/83b1533042af48d69bd3e56f90b944a6 : SUCCESS in 4m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0a7f71f29ca456bba8d48da8ee60a14 : SUCCESS in 1h 00m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83e233ca20b547a3a64fec484e84a3d6 : FAILURE in 1h 24m 30s","accounts_in_message":[],"_revision_number":11},{"id":"09e4201c1193c917f0e9fa1145a93d9f1f0ab565","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-17 08:12:22.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"e630d166de5f10d26dba7fa05b3a0fbd2b226992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-17 09:33:17.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dd6975989b44ed5a4cb98311795e282\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b48ba458e1ab40998827933a7aa15522 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cece94c1458c40e39b11b87721e3c6f3 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae029d7a7ed84cfa99ce29ca516b497d : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f86407bb241b4d8d8a4493751007e1f8 : RETRY_LIMIT in 1m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b491e89e01bb4b07a42732c3552ad9bd : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2db1c1d3eeb747648ec46cf595f83553 : SUCCESS in 2m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a96220e41f794897a1e0a061a85cf164 : SUCCESS in 5m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bdfaca64f7514de5baf9a76be251bbda : FAILURE in 3m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4b6b7ac55c354f1db4aee480d0929886 : FAILURE in 3m 59s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a1cc22d5ff754011af292741c29589cd : RETRY_LIMIT in 1m 38s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/bef2d3d3c29941df85ff0c8bfef348d6 : SUCCESS in 4m 45s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/38b8dfefc88a4d599ab0ae28775fe3d4 : SUCCESS in 5m 23s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/430ff0b51f1e43c09bbd75e17de46214 : SUCCESS in 4m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/46691ada6e4047eb9802b4750063094a : FAILURE in 8m 53s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b40f78d13212452cbdbd56245c31d051 : FAILURE in 3m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2256e29685754926b0e77861266a1ea0 : SUCCESS in 5m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/09a293f52df94404a8003d2d72fe9293 : SUCCESS in 3m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/99f20328e1f24552b258b70d541851f0 : SUCCESS in 17m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ab9ee24832064dbda05174a70f3dcebe : SUCCESS in 8m 11s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6b1c1c8fe4ea4ca0b63e0b213b9215ce : FAILURE in 4m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/425df557546441a4b12129b1cb728946 : SUCCESS in 8m 16s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5311c9bdc7884883abcfc6ef458b5dea : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/07c2957f3c4d4f62ae7246dde3ffcf85 : SUCCESS in 9m 30s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fc1e9bfaeaf945eca7e6b5638abeecb1 : FAILURE in 5m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b58159cf88a74a2e87b263bad5307e9a : SUCCESS in 4m 43s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ff4a8ffd55a34079a0c47ec740e3bde4 : SUCCESS in 11m 50s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0172b0dc85b445569abdf3388618d0b1 : SUCCESS in 18m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ed2d1683d3a2481bb2e099e55c7ffb57 : SUCCESS in 7m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/edeefaa90ee4407799208c3370a8d22a : FAILURE in 5m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/866504d2cd9e4bc1be51a1a4f40e2f76 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ddbc5841d93d4a2fae2e443417d8ee76 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ddccfab0dc1845cc84e1d4d6514db8b2 : SUCCESS in 24m 04s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3c239942e3a843359002fba8b1ff441d : SUCCESS in 5m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a78941befc6e4b35957f99c0bc2969fe : SUCCESS in 3m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/047b4c5d01804543946f77f400ea84b2 : SUCCESS in 3m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0bf709c3747e48e2acd0482dbc7fbf27 : SUCCESS in 17m 54s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d6cd2b7a76b546a5b5ac842f50113359 : SUCCESS in 5m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dc7cadb193d46af81d657013316d4f6 : SUCCESS in 1h 01m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55368a20ac3041b88d6faf99e0817efa : FAILURE in 1h 19m 25s","accounts_in_message":[],"_revision_number":12},{"id":"d2d1c92670a42ec1464096ac33afa05f9496ccef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-24 08:54:57.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"36570e47fa73bc63d7c72ecd0b056de657857e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 09:43:34.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76a8b45755d34e618ce77e7aa0578237\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d0061908e5497794c79ab44d6b6ecd : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bfed1ca9d9142db997bd3ca2999b735 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f8bbcdda4d9437b99e90712d12bb104 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b308d49e53ad4c788ebcd8aa40f06675 : SUCCESS in 2m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c82d3886d23d47deaa33f7a1d30aa307 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9836bc74e93348b8b1f20e9609ff5e79 : SUCCESS in 3m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c562fc8151e74b84a2a09a8fb1d7a4ba : SUCCESS in 5m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/55e6e594ab424823b808f5d75d8c1728 : FAILURE in 4m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/174d22b09f274cd883e5461c6238b651 : FAILURE in 2m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d1c8fdd46080405c8a01a08859830c20 : FAILURE in 2m 26s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3d0a721970f240c6860e720390bb9be3 : SUCCESS in 4m 26s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ea33a8268b5540ebb79a6bf22fe3cee2 : SUCCESS in 5m 10s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/eb58a22b12914ec9ad74025a6164f010 : SUCCESS in 2m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1694bcd027444c53830f38da3f37f777 : SUCCESS in 8m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/045798e93a554593ba05bc2a37dad573 : FAILURE in 4m 25s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/98d416925049453aabd712d972e147fb : SUCCESS in 5m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c35e7c9df65c46679233410eaac4da4d : SUCCESS in 3m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/997b568749f7403a843c18fc46319377 : SUCCESS in 19m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f16da720c4e49708b9297ef7e1b5828 : SUCCESS in 6m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1f3a41c873eb438ebfe5388243f6b4b2 : FAILURE in 4m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5f406e5d6d3b4031a30cb9bb51dc7431 : SUCCESS in 9m 17s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/245d56afe7114e8894a702515cdd9413 : SUCCESS in 2m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c233b8b0b5734927afbc03e161a22bbf : SUCCESS in 11m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/846689fd7dee4803b5ba9050501d9f71 : FAILURE in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/92f8a1f55d954c4a8fb9e2f2d64f34f5 : SUCCESS in 5m 37s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/32bd58ecf52c4566ba3907b48b18c9d0 : SUCCESS in 10m 55s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/82c819439ebb4af0802ac8d751e4baf7 : SUCCESS in 33m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dc8ca4bf851f4acbbf623f7ed84d8ba6 : SUCCESS in 9m 50s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9be08d31a2a3487c9f16ad563d206e5f : FAILURE in 5m 51s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/44e07b6ec8874f8eae47982fc2553729 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/70a59abe490d4f7a9628f7c49541ef0f : SUCCESS in 2m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9acef3a7c6d44e409758990c00344266 : SUCCESS in 25m 04s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b1834383dd4647479ace1a5d09672ba6 : SUCCESS in 3m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/04aaf645cf0e4bd68732cb9dee582f91 : SUCCESS in 2m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2511d915af6431391bbcf5fba46ae26 : SUCCESS in 7m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2ab37367da734191a10adff852e91378 : SUCCESS in 7m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2162ef1490bb44d19714866e9bbcd1f5 : SUCCESS in 6m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c281194545c74accb34f48d66c12940a : SUCCESS in 36m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bab9501a540c41a792dedfd2f0e479f3 : FAILURE in 47m 54s","accounts_in_message":[],"_revision_number":13},{"id":"3ab525c783a260344bc68c819fb3e07ea7bb1ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-01-31 08:31:03.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"9ba44949deaac1966d5725b2f687ba2159ad4b3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-31 09:31:53.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/baa379ee31d64c9d86b21052ed251020\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e198c5ed18043d48f81de5a8705561c : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f416fbee5aef47e898aa8fec6045f9ea : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6d2672abff3420aafc18b1d0a7c3ae2 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e471a74a451647b5ae15f1a7953f5f38 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1cbbe08344044d92979a1a9bc749343d : SUCCESS in 2m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55b3dc3c352c4a169cb8c7c4d4349437 : SUCCESS in 2m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/befdc424f2424314acf1f01ccbf656f7 : SUCCESS in 6m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/95703f10fb3f49deb03745d1350f72d3 : FAILURE in 4m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d399c5edf5024acfb5c3cf72f3c1c8a6 : FAILURE in 3m 38s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/fc03f68252b14825a1d2e1cfcf19ddeb : FAILURE in 2m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e6c10112471c44229eb9274abdcb923c : SUCCESS in 3m 18s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2b0f6066a95043d09541808bb4e94020 : SUCCESS in 4m 29s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1ca816833cb944c495e55356cf32b9a1 : SUCCESS in 5m 07s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7603d466761c4f2198520b55c5a40f51 : SUCCESS in 11m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/280fff80e7074512b0d460349d42b201 : FAILURE in 4m 54s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a2e5e2f2b46a440bbe0e5958fcac06de : SUCCESS in 5m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f06e5d15dd7047a0b6391e78ec8cdce3 : SUCCESS in 6m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/eb6ff1c9a97840c2a425c9ff825b1c29 : SUCCESS in 19m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6eb35722e5f74a06a8fd10502eb8d73f : SUCCESS in 8m 05s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0b146531d82444389e240c26ecc530fa : FAILURE in 3m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c00f319273a14123a37b78b2a91d17f5 : SUCCESS in 13m 15s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/278978fb40714ac59ead221c2e8810a3 : SUCCESS in 2m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4ffb968d0c2245b4af97f69e3b7aa619 : SUCCESS in 5m 15s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0524892128dc4e998a1c9568966a84f0 : SUCCESS in 3m 49s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e4dece2001e04adba4d13a238377e3d9 : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7232145cbe30498294cb6aac8c3f0146 : SUCCESS in 12m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0e9244232acc420ebd427f47a5e2c81c : SUCCESS in 28m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/efe8901b77a84aa3be801ab09b999795 : SUCCESS in 7m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ea270894b17548e6a43b5dc74e5fc794 : FAILURE in 5m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/db24fb78250242efaa6b20832809127d : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/81402c677b584345a42c92ca153e651b : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7d9d29542b64e29bbe6e9e0f9e3a3cc : SUCCESS in 26m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/40a3e0e01d464cb9aea3d0b9821099bc : SUCCESS in 5m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/39ba6520bfa84fe29e0658de44cfbeab : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc06dd987fb84be9941c05d8a7633391 : SUCCESS in 6m 48s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/51fd0d0a47624342bd47ac73a1513b53 : SUCCESS in 14m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ea98748f85c6408d8d37f1ab9acf2bc2 : SUCCESS in 5m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83d2a2b35e984060ab5058c1e9fc6fff : SUCCESS in 50m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43e8030e4a044b1cb05f13dfd20f11d2 : FAILURE in 59m 20s","accounts_in_message":[],"_revision_number":14},{"id":"cc208ea8d51f5a21780000b6b0c2ff6c53b4b71a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-07 08:22:34.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"304aa5e84941ffbb342454f9528172314e3ebc3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 09:22:45.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ed473a84fa94c6c90e6416daa217cf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b57cfe340d4e3194a1a2c8a730213d : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c0a515a7c3f469db55c41c6587a6671 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7e252f2422c472faaed38048c6fe9a4 : SUCCESS in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f9253c8ccf84de0bfd516c5ece97755 : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d6a35b6f67a452e8e6acfc89c9cc28e : SUCCESS in 2m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d19808504b3b4724b4fa40495e83c451 : SUCCESS in 2m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7075664e1c5f4d8ea0a576bb0868ac44 : SUCCESS in 2m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/93c29b967dd04e05ba0c1886719253ad : FAILURE in 5m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ce73202336a24e7cbf898d0971c600d6 : FAILURE in 3m 44s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ec184b13debb45ff858e25c875752840 : FAILURE in 3m 41s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/79352c9910f94c9781013ebf7bbf99a6 : SUCCESS in 6m 00s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/09a0d21e9a074059bb80c7d13b79f8c6 : SUCCESS in 2m 51s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/971302567e5c408f91f9d0f7bc62d6a3 : SUCCESS in 3m 58s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/85c889a0906c43c09019bddad7c41490 : SUCCESS in 8m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2eaaf763535c4ace8b0b86d492ea7b81 : FAILURE in 3m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/81eda932ea0d414487fa442db4c60640 : SUCCESS in 3m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/836d4a8bd6a84baab49080af18791edb : SUCCESS in 6m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7c05158fcffb4e3894e0ee7993009aef : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/417ae0a66bf64b2aadb79aa161730a96 : SUCCESS in 6m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/71a64732d1db42f78442a2b71219e134 : FAILURE in 4m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b0adacb83d5d41bfa0a063413b387fca : SUCCESS in 11m 05s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ee4713bdab72438ab3cefe4a46939ed2 : SUCCESS in 3m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0996326b7a7d47d6869d85030f8388e4 : SUCCESS in 8m 19s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ca34d6d3d8db4305a584b075d4c213fc : SUCCESS in 3m 09s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6a39a59d3bf24827bed57f08b75d9e2b : SUCCESS in 6m 06s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/070f90e2fa344b9090254210212219d8 : SUCCESS in 10m 05s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/8840e005fd344a78897c75c4cc534537 : SUCCESS in 27m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b0d2f476bcf04ff2ba8eaf735819b455 : SUCCESS in 8m 13s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/22aaa03e30844da6a17e0967594917bf : FAILURE in 3m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6eb8865d81d94315a8d748cc06a6197d : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e15abc2c7db48ad972e3ccf25de6f80 : SUCCESS in 5m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c2de1b11e10746b1a72bb33692b051f3 : SUCCESS in 30m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f90b4d7d2d9d4e8cb5ea386375f30d28 : SUCCESS in 6m 51s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/943be91d22eb45f39244bddce60c9916 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a23eda52ded1474eb3ea06b693f21321 : SUCCESS in 5m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a5ceb0e96cb9427f9dbe72071e4d56cb : SUCCESS in 19m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ae71073618c34194b4d70eedb4b1794c : SUCCESS in 6m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bd5fcffd06a4b47afbcb554e0341b46 : FAILURE in 55m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c072607213164c85ada362ab2778ddaf : FAILURE in 50m 03s","accounts_in_message":[],"_revision_number":15},{"id":"561b57ba5e470be3630503e64c9b3c5c868bceaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-02-21 08:22:23.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"54e6c01b86939e5be58e7ab7adde162b6be8d881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-21 09:35:03.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4600823a272945b2b92bb9050916de81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/462adba9f66b4fc0a3464b19a2573466 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bdb9a0606714b9985b92cb2f79f2750 : SUCCESS in 1m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ded297104ab40919b84b4d5df028343 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aeb1642c568949079d152349193039e5 : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d4e6638cf3164d6faa54ed58676e6a75 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf182a7ff142468c88059eebae823e06 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b837e8b243814b2086dc8256bb676a41 : SUCCESS in 2m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bace6a5d18944241b14b161cd395319f : SUCCESS in 5m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2780eb18bdfa4384b666b295899dc1d0 : FAILURE in 3m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b2c7bd143a0a42528999e034351e29f0 : FAILURE in 3m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f78f5b922c3b43c2869a21a1161dea4f : FAILURE in 2m 26s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3b5449e9d1b64de6b81648b52b00f91a : SUCCESS in 4m 20s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/153815700a8048b6ba06d85591d8363f : SUCCESS in 3m 41s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4c207061a301452faaee7bcb95eb07c1 : SUCCESS in 5m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c81af16a97c840b18753afa9b85402d5 : SUCCESS in 8m 50s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/71b6b43db08642f5a668a828dbb52043 : FAILURE in 6m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c92ceea724ee4aa8890a7f9494fff243 : SUCCESS in 4m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3768cc459aed4e3e8e2382462ee04e82 : SUCCESS in 6m 29s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6483afb387b444a58e127e2fe5dc099c : SUCCESS in 16m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f117285bd80f42dc97c7bb6505356720 : SUCCESS in 7m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d42828a9ac6a474f8979c6cc76551e4b : FAILURE in 3m 49s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/052b994925074c659df64fae43770d87 : SUCCESS in 10m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6ab432c284284a488a9732aceeeffdf8 : SUCCESS in 2m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/da4e06f4dfb04363836f16992497f302 : SUCCESS in 8m 54s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b0d2258f85374321a91d416bc87e702f : SUCCESS in 5m 45s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a945df92d34b465dbd7cc9e841aef6dd : SUCCESS in 5m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e6ef9189738342138d1b33808841d35d : SUCCESS in 11m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/410ea877c56745b6b6311f68b836c7e5 : SUCCESS in 33m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0d2404ae73814973b3deaa0e04f09ef8 : SUCCESS in 6m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/154b5cd40d2a47c2af0939df9b897fe1 : SUCCESS in 13m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/31736aaaaf534d7190636810b108943b : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dc828e9674c14788a1d39f812988e098 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/696ed709347a4e73820664e245d068ad : SUCCESS in 25m 25s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/98166cdf751c4763abe3b89b43b7a70e : SUCCESS in 6m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7f85c6c44c844ef6a95d2dce368a6375 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ada004241ac43f9bafaf62fcbc6c1c0 : SUCCESS in 5m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e82f55ec3fad4726b18248028151b244 : SUCCESS in 12m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/3cb1b2e4a781457abb4118b79f2dc712 : SUCCESS in 6m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b9abf25edf14ffc8ce29d5d49c2d424 : SUCCESS in 57m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e8b045b3cc440eab1c3b889123b8cb6 : FAILURE in 1h 07m 48s","accounts_in_message":[],"_revision_number":16},{"id":"2949e6abe754d93fa33e4d4372d8de41766167c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-22 09:35:11.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"7ba0ecaee2a9b5299f9646d723bc251a0a211b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 10:56:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75c1b4913d2044a5b747a1e49c3615c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d883ac5bccfe4f41a101ffbd01923438 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc35388c364a4b148dff088fe3988ac9 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4aa74e12a9b47aa98b42faa06ca24a8 : SUCCESS in 2m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/550e1cf2ea5c49daae62ec4bd533de73 : SUCCESS in 3m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0b22575c461949c6becbaeab7f953529 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0060056e43a8485d97ad6d882b0b3283 : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4df0fd18d7146bda13331835d0c8832 : SUCCESS in 2m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/093fe03f55d546afafff0a0ccc680195 : SUCCESS in 5m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e5173f47824a46ac8dcfd3a416d86534 : SUCCESS in 4m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ba3ea24cfb3b4d5dbe2ed784a845e9e0 : SUCCESS in 8m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8d9d6e123783471197dfbd562a273281 : SUCCESS in 3m 13s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f2628899d0dc4bd4a069a2fbe8e8690a : SUCCESS in 2m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2385142f2a1843619fb3879e9072abac : SUCCESS in 4m 57s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/cbc24875bc934abe8ca0ac921bf9f083 : SUCCESS in 5m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7693588d734a43918e3f4cfc93f4a1cd : SUCCESS in 9m 51s (non-voting)\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/198c3beaa941442f84a5d3a048c08d4e : SUCCESS in 5m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/71083662af73466782cf6855edc8a4a0 : SUCCESS in 6m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/91e90e1816934d53a1e9dc43658582a2 : SUCCESS in 6m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a598b3ef4701464889e3d54a0de6cd76 : SUCCESS in 8m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54822110bafd47219c8301e5e3031a49 : SUCCESS in 6m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9fe5bf636db748b2bbbb8de9c7264fdc : SUCCESS in 11m 02s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e7e900a887b24c81a0381f2249a485ba : SUCCESS in 11m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a7aadfa4172b47c2a79bce790353afcb : SUCCESS in 1m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/828db75507de47e7b7f8570f4afaa63e : SUCCESS in 8m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c3b90edb737945ee9478db6c5f86d4b5 : SUCCESS in 5m 57s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3ed5e58a9b75498192a5e0b3d4bb191a : SUCCESS in 2m 43s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0552ffb3c58546e3af0995665388fcb6 : SUCCESS in 12m 55s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/45c98a476b0540b6b92f39f45ba10254 : SUCCESS in 18m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/24885ae1da164871b17298f37fde43a6 : SUCCESS in 6m 35s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/41764e82023b491ca489431ee6d770b3 : SUCCESS in 16m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f8bb380e52624df88d2fc90c94d81f12 : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/638b08f38fd8499ba0417a3c5fef4648 : SUCCESS in 1m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bbd6a505f3aa438b82666db051324ee4 : SUCCESS in 25m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3d2628399ecf494fb76902d1c7333aab : SUCCESS in 4m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e20e1bb6b7834c7899da11d19ab77f5b : SUCCESS in 2m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0471dd9e35764852bcd62fbd7d69190d : SUCCESS in 5m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/289ccf5d24e34f2995d574c81ae3559e : SUCCESS in 12m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a226deb47aad42a992ca9dc3c4d2759a : SUCCESS in 5m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cbc4c15f93849beaf19d55f7efce6f2 : SUCCESS in 1h 06m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4faa6f353aa42348b189e4406a93f2e : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":17},{"id":"9fa279172ebfd0841a74f7f7802645e420733366","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-22 11:22:43.000000000","message":"Patch Set 17: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"f2f9f484edff36d969c77d93cf62cc5a74f7a0e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-22 13:57:12.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":18},{"id":"c641c8d34f8277bc44d2b6f08c0f33dedfcf625d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 14:59:25.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1b99d250df84bba8677ed54e9cc8f92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c2d8e8a06da49fb930f50d385b4d509 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647d879d26b94a68bdcd182a0e241f91 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/942a214c983e4cd9a148f0fabe84897a : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b38585ec5c1e41659d8b0a0196dcdccb : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f418bc09082477f8b33a23b2d853de8 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b00bedef47e04093852b4ce15076b364 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f1b40d6fb42462399062ffc88325e16 : SUCCESS in 1m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0d631224e6f64634b01971d25055b784 : SUCCESS in 3m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cd9a48532d33429d85c2e0fbec68379e : SUCCESS in 5m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/303f2eab50f943b7afa8ec3a1e09e5bb : SUCCESS in 2m 53s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d8d6b9e9b2e84275b427bf8cd0443eae : SUCCESS in 5m 40s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9d814810ba134859a98e042da4dedb0e : SUCCESS in 6m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/01a50f5ba6094b3d92d62defb24ef397 : SUCCESS in 3m 06s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2242ac6ecdd04212ac709cd6006c732a : SUCCESS in 5m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/95cc3d209f5545a09a204cd4854b6689 : SUCCESS in 12m 38s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/856ff66d5459472290d2411add20eb78 : SUCCESS in 4m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3a08c8f2167740c593888cc6ab9ab20e : SUCCESS in 4m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c46c4eacae894f3885baaa539a2806e4 : SUCCESS in 8m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e93483505bf6429a89df6e3c7565d656 : SUCCESS in 17m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e8bd6a912f0b446d9eaf4483cb7ee7fa : SUCCESS in 5m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/47169627eb61446d81da539f216077c6 : SUCCESS in 10m 14s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/59af0628c85342bfaa5498d21e6713a5 : SUCCESS in 9m 15s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fd1da0a47ad54a4cbb7e244a78e8b0e0 : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/52b0f19c603f411193954cc87c1bd773 : SUCCESS in 6m 01s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bc39d3e3f6b849f1a787b5f648fc9def : SUCCESS in 4m 06s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6efd631e9f004950bdefb22cbff0df5b : SUCCESS in 3m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f7c70d6123904c0eb1142b0377502567 : SUCCESS in 11m 52s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6dbd7d660fe948c0a89d248917b30148 : SUCCESS in 32m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d904ca05a7144787841fc118a7d06c50 : SUCCESS in 7m 15s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3b9be4c453ab4adb8dcf3eaf2046c869 : SUCCESS in 13m 42s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6cb6cb38aa0642a187230a5fe51df5ed : SUCCESS in 2m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2d3b0e04e35e4ea2977e43a61039e593 : SUCCESS in 4m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0ad9375b6464a07bb81f3fc1973a42d : SUCCESS in 21m 53s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9925ba0185ee4dd5b2b13b358cd71a1e : SUCCESS in 3m 25s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e846975dfe6047cfb042159590023c6f : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/678d34a058294ce6ba8dc0c19067564d : SUCCESS in 4m 07s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b08472a94ea243e8b14201790ac6a0cc : SUCCESS in 19m 39s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f94b5b4e04004c56bf847dbfcf849204 : SUCCESS in 3m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc2c213a6ec4463db3d4a54b62eeb298 : SUCCESS in 32m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bebe6ed87e674198869d6158ebbe8360 : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":18},{"id":"8aaa90dd541003589416466e92eca825f2001c5b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-23 09:58:01.000000000","message":"Patch Set 18: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"aad662ca1478e57203d6972e5be33c4922285798","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-02-23 10:17:12.000000000","message":"Patch Set 18: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"61fdcb37f1e88bc558847184317c84cef960c837","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-23 10:18:42.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"eb028ba58452d431ff8f2ab65521a7b8e5194443","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 10:19:14.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"1189802a392434d0f648c762ae1c19e1a9b4a25c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 11:47:11.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26d25ed42e174eb9952da7e3cd589f71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc57927733643b4b4166a8129d5aafa : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a75a30757d7f482ab2cbf925161a26ac : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/599d4c57fd6b47de9c253abfca0ed602 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f223b2b7c0224e0ca16c07bb40a88681 : SUCCESS in 2m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0314f555b10a4a91806313d6a2992dc8 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2803c4095b3640ae90ba3e61595ccbc2 : SUCCESS in 2m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7b4a194a96614ef5802f2201543aa35b : SUCCESS in 4m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e90268ca645f4f629a29c663fbe1bc67 : SUCCESS in 5m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a523606dbb64472aa57d19a4aa9403e2 : SUCCESS in 5m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5783af0c32094681b48533ecedda93c9 : SUCCESS in 3m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4b00babc7636484897ff0ae053019356 : SUCCESS in 10m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/bec69edc20c04503ba2354a5cebd6e20 : SUCCESS in 4m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cc072ef3a34e4145acff221c99054ed5 : SUCCESS in 5m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/562e32c296d14479910f46c22e6e6ed7 : SUCCESS in 6m 31s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3a3d49e0d6724c3ab5a24eef420b6d4c : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dce9d5c91834db8ad3b2d241edc120d : SUCCESS in 6m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2d40a71081a141cfa8a379f8ab718228 : SUCCESS in 4m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b56d214aec3d4bdb963553a53df9d3f8 : SUCCESS in 11m 38s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/11f4255fbf514e728c99f58da151e26a : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c3927ec715644e08b4cd28b2477b06be : SUCCESS in 6m 24s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d06ed3f6dae14bf0b907d3bfbd66b9a8 : SUCCESS in 24m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c516214da83345688cb8811899e92dae : SUCCESS in 7m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ef83e8736e814e1d891a7277ec377057 : SUCCESS in 16m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9a50a9f1aa4b40f4954ff6b6698581da : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/24ca8b6c3ec942faa224168fae97f8fd : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8fd0054c6b3f46509b7ba91b9780fee7 : SUCCESS in 24m 49s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1665f2425816473ca9fef11ad3fce86a : SUCCESS in 4m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/220e58b75c484f40bc24e923feb1013b : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c8c09dc124042fe844563e2213e630d : SUCCESS in 3m 53s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/99745f99a1f54db0a7b050c397078bc3 : SUCCESS in 8m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4a0179a7db854ef09b5d7ce5c6e8f9b9 : SUCCESS in 5m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/626cfc5f59214ae4ab7ac4dfe09a9495 : SUCCESS in 36m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e9528ecadcd41719b6b12f3b3012d95 : SUCCESS in 1h 26m 10s","accounts_in_message":[],"_revision_number":18},{"id":"6636b77517527ecc8738a3108c891d2b3e247882","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 11:47:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"323fbc669eb23b9b24da69e0036534bb60672475","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 11:48:00.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6c8951cc00245ada20762dd5cda89cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5233e813a2e4d738fadaaeaddfdd1ea : SUCCESS in 35s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"5d11acecfaa7cc14affa3697a1d0fe2931417df5","revisions":{"488831d21d12729d6b873ea950e50bd3745e25b9":{"kind":"REWORK","_number":1,"created":"2025-11-01 08:20:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/1"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-01 08:20:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-01 08:20:15.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/488831d21d12729d6b873ea950e50bd3745e25b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/488831d21d12729d6b873ea950e50bd3745e25b9"}]},"branch":"refs/heads/master"},"a7b8a981bcf89c7c362b859565489f4ebeb33894":{"kind":"REWORK","_number":2,"created":"2025-11-08 08:32:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/2"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-08 08:32:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-08 08:32:37.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7b8a981bcf89c7c362b859565489f4ebeb33894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7b8a981bcf89c7c362b859565489f4ebeb33894"}]},"branch":"refs/heads/master"},"f6c889bb33ec9a97d51c6017cdd16937b1619a5d":{"kind":"REWORK","_number":3,"created":"2025-11-15 08:27:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/3"}}},"commit":{"parents":[{"commit":"8b17b6b0f89e941a1b065aa608f44491bd3b832c","subject":"update constraint for os-ken to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b17b6b0f89e941a1b065aa608f44491bd3b832c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-15 08:27:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-15 08:27:39.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6c889bb33ec9a97d51c6017cdd16937b1619a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6c889bb33ec9a97d51c6017cdd16937b1619a5d"}]},"branch":"refs/heads/master"},"d60fda9f1673954435144617939273ae33a5bfa2":{"kind":"REWORK","_number":4,"created":"2025-11-22 08:19:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/4"}}},"commit":{"parents":[{"commit":"6b2bf0feb131abed508d0cacf8acba6f1d9ac3a9","subject":"Merge \"update constraint for python-zaqarclient to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b2bf0feb131abed508d0cacf8acba6f1d9ac3a9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-22 08:19:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-22 08:19:26.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d60fda9f1673954435144617939273ae33a5bfa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d60fda9f1673954435144617939273ae33a5bfa2"}]},"branch":"refs/heads/master"},"9fc8ee7aa4def5bda480022520b5f61c0b9a122f":{"kind":"REWORK","_number":5,"created":"2025-11-27 09:28:23.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/73/965873/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/5"}}},"commit":{"parents":[{"commit":"3bbfa41941834981326b00eee6bf0f3d8157e2ba","subject":"update constraint for ovsdbapp to new release 2.14.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bbfa41941834981326b00eee6bf0f3d8157e2ba"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-10-25 08:15:31.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2025-11-27 09:28:12.000000000","tz":660},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nKeep version specific pins where applicable.\n\nPin versions for packages that are not ready for updating:\n\n- oslo.limit:\n    See: https://review.opendev.org/c/openstack/requirements/+/967742\n- gabbi:\n    placement: \"AttributeError: \u0027NoneType\u0027 object has no attribute \u0027api\u0027\"\n- infoblox-client:\n    designate: |-\n      [designate.plugin] Loaded plugin backend:infoblox\n      [infoblox_client.connector] Cloud WAPI version detected: 2.10\n      [infoblox_client.connector] Configuring session\n      [urllib3.util.retry] Converted retries value: 3 -\u003e Retry(total\u003d3, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)\n      [infoblox_client.connector] Authenticating with username and password.\n      [infoblox_client.connector] SSL verification is False\n      [infoblox_client.connector] Connector initialized with options: {\u0027host\u0027: \u0027192.0.2.1\u0027, \u0027username\u0027: \u0027****\u0027, \u0027password\u0027: \u0027****\u0027, \u0027http_pool_connections\u0027: None, \u0027http_pool_maxsize\u0027: None, \u0027wapi_version\u0027: \u00272.10\u0027, \u0027ssl_verify\u0027: None, \u0027cert\u0027: \u0027****\u0027, \u0027key\u0027: \u0027****\u0027}\n- pyasn1,pyasn1_modules pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n- boto3, botocore, s3transfer:\n    glance:\n    - https://review.opendev.org/c/openstack/glance-specs/+/962248\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nSigned-off-by: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fc8ee7aa4def5bda480022520b5f61c0b9a122f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fc8ee7aa4def5bda480022520b5f61c0b9a122f"}]},"branch":"refs/heads/master"},"de6dea1a49a8db83b6a2f1c87749f3ca61e735e3":{"kind":"REWORK","_number":6,"created":"2025-11-29 08:29:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/6"}}},"commit":{"parents":[{"commit":"44dbd103640a38272df3ad29048b4c5dece2c1c7","subject":"Merge \"update constraint for sushy to new release 5.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44dbd103640a38272df3ad29048b4c5dece2c1c7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-29 08:29:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-29 08:29:21.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de6dea1a49a8db83b6a2f1c87749f3ca61e735e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de6dea1a49a8db83b6a2f1c87749f3ca61e735e3"}]},"branch":"refs/heads/master"},"d895f3d07fc710c7a678386045952bcad4b2d647":{"kind":"REWORK","_number":7,"created":"2025-12-06 08:37:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/7"}}},"commit":{"parents":[{"commit":"f5707cc58f7152a82d93933a9863587af5f0ffa7","subject":"Merge \"update constraint for microversion_parse to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5707cc58f7152a82d93933a9863587af5f0ffa7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-06 08:37:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-06 08:37:52.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d895f3d07fc710c7a678386045952bcad4b2d647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d895f3d07fc710c7a678386045952bcad4b2d647"}]},"branch":"refs/heads/master"},"596224798cf481c166a08284ca36bff561e48d73":{"kind":"REWORK","_number":8,"created":"2025-12-13 08:18:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/8"}}},"commit":{"parents":[{"commit":"0cd36c4f8da2ad8c6a02ef71ab9f239734593760","subject":"Merge \"update constraint for glance_store to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cd36c4f8da2ad8c6a02ef71ab9f239734593760"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-13 08:18:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-13 08:18:21.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/596224798cf481c166a08284ca36bff561e48d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/596224798cf481c166a08284ca36bff561e48d73"}]},"branch":"refs/heads/master"},"2e7da3eda4f85d2624025b9c7dc87e6548298f6c":{"kind":"REWORK","_number":9,"created":"2025-12-20 08:21:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/9"}}},"commit":{"parents":[{"commit":"0cd36c4f8da2ad8c6a02ef71ab9f239734593760","subject":"Merge \"update constraint for glance_store to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cd36c4f8da2ad8c6a02ef71ab9f239734593760"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-20 08:21:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-20 08:21:22.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e7da3eda4f85d2624025b9c7dc87e6548298f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e7da3eda4f85d2624025b9c7dc87e6548298f6c"}]},"branch":"refs/heads/master"},"53c93fb72b48059744d288d2ce4e340fa432ff31":{"kind":"REWORK","_number":10,"created":"2025-12-27 08:21:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/10"}}},"commit":{"parents":[{"commit":"0cd36c4f8da2ad8c6a02ef71ab9f239734593760","subject":"Merge \"update constraint for glance_store to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cd36c4f8da2ad8c6a02ef71ab9f239734593760"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-27 08:21:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-12-27 08:21:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53c93fb72b48059744d288d2ce4e340fa432ff31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53c93fb72b48059744d288d2ce4e340fa432ff31"}]},"branch":"refs/heads/master"},"f35f1cc489646e499ca5edc41ca6ed0842abd067":{"kind":"REWORK","_number":11,"created":"2026-01-03 08:23:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/11"}}},"commit":{"parents":[{"commit":"0cd36c4f8da2ad8c6a02ef71ab9f239734593760","subject":"Merge \"update constraint for glance_store to new release 5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0cd36c4f8da2ad8c6a02ef71ab9f239734593760"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-03 08:23:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-03 08:23:05.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f35f1cc489646e499ca5edc41ca6ed0842abd067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f35f1cc489646e499ca5edc41ca6ed0842abd067"}]},"branch":"refs/heads/master"},"2e4709295014c98563530ecfc3ce105d4f07be8c":{"kind":"REWORK","_number":12,"created":"2026-01-17 08:12:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/12"}}},"commit":{"parents":[{"commit":"179c2a9f8d4c801210f99f6a28846df7238a72d2","subject":"Merge \"update constraint for oslo.limit to new release 2.9.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/179c2a9f8d4c801210f99f6a28846df7238a72d2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-17 08:12:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-17 08:12:21.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e4709295014c98563530ecfc3ce105d4f07be8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e4709295014c98563530ecfc3ce105d4f07be8c"}]},"branch":"refs/heads/master"},"864f04c46c7f82059505776b54084508e3ebce44":{"kind":"REWORK","_number":13,"created":"2026-01-24 08:54:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/13"}}},"commit":{"parents":[{"commit":"b6633b1e375b7fbb288f7c14e84cefdcfd1d276f","subject":"Merge \"update constraint for kuryr-lib to new release 3.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6633b1e375b7fbb288f7c14e84cefdcfd1d276f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-24 08:54:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-24 08:54:56.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/864f04c46c7f82059505776b54084508e3ebce44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/864f04c46c7f82059505776b54084508e3ebce44"}]},"branch":"refs/heads/master"},"15684750f567d733ffab171a6009ed2eca93552e":{"kind":"REWORK","_number":14,"created":"2026-01-31 08:31:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/14"}}},"commit":{"parents":[{"commit":"cc2a90b26f418a8d6a7e2a0ffb0586bd5a92f034","subject":"Merge \"update constraint for oslo.concurrency to new release 7.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc2a90b26f418a8d6a7e2a0ffb0586bd5a92f034"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-31 08:31:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-01-31 08:31:02.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15684750f567d733ffab171a6009ed2eca93552e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15684750f567d733ffab171a6009ed2eca93552e"}]},"branch":"refs/heads/master"},"bde73bb2a70de20ebbb86e97054483f49658e50a":{"kind":"REWORK","_number":15,"created":"2026-02-07 08:22:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/15"}}},"commit":{"parents":[{"commit":"a3bc8a52ae4767d5df609797e46c86ab633a044c","subject":"update constraint for os_vif to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3bc8a52ae4767d5df609797e46c86ab633a044c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-07 08:22:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-07 08:22:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bde73bb2a70de20ebbb86e97054483f49658e50a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bde73bb2a70de20ebbb86e97054483f49658e50a"}]},"branch":"refs/heads/master"},"8b4ae52af8c192a964f83d8164170cd7b460301d":{"kind":"REWORK","_number":16,"created":"2026-02-21 08:22:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/965873/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/16"}}},"commit":{"parents":[{"commit":"33bac846633a046569e3b01a70c8ce27ec2970b0","subject":"Merge \"update constraint for neutron-lib to new release 3.24.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33bac846633a046569e3b01a70c8ce27ec2970b0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-21 08:22:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-21 08:22:21.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b4ae52af8c192a964f83d8164170cd7b460301d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b4ae52af8c192a964f83d8164170cd7b460301d"}]},"branch":"refs/heads/master"},"a0d6cf3e8c184718d779d5ccdd714e103a65cfd0":{"kind":"REWORK","_number":17,"created":"2026-02-22 09:35:11.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/73/965873/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/17"}}},"commit":{"parents":[{"commit":"33bac846633a046569e3b01a70c8ce27ec2970b0","subject":"Merge \"update constraint for neutron-lib to new release 3.24.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33bac846633a046569e3b01a70c8ce27ec2970b0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-21 08:22:21.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-22 09:31:26.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- gabbi:\n    placement: \"AttributeError: \u0027NoneType\u0027 object has no attribute \u0027api\u0027\"\n- infoblox-client:\n    designate: |-\n      [designate.plugin] Loaded plugin backend:infoblox\n      [infoblox_client.connector] Cloud WAPI version detected: 2.10\n      [infoblox_client.connector] Configuring session\n      [urllib3.util.retry] Converted retries value: 3 -\u003e Retry(total\u003d3, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)\n      [infoblox_client.connector] Authenticating with username and password.\n      [infoblox_client.connector] SSL verification is False\n      [infoblox_client.connector] Connector initialized with options: {\u0027host\u0027: \u0027192.0.2.1\u0027, \u0027username\u0027: \u0027****\u0027, \u0027password\u0027: \u0027****\u0027, \u0027http_pool_connections\u0027: None, \u0027http_pool_maxsize\u0027: None, \u0027wapi_version\u0027: \u00272.10\u0027, \u0027ssl_verify\u0027: None, \u0027cert\u0027: \u0027****\u0027, \u0027key\u0027: \u0027****\u0027}\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n- testtools:\n    tempest: \"AttributeError: \u0027AttachVolumeMultiAttachTest\u0027 object has no attribute \u0027skip\u0027\"\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nCo-Authored-by: Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0d6cf3e8c184718d779d5ccdd714e103a65cfd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0d6cf3e8c184718d779d5ccdd714e103a65cfd0"}]},"branch":"refs/heads/master"},"5d11acecfaa7cc14affa3697a1d0fe2931417df5":{"kind":"REWORK","_number":18,"created":"2026-02-22 13:57:12.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/73/965873/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/965873/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/965873/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/965873/18"}}},"commit":{"parents":[{"commit":"d179584aeb59a023f065cbf763204525586f56ef","subject":"Merge \"update constraint for oslo.utils to new release 10.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d179584aeb59a023f065cbf763204525586f56ef"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-02-21 08:22:21.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-02-22 13:56:37.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nFixes applied to automated upper-constraints.txt update:\n\nPin versions for packages that are not ready for updating:\n\n- pyasn1,pyasn1_modules\n- gabbi:\n    placement: \"AttributeError: \u0027NoneType\u0027 object has no attribute \u0027api\u0027\"\n- infoblox-client:\n    designate: |-\n      [designate.plugin] Loaded plugin backend:infoblox\n      [infoblox_client.connector] Cloud WAPI version detected: 2.10\n      [infoblox_client.connector] Configuring session\n      [urllib3.util.retry] Converted retries value: 3 -\u003e Retry(total\u003d3, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)\n      [infoblox_client.connector] Authenticating with username and password.\n      [infoblox_client.connector] SSL verification is False\n      [infoblox_client.connector] Connector initialized with options: {\u0027host\u0027: \u0027192.0.2.1\u0027, \u0027username\u0027: \u0027****\u0027, \u0027password\u0027: \u0027****\u0027, \u0027http_pool_connections\u0027: None, \u0027http_pool_maxsize\u0027: None, \u0027wapi_version\u0027: \u00272.10\u0027, \u0027ssl_verify\u0027: None, \u0027cert\u0027: \u0027****\u0027, \u0027key\u0027: \u0027****\u0027}\n- pysnmp-lextudio:\n    ironic: \"AttributeError: module \u0027pysnmp.hlapi\u0027 has no attribute \u0027usmHMACMD5AuthProtocol\u0027\"\n- testtools:\n    tempest: \"AttributeError: \u0027AttachVolumeMultiAttachTest\u0027 object has no attribute \u0027skip\u0027\"\n- pyroute2:\n    nova/os-vif: https://bugs.launchpad.net/os-vif/+bug/2109396\n\nPins only relevant for python_version\u003d\u003d3.10:\n\n- backports.strenum\u003d\u003d\u003d1.3.1\n\nChange-Id: I66bfcc73f1e03b61dca08e2cf3b35a581aaf7568\nCo-Authored-by: Jens Harbott \u003charbott@osism.tech\u003e\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d11acecfaa7cc14affa3697a1d0fe2931417df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d11acecfaa7cc14affa3697a1d0fe2931417df5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
