)]}'
{"id":"openstack%2Frequirements~965952","triplet_id":"openstack%2Frequirements~master~I4f91343247366456ee27db2628b9deec28830031","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4f91343247366456ee27db2628b9deec28830031","subject":"update constraint for pbr to new release 7.0.3","status":"MERGED","created":"2025-11-03 17:22:10.000000000","updated":"2025-11-20 05:50:51.000000000","submitted":"2025-11-20 05:50:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965952-new-release","meta_rev_id":"e3b0b09a946011677f377e18722b8858297f8e9e","_number":965952,"virtual_id_number":965952,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-20 05:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-03 18:37:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-12 23:46:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-12 23:46:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-03 18:37:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-03 19:10:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-04 16:55:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-11-11 00:26:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-11-12 23:46:59.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"9ce07d9badee4c6a0665b0af3c74786e8b894103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-03 17:22:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bedabb07bd64e403fd2901a12256c6311f727621","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-03 18:37:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0bf21d7c46d9719e945180f5b530f0dfb21f5492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 19:10:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5a6d0eecf6f4450a2d368496cd1f4b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70bad31b96d44793900e73a1481665f3 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3573c767d70349759ef54195539d677d : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/332fa819be1948458e46e79c262ee202 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df3b6b3336af4193815f6b9a5b9c4fab : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/966ae145857840cab72c178972249223 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a748966bc8564a2eb6734a1341a7f31f : SUCCESS in 2m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d84f90b6950f4191bee44dd47fe86bc0 : SUCCESS in 4m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/14ea04d4670749af9e7df94fee7a3402 : SUCCESS in 12m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cb47994f1f0c43659afcefeda829d1cc : SUCCESS in 5m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/bfdf369f2a0c475ca5fb92cbdd27d8ae : SUCCESS in 5m 02s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6d070fafd02345d09d5514005b739f6a : SUCCESS in 5m 49s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d029f741a1fb48b4a105e4188f1dfe6d : SUCCESS in 5m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5f3fc357481c4411ba44e252b21b7974 : SUCCESS in 4m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9e00a11e2ca04d1bafa5af2356135c98 : SUCCESS in 7m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/48ef802d499d43e0b40e532d82010e05 : SUCCESS in 3m 12s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a0edb7b0f0ab4be791302c6046a3f6aa : SUCCESS in 3m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ea182663b82c404f8cde111fc17da377 : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6c5b741c67b7455786b130642479f682 : SUCCESS in 8m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/066678d5d153443b8361718aca45f462 : SUCCESS in 7m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/14ecb940ded342f6918be74088677676 : SUCCESS in 8m 55s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/fd092ac837cc4daf9bbb983d2497737c : SUCCESS in 9m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1943b17a2ace4b8e8319290984ea4877 : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5c4e96a849954a96814cb239a582df39 : SUCCESS in 9m 56s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/594a3bcbc4f24cbe818e5d81a324a49a : SUCCESS in 6m 05s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f9b04ef5f2f5425dab17b860d52ef952 : SUCCESS in 4m 37s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d66c3869d7534387b7586cbfe204bbe5 : SUCCESS in 12m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/378164a002514c24b4a29a90ecdfb5ab : SUCCESS in 27m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8fc4f2ec1c604a328a2a3b8776947a30 : SUCCESS in 9m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ef756aa455e5464e92d997ebf37b26c8 : SUCCESS in 12m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ad289b95e3ae4695bbaa8421a642e0cc : SUCCESS in 6m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4391fd5244449a5aa2140ff1d7ff9d8 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cfb9c0cdfa664179955144749a05d383 : FAILURE in 31m 04s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d3609368ec0445528c2b3f711b7ec4e6 : SUCCESS in 5m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a23f706d69c94893b540cd9d8712886b : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b6030ed68824fe6a102242e59304109 : SUCCESS in 6m 52s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/fb5a20ca6cb44fd7a7a2a138471d1d87 : SUCCESS in 13m 30s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7fcfbcf690e246d0a78c4fd8ed8f9f00 : SUCCESS in 5m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9c1e8e5b9d246d7b68c4db9c32972c1 : SUCCESS in 48m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c15c1487d3af4641b572e2cb32e0fd3e : SUCCESS in 1h 42m 41s","accounts_in_message":[],"_revision_number":1},{"id":"462136de4330277dab7c643445b7f9226f0f4627","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-11-04 16:55:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17a36681bb6f86274c69e61fd468d4ec61571a34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 18:39:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/adc8a79205f04f39a1e66c97842c74da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80784d052e5140e0ba259e21229d102f : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25fe502f776046ddacf4fba9121f31f1 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d01972f03cc417db13e87198984bdbd : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96987630ed4d4017bc6e29348be802cf : SUCCESS in 7m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23320129f9ba446d81ba510cbe899387 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f18193706a1c479f99e0bf831d5e3407 : SUCCESS in 4m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c2e31429a65841d3ad5c1afd137424dc : SUCCESS in 5m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f57f4f9370524cb3a3c438ea208d87a6 : SUCCESS in 7m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ec1a45edf95f4a238a29c255ec5b7146 : SUCCESS in 4m 12s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/f6dbde9e2cc44b7a91754067713c046f : SUCCESS in 13m 19s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f94dc45611e54c53a6c3fc3f9ed13c74 : SUCCESS in 5m 03s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f994984f037044a79c29b9893bfe90c1 : SUCCESS in 5m 31s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/98ee0c1088394e2884298e02d016d666 : SUCCESS in 4m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/691daed3d1d847a3a85c7014d5f38580 : SUCCESS in 8m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/af654cdc58fe477b81bed0a66f3e15ef : SUCCESS in 4m 54s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6cf7cf50c3fc4e8aae48e52b966087c0 : SUCCESS in 5m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c6890ddbcf1c4a728ad4c48f2aa7e35f : SUCCESS in 7m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/672eebd26f01469e88ceeedd1d775da0 : SUCCESS in 15m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5fab9eacc67445158130b74e86c6af02 : SUCCESS in 7m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7d9e67654c9544efbc97d8685cb519b6 : SUCCESS in 10m 04s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/72e847db15bf4fe2ba66619a028544d7 : SUCCESS in 12m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4df2be6b279c4f7a8e2f3729f617bbb5 : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/5ebc9091a1e44caaa5f5f6a5ff6c9562 : SUCCESS in 9m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/27a28eea59e74e8789691f2c31c61890 : SUCCESS in 5m 09s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0645758f9bfc4df2af0edd671c1939cd : SUCCESS in 4m 56s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/522796bc7ed145c1b2b0abb76156932c : SUCCESS in 10m 56s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e0c27345702d4bffb4791fbddd20e463 : SUCCESS in 22m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/39d3013e44164726a731dee5e712b06c : SUCCESS in 5m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7e66fa239ac541d6b5091f959967f406 : SUCCESS in 13m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c9b5177235d149999aa7b85a9ef7b43f : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7e98138667914d70b227eba9ac227b82 : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dc9f6248c3f849d2a0a3a03158086976 : FAILURE in 29m 31s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9de5ef42abb848348eeb0e517e4eb158 : SUCCESS in 6m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dbf28688b2f04062830db51d5b67f4ef : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ae72874fe0a436c8cd8fa03f3688393 : SUCCESS in 5m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2a50697353de4b7e9c1213090ffc41c8 : SUCCESS in 19m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/946e4c7527314509900c7d620c3117ae : SUCCESS in 5m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ccd4acf59a84eb1bbe6c6368f02190f : SUCCESS in 1h 00m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/adcca61a52f8467ab194740b9718ed93 : SUCCESS in 1h 39m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b4a096fb56c2b6faa71c8df957b04d51821ac320","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-11 00:26:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24fb984ccfcadbdb7c3627bf1514b0180ba8b372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 01:44:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9db422304bf04925a2b150618d01a080\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d82628ab137f4505b6153aa1e6df1222 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af454bf9a7b84be3ba6e3b792686bb62 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f63e4dac2a75452e981b8ef099588868 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4142c773250a4378b1b5b0c50ae008f8 : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3ab0f4428c24dd680a9d366fc404c63 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85eed2373e404d49a583145498a9035b : SUCCESS in 3m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/25481ea56e8f4bfabf7b6bea40922c37 : SUCCESS in 4m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7f128610b5b647b5ad937d0b416f5d52 : SUCCESS in 7m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3bbe79e57066449e9b800185165a65b1 : SUCCESS in 6m 49s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/016b1eb85e9a423db0bf04f57a4b4175 : SUCCESS in 10m 01s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9ff7333e9c9741b09406b1cb8a44032a : SUCCESS in 4m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/8cc1bbb8ca794e03ae36ec8b3962009a : SUCCESS in 5m 11s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2ea8a35c978849d2b2e2c25bebc41a9a : SUCCESS in 3m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6e3e05a31a9b4f719330e13a235d11cd : SUCCESS in 11m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e0c7ab49b5da44ceb09f72bbe9093133 : SUCCESS in 5m 13s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6e0be562a00b4844a0286429bf9705d6 : SUCCESS in 5m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e2e053034411480f81506419bc3f2c9d : SUCCESS in 4m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/be2988e2bd2a4696a41d94c9bffe39c7 : SUCCESS in 17m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a6b7bf9396247289381f804ed77c235 : SUCCESS in 7m 45s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e430ef6952f441dc854d849a4a8a65d3 : SUCCESS in 8m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4dd4e7a6487a49bc909e90ba6b54fffa : SUCCESS in 12m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0650827ee3b14ed9ac0564dcfab035f0 : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ed7f70b297854347aca323db655069f5 : SUCCESS in 10m 02s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bf6a4ec7d5d34ad08394e837aa245246 : SUCCESS in 4m 51s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/f62b85cf9b444536ac810c8328249f48 : SUCCESS in 4m 59s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7523ad67ec8e40fabcb77f1263dd9659 : SUCCESS in 11m 18s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fc23957331064632a3fee2d4e4a4c726 : SUCCESS in 30m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e56ce1f1e78a41f981d729e1e25b491d : SUCCESS in 7m 16s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/bb03bf4216a1485ba275283ea4315ad9 : SUCCESS in 13m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8c426f5be00a481f986c8c5a510c61c4 : SUCCESS in 3m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8718f9baf6c34145949f4f46321adadd : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6364c8652a894d3f8e79c0d3642c9585 : SUCCESS in 27m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4f19453360a5450c960c9dc99dd18217 : SUCCESS in 6m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fb5334311c32440f97d2f443f9579ae4 : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a50db62bfba40f39d66af32241d27b7 : SUCCESS in 5m 06s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/312706cae12e4289aa631c618bf4fce8 : SUCCESS in 16m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/592a06f0002e42e5af44c0926d911523 : SUCCESS in 6m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/111cbdf0c26246f19c17dc4abffe5051 : SUCCESS in 44m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22c3eaa5582b4e9da8ad404356358d68 : SUCCESS in 1h 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"da8a3acad43e6a8d39a65e9c7d7e77fe43fd19cc","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-11-12 23:46:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b7d48929b8c8d3be24816d713e259ff21a6a53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 23:47:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d59481c3c1483310ba79850300387d2a72eca0fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 02:05:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4987bc131a534bc4906d615019196959\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc2ddc7c0cba41cb9c59bc410c2b5e36 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae0b5503e854711954f4b3552b0b2a6 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca098734dc504dd39a6c8c52acc4d0c2 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/490597680c5e433eab6ea5c44f8e82b3 : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52ebfd24fd654019a3274961bb928dcd : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a4ca0f18d744c18b8f68b5ee0c9abac : SUCCESS in 2m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/138f2f1f27a24f6da53dd204535bbcc2 : SUCCESS in 14m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3f55274cca5a43b8a50c3a315cc440ed : SUCCESS in 6m 29s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f36923d372df459fbb5fecce8bf1742c : SUCCESS in 5m 30s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6f556a44709d4cf090fcec1eb52c06b1 : SUCCESS in 8m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9d17a195bf034f83a32767b51742d2b4 : SUCCESS in 6m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2be3f0117d704fcf9a59478dc63287f6 : SUCCESS in 4m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/529aa996e1cf41d49e008781f2c9bcee : SUCCESS in 7m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4ef74c0bd82b4f3f959888f0ed9e6210 : SUCCESS in 15m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3481ede2738f488f913f25b6a64107a2 : SUCCESS in 6m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b10d9d3824e847c190cdda929b88f7fb : SUCCESS in 8m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e31efd65acc440e992b63f3fc85c3f75 : SUCCESS in 8m 18s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/09e07716376d4009b9d4f75ccc1ffc6a : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/288d2f019fa64a829abd7d206440f238 : RETRY_LIMIT in 1m 20s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2560920d00284dddb593c14c1316c83a : SUCCESS in 18m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/730a8e014d1045b3ba500debed839880 : RETRY_LIMIT in 1m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fb38bb5268cc4abb9440802fd8cd6f8d : SUCCESS in 11m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/58cd4fd3b5394a689ec77c928d524fd3 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3177964ed09c4d179e1f509c7eb8bc49 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/128a24690a6142a9af72b10e954cafe2 : SUCCESS in 25m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/15e16a21daf94b6b886ab81373c58168 : SUCCESS in 6m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b6941cb16d1143f2a971d2df8cd7ef92 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72cae62050f744c99308156f62736afa : SUCCESS in 6m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e5898f2001a9471b996041a8f6ad04ea : SUCCESS in 18m 05s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d51027f4eeb8440dbf2fcdfb0a786d02 : RETRY_LIMIT in 1m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c09bf31caf44136b91afc62798c3f89 : SUCCESS in 1h 05m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7765cd5b27f4403b81f8aab3c58efc3 : SUCCESS in 2h 08m 36s","accounts_in_message":[],"_revision_number":1},{"id":"48d70983a1dc0ee40054924659328be7d004b3a8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:53:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"366b82dd449b14b6c10b8b9c69bf73047bae1548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 13:27:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f7664b62d3a4c20b72725655963f4b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a65023009794bcf81c54b2fd64d4f47 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408bd91aae534946a29459e917c6bab4 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93c36058b28d439098125e43a433bcb4 : SUCCESS in 2m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d42fbe848c949a788aa6ea074c8f055 : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0bacbde3f99491cad3a5ebc8831ff81 : SUCCESS in 2m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9acf9e3470f14f989021e937fd27b009 : SUCCESS in 2m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a865cf9696114a799b407f7a28da51e8 : SUCCESS in 2m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/adc11250c0684891b70cd5b8413b11d9 : SUCCESS in 10m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2c3a517830b746bea8c671eaa9973090 : SUCCESS in 3m 42s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/84532d775e694dc19edf9f3208fea47b : SUCCESS in 4m 51s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4882d61510974773adb2b1c77a8460ec : SUCCESS in 3m 01s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c11b49d7b04b4fb9b8f6505da0e50f5e : SUCCESS in 5m 17s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3f0c88fae4874b699c687757dcc9e8e0 : SUCCESS in 4m 20s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1cf95d27dd97448398123925b4e55055 : SUCCESS in 10m 45s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/df718950df67402789456a0b505b695b : SUCCESS in 5m 23s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6115ebf2bad243ec83bbc9f4e3ee7097 : SUCCESS in 4m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f6b46fcfb68146b7bf91f263df1c6530 : SUCCESS in 4m 54s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/00280dbfb19047948850f18607dcb337 : SUCCESS in 15m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c20223d316c44fe38f29871f4d2b8e5e : RETRY_LIMIT in 1m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/01a39f47e00646fbaaf3ab001729c81f : SUCCESS in 10m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/34b42bec9cc547d69786b2395e6489bd : SUCCESS in 11m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2cb6dbb2db7a49889d93503ca8420e80 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ea3119f6a2bc4c74b51ac5ff09e87640 : SUCCESS in 6m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d94a985085564ecba2195ea98e22af49 : SUCCESS in 4m 43s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/32e3922c96f0495b8ee023849b07698c : SUCCESS in 3m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2f2da3c5d1b944fe9e3a2ba168a6f630 : SUCCESS in 9m 11s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f3ec09de926247ab9ff38765bd09a548 : SUCCESS in 33m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/200e1286a0b547b4bca5aa714c1df49c : SUCCESS in 5m 02s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/4561d701561f422d9f8b8c83abd115c4 : SUCCESS in 13m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e571f978403946c2ac0ab4e51023ea5e : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/12cc04dc763f44c6af193734126fa933 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6352122d5a4341eab6f927bbacb29d97 : SUCCESS in 28m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e45cff2c04ea45afb76db2e13d1d4a41 : SUCCESS in 6m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/330e1f14749b44cfb10012655cbecffe : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8e88a509b254d96931c2d425d7d1925 : SUCCESS in 3m 52s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f9608a240fd5469a83ba6fc3e7c66ba5 : SUCCESS in 18m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b5b3841ed88641079d4f88d0b72fdf80 : SUCCESS in 7m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14fe071212784f8ea42f285df3438b09 : SUCCESS in 51m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/215590bd2e0944c190f1e69b744c0a25 : SUCCESS in 1h 32m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1fdd942e9d86fde911b59b5cae4d57413255d18d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 13:28:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a01fc95967eb582f70281d741c2be3b90636d93f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 15:18:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d6d014af96645dea678b142ad6e4816\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a3eceedca0436a9ba4fb12952fce6e : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2595dc3dd5d4fe4bb9f4dcd4739a6b7 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abaf5bc502414fd8a3a588807dd0afbb : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87a9042a5f12464891d0ea01e6018c1a : SUCCESS in 7m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c3f64facc5e47a0862b460a1f8c3f76 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/891733a091b446d48d0e79bc47655466 : SUCCESS in 4m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4334693da31745f09939664368310c10 : SUCCESS in 7m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2e30710cf9294856b8fda02f5b2a504f : SUCCESS in 6m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/03c8d3f9f0a14cb6bf3b59fa3000752f : SUCCESS in 6m 58s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e0740fc1e4f64570bddfda5cb8f4e1ea : SUCCESS in 5m 18s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/82ed52cb34c949abae4274ec200654c3 : SUCCESS in 5m 26s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/10aa77fe44dc42b398f618fdc0e70be2 : SUCCESS in 4m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/362b7fb6e2b84d0b84197a0fa94bcf48 : SUCCESS in 6m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9c59a1e6942f46d989016c2c6ab125c5 : SUCCESS in 11m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/78290f1e4a344ae29dc9cb1300f25b73 : SUCCESS in 6m 51s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/803e51d1b53c406c99a216163c5fce3c : SUCCESS in 7m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b1bd352e7cc64892bc0482f6f26a7dda : SUCCESS in 4m 51s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b66fc2539fcb40e7b8b3677dc4bba661 : SUCCESS in 17m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7377f9fb860b498497481b01912d2655 : RETRY_LIMIT in 2m 01s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/77819e4bcad449199c0d0d54b388e2a9 : SUCCESS in 9m 15s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b586af8f4603470c8f576ac5554b7c26 : SUCCESS in 12m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dff31efa67f841d5930ce7c6964b93f9 : SUCCESS in 3m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/60e9b5c64e91461e8813a2824fed77b8 : SUCCESS in 9m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3f8b4281e2e7407ebfc7a7945c0a80c8 : SUCCESS in 5m 08s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/81568c59a5a4408e97f8f370c1158a08 : SUCCESS in 5m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/08a75d99c81241d899e56bc0c3a057f5 : SUCCESS in 8m 31s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7964c81a271b4bb28182565870af8a67 : SUCCESS in 30m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e1d81a317a4c42b280a215667b19d3f6 : SUCCESS in 7m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9966f6fb93534ba58bc8c1b0dc3ff7cd : SUCCESS in 14m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/097de6c5832745fb9d428ed922c5e647 : SUCCESS in 3m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b91d7aa940f443d8e705f3c634b59ff : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd6e65b62b1f475cab8a44394ef9510d : SUCCESS in 27m 27s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2fb0551851914445a6cb13e37b5c5faa : SUCCESS in 8m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/bc875f7500b64eb2a57fae827b02b817 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d9e20f185334a78bc22af58f242928e : SUCCESS in 7m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/acbb2b2dc4fe4a308c3522c92ccab6fd : SUCCESS in 8m 10s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4864886e003f411a8ec3dc4608d55c74 : SUCCESS in 6m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/515a689c3ee649f68d07939f726903c5 : SUCCESS in 33m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e549fef937ce488ea431e04024b4d642 : SUCCESS in 1h 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"25e9ce6f6d47290e00a55a220c372a8e1708ac44","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-11-20 02:23:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f983e36045b781df80e0563f965c8073fddb73f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 03:57:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d91a0e0974f40b2830212206450ce2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e8022bf9a5466c97108e6accccc086 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37e5816c7cf84089a656149527c34122 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/817f8c91ad2847b18fde88b78bd3d252 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e62425684dc14168a5679ddddcf85408 : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2171ab4a30a479db6c1b959b477f680 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb002550513c4b5cb995097ea10e7190 : SUCCESS in 3m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/950349f0bf7f447b833926238951ee7f : SUCCESS in 6m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9524b6a0063b4b4f8330ea092addbffb : SUCCESS in 9m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/edc20a9ca67444789ef36bd916244657 : SUCCESS in 7m 37s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d8876f6a2db24745addd5bf04bbd4fa7 : SUCCESS in 7m 27s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/13b6832197894955bb1caf0b3472ae67 : SUCCESS in 5m 05s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c9e3f53bfb7742c0a6566daa588ecedd : SUCCESS in 4m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/29331d4a62ed4a70bec35493e6caf76f : SUCCESS in 6m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9c40eaef933e491bbb2881ad3e46b32e : SUCCESS in 10m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1f859f1a62a640f29c5bac21b84595cf : SUCCESS in 5m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bf3bd08d62a14b04b7fe13185d7a94d9 : SUCCESS in 6m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3cdf0d477c5e4555a35db9c1087b4555 : SUCCESS in 6m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/23aa4e8784874d198e3d831ccd4df8f7 : SUCCESS in 17m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e6c0cedfee5b4eed91e1097f3a4ccb18 : SUCCESS in 10m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/14509c717069412ba9b5fb0a8af29da4 : SUCCESS in 10m 38s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/fa3b5a8883454065a45b71b7bf7e5a1c : SUCCESS in 11m 55s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2ff336c7d0614e1aaf4f6eb1feb5ab14 : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/78e9d4c8426541febb5ba74331050da1 : SUCCESS in 10m 13s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8d98dc27f7da4e73b669e543e47eadf8 : SUCCESS in 6m 05s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d53ec1e2444541ffbbbb96eff3d040be : SUCCESS in 7m 33s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2b5a73e8d9e34e7d9c2d0809940bff43 : SUCCESS in 10m 10s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7c2e23d6bb1e4c3b93b52763c64ac0ec : SUCCESS in 27m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1f30ae51a4c8481191721272c405fc89 : SUCCESS in 8m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fdc7845bc04c44d1a46c4f0b95e4645f : SUCCESS in 12m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/81140b2452f14c9aa734d1f366bbf005 : SUCCESS in 6m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/953ea355317c48c4adb2ad8e3ccd8e3f : SUCCESS in 6m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33e3c054bf464ad8aa6ba8ac0b825bf8 : SUCCESS in 29m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e8502cd470f242d98e99d81beba6ce94 : SUCCESS in 7m 50s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0fb691ce406f4e82aca4869d72a7c447 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9491ced95ea74480b864467d37033558 : SUCCESS in 5m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/bf97cd1676584a25969878a297009d4a : SUCCESS in 9m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/660d3dfb8a3345ab8bd786d5759faf58 : SUCCESS in 6m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e5f844f69b742119c95255cc84ad558 : SUCCESS in 52m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e5951c3859a43709fa5a450edbfad77 : SUCCESS in 1h 29m 31s","accounts_in_message":[],"_revision_number":1},{"id":"fb66e194b03d7bc70070052b6644c0b4703d345c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 03:57:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"425afc264058fbaf229b92157b64dd0af559f69f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 05:50:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ad15b216524e37981453ca99fc5bbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d004d12bdce49ffbf4e004cbd0ea677 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e449259978e54dcda6bf61a8e198d334 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01b7bd64cda041599240936824751e5a : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59cc698a26c74d69a7750568e0b36af5 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/952e89fb60ed4607a8e02ae6951d7b9d : SUCCESS in 2m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b0bfc54f4f1401ca31b10fa1ab6c363 : SUCCESS in 3m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3c4f7418def04521963f62ee83adb91a : SUCCESS in 4m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ebb59b8dff7e4aa0bfe0f85e169c5e2a : SUCCESS in 10m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b013a982c6f24b1c9abc9ac3e1963ebb : SUCCESS in 4m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e0f48ad73bcb47b8a00480f1a0ff129b : SUCCESS in 11m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5aa9bb41e8cb41ca8fc7fcc2a599193d : SUCCESS in 5m 14s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/486247cb93ce4741a6376e062aa7ee72 : SUCCESS in 4m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/69553a7ac212499c9c6566e785a4720b : SUCCESS in 6m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a39751bc1c494d8e826b8af7e97ff9df : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9255b0d01552465d84269e5df35f6686 : SUCCESS in 8m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0ac4a00df2704a90b079c9bb7094222f : SUCCESS in 9m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cfc6cb91d4f8433eb9380a92f75789d5 : SUCCESS in 14m 20s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/175396e0afd94657879c75da56c16644 : SUCCESS in 2m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/53a55b7606234d2fb70124a8d46b32df : SUCCESS in 10m 26s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c54c79b9399b44789f29eef92064ab58 : SUCCESS in 18m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/57b8abda6de84e1cb3706239a592835a : SUCCESS in 7m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b0a3cd2833884eb981b38fef1eabbd16 : SUCCESS in 16m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/fffc06548611438eb0926a5a1be8409c : SUCCESS in 3m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/40a3252da49d4e39968ce3c15b32efcd : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4e45ad946cc3492fb8c16ce75f78a3b9 : SUCCESS in 20m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d56408865e9c49ae9d837ec4b630503d : SUCCESS in 4m 53s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/78687da14c474fbb84923c166c4847cc : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a4c97363e1d94b69b792a6c4dc6d5646 : SUCCESS in 4m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/60d0346e9bc04f15b69473e4c0c2f3f1 : SUCCESS in 8m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/713e488b53254cd49b462fc4a6995c33 : SUCCESS in 6m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e6a20f530dc4999b0cd02f12a0b5858 : SUCCESS in 47m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44a2ce778d5045e88c855170e558b005 : SUCCESS in 1h 48m 06s","accounts_in_message":[],"_revision_number":1},{"id":"02f04cd7a96da2be0a543a157fc3d5ac996c8b8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 05:50:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e3b0b09a946011677f377e18722b8858297f8e9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 05:50:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17f34b2bc0794eec8a5d874390bbaa12\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa1e5fe95664ae7ae7475ad55ece187 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3dcf7a433f367f3ed8af9a655823c80efe69ae3e","revisions":{"3dcf7a433f367f3ed8af9a655823c80efe69ae3e":{"kind":"REWORK","_number":1,"created":"2025-11-03 17:22:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/965952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/965952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/965952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/965952/1"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-03 17:12:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-03 17:12:09.000000000","tz":0},"subject":"update constraint for pbr to new release 7.0.3","message":"update constraint for pbr to new release 7.0.3\n\nmeta: version: 7.0.3\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Commit: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Change-Id: I80d65f9f4c82ae66a585e5f13ea004bc50d513bb\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Code-Review+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta: release:Code-Review+1: Clark Boylan \u003ccboylan@sapwetik.org\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I4f91343247366456ee27db2628b9deec28830031\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3dcf7a433f367f3ed8af9a655823c80efe69ae3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3dcf7a433f367f3ed8af9a655823c80efe69ae3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
