)]}'
{"id":"openstack%2Frequirements~967031","triplet_id":"openstack%2Frequirements~master~Ie5136504837b55c6b69f40da0932f83dec57eead","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-11-20 09:48:25.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"Ie5136504837b55c6b69f40da0932f83dec57eead","subject":"update constraint for mistral-lib to new release 3.5.0","status":"MERGED","created":"2025-11-13 13:56:01.000000000","updated":"2025-11-20 12:58:41.000000000","submitted":"2025-11-20 12:57:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967031-new-release","meta_rev_id":"ee1a950a05800a425f4498e8e42316f6e68aaa52","_number":967031,"virtual_id_number":967031,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-20 12:57:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-20 02:25:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-20 02:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2025-11-20 09:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-20 02:25:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 15:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 11:53:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:13:28.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-11-20 02:25:47.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"31856cc4a69306022be288f1516be0fe8817f1b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-13 13:56:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b97931a97ce9466de351a96e01cb7957971c40b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:11:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6b72712a0c44ebd9217201fc9fbf5f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009e6d306423494f92f7f3028e97823c : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d690c5be9c4ea8b8d5ab06501f6bdf : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6273ae709e354aae93ae97feca6a80f2 : SUCCESS in 2m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2330c538d9e74dd4bb737a6b4f44bca0 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8941dda74cbd4007a83c21cd78cb8134 : SUCCESS in 2m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2fb8b48dc51b441da4c0ab92d642ee3c : SUCCESS in 2m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4506aa868d99438383d8f6cc29e4751b : SUCCESS in 5m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7485a89b5fd84638ad0fc9c5c97dbad0 : SUCCESS in 10m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4a2f4f2a582b4508a07b92b8d44ea3dd : SUCCESS in 4m 24s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/6f397ae3bc784a0dba4a574fef0ad62b : SUCCESS in 5m 31s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/487ca722aa474c578c55d828c3363efc : SUCCESS in 3m 12s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0987cca473234b9a91d5efde4a729eb7 : SUCCESS in 7m 26s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a6c7bc46da1f4425976884f9d8c1f8a3 : SUCCESS in 4m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a4131a39a2c04696899c89501d17eab2 : SUCCESS in 12m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ccf0ec441a5144d09f5c50b1b8496082 : SUCCESS in 3m 50s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/274f5c39fc8a4bf3ac2dc7bfd75af5a8 : SUCCESS in 6m 11s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e5b94d10a2724164a52e03d5b0ab64ce : SUCCESS in 4m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e21e0376e4624547b51e28920016cdb3 : SUCCESS in 19m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/59bc9d12903f4153a45e5101551b73d6 : SUCCESS in 9m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/39c5f56d00064e018f817b23e8785304 : SUCCESS in 10m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/572aa4e283e24de1bf761664ee63a566 : SUCCESS in 12m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9803000699a04f708046d004fef534d4 : SUCCESS in 3m 39s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/eeefebc3533d4dd09818c2b952b541b4 : SUCCESS in 6m 35s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f91f572be29f4e818bd0e735c97af317 : SUCCESS in 6m 19s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/6235d4e9c1f44057a4dfed89d92dd8cb : SUCCESS in 5m 58s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0d1f6d42a4cf4701a724342ed23031a1 : SUCCESS in 11m 15s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4fe213e2947548849fbeec63331185f6 : SUCCESS in 28m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/35eff3a634b1488b9a5c8449e03dd3e1 : SUCCESS in 7m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3e9069953fc749d393180850bd370158 : SUCCESS in 14m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/bc3aabed590f4b0d92c37b61a06ce807 : SUCCESS in 4m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e26c8cd4992e40549c445c98842b100c : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/393d0670dd2549f9ad594b052c4be863 : SUCCESS in 28m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cc3855b21f704a288308151efa6402df : SUCCESS in 7m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fffd6532952644bd9ce7aee707679417 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/084760ed1a24425aae664a1589e79d68 : SUCCESS in 5m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/59e23c039f954f7581cf36e34e8d55a0 : SUCCESS in 18m 28s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0e8d2978cca34f48bd5e34e2b96a3d96 : SUCCESS in 4m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31f389054b924585b2b1646ff17d3012 : SUCCESS in 1h 08m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38acdbb8eade487d851326c9fc9cd93c : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8d772c78ab8e6347f37b7fa61fe5b72575cf4c7c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:53:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e0ead62e2a52fa71d48d7f51536e78d16f7c70f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:13:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81a2fb9c55b825cacdebc0e1ce69019b6b1d7793","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:14:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c461d394c812700d61ce05e395561507eb2350f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:24:46.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"b487c623a06bf519f0fc56e6bff6952e51bb6f45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-11-20 02:25:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"7d39ed854c0ee671b3f150e1351c47131dd3eb57","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-11-20 02:25:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7f2c3b97d3391bf8d1a8d01dd9dcc6c28403a0a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 04:16:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bbf4971b72d4c17a43e9c5ab4c28b9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/456092039fad43289f5d2606128593c7 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6d7c560a87747c48fbf7a5c4b70e650 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae442a6e3fd648a7987505f855c6ae5e : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/41fa1a0633ab45a49008a7bc786242c1 : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6647c8d2ab7d4a1b825b9f032c16947a : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5954b9600986464990193c6f963a4bda : SUCCESS in 3m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5470b6aad4134755a2518cf25daa19b0 : SUCCESS in 5m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4293dee60de749c7a679c8004ecbb14a : SUCCESS in 4m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c576b7bed96540d283234e22015e79c9 : SUCCESS in 12m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0f7ed907c4b648a396ce208abb7ff768 : SUCCESS in 5m 49s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/868265c14fb74a0682a8e0cc624df4f3 : SUCCESS in 4m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/71390bda66d34f8f8fdc19542fe466d4 : SUCCESS in 5m 29s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7d8a8b30a22343c8a72875c4e9c2ae1e : SUCCESS in 4m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4abb6508f35c4a288bb139ee294c87ae : SUCCESS in 11m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7d6f984b5a26417685eddebc005b529d : SUCCESS in 4m 32s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c24ea1fddcc14a8b9ba7a07fa14d62be : SUCCESS in 4m 06s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/53a969120cf74be2aa6f85dff23490b1 : SUCCESS in 5m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/05f244e4619f4c11b4122bf5f38f273c : SUCCESS in 19m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d6fe19b339a4245b52ac1a065671af8 : SUCCESS in 9m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8703137ddc244db28f70d869b1367180 : SUCCESS in 8m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9d826c79090842c395ea477c0140f171 : SUCCESS in 13m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dafcacac683b4766a9315a79a64c14bb : SUCCESS in 2m 26s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d0396825e87a43af8c2a0a80564b353d : SUCCESS in 12m 02s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0acc3e40dfc8424aa93b5713455b10b4 : SUCCESS in 6m 37s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9a32201c38c448d9973476a607013568 : SUCCESS in 5m 26s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0489d84ec486461183af7f711d6dba49 : SUCCESS in 11m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/87310737b9ee407a8d55368c067116fd : SUCCESS in 23m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d92bf1f0f58443879f65e23ac9b646d7 : SUCCESS in 7m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e54a20660e32457099672bda0b83ca77 : SUCCESS in 14m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2eb093bf83374e058cfb5d2fd28d6e90 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1059eb2def9540269a687ade7e093677 : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7dd195bc29b945efad41711aba31f53f : SUCCESS in 27m 39s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/adb4abe546104a818f6a0f1b63edce84 : SUCCESS in 7m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5ca789dce97f47a2bdf18c7d7db6bc68 : FAILURE in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e77703824164f42ad4eb1817eb7166a : SUCCESS in 7m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1e7f9036ac7b4804a99497a052735257 : SUCCESS in 21m 17s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/aa6a25709ade46059a90fc949bdc83a2 : SUCCESS in 4m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9c13b01ccfe4072a91551cbe1b78ae1 : SUCCESS in 58m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b612c654662c403492a67688963d6732 : SUCCESS in 1h 46m 19s","accounts_in_message":[],"_revision_number":2},{"id":"bc0c1896d77b51810d513c4e8e1a209a38775e47","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-20 09:48:25.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0dccd01d82adcbe53c636f35acca6d8d7dc67d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 11:44:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c2ed920f878463683574abef5bea0d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba6bcb35147247e3857b24e4e98205d9 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4093cdfb55b1454c95181faf76869181 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/516031ad48724e62af1eb9d77acd4c8b : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/907944b9dd25461e809af19437c87053 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1abf078826b44a65aa73fb3baa7b664f : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/11dc53d1f65c4a0f914a7b80a65675e8 : SUCCESS in 2m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e2822666c2f2437fac530cec08479aa2 : SUCCESS in 4m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d6cbd3bf341a425eab504f4f4548087c : SUCCESS in 9m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/761abb46e25445feb54ddcc9fe0dbf64 : SUCCESS in 7m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/67fb3f2107744eba8e2bea9e6506a6c9 : SUCCESS in 6m 02s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9c5523153a74462aae475f84980dad96 : SUCCESS in 6m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1bb681feaf3e4de5939636cc2671b8ca : SUCCESS in 5m 55s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/75bc10c77cda4b97b6c999b6745e95c5 : SUCCESS in 4m 21s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/90580facbea84014965ddc33fc766535 : SUCCESS in 7m 41s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f66861deab8449d384606a87650fa418 : SUCCESS in 5m 21s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c3157716966c41ab80f043e6e3d40f1e : SUCCESS in 6m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ed1335f222c346e49723ad7ef0818505 : SUCCESS in 4m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cc46c34d19c84c1e81c503b4de874c80 : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20772840c2b1482da6f4c2db58c2be30 : SUCCESS in 7m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/05892a30e0f1493ea610098f4705b2eb : SUCCESS in 9m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/888cff4d6d1a4b78b13356ff865c799d : SUCCESS in 10m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ce29ccbc26a342af8476cc61f7435125 : SUCCESS in 3m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d56f76c4472c4e74a091a5c8de3695a0 : SUCCESS in 8m 45s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c113f4b46a184883afa47f4117f2555a : SUCCESS in 4m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c77d1e586aca457794d51e0c9f1c113b : SUCCESS in 3m 34s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5ebddbce442c452a9fc9f62918fc9374 : SUCCESS in 9m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f1b966bd3a984111b80f8422f37e6fd4 : SUCCESS in 27m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c5549d84ec0446a09ae35dcb2a42f80b : SUCCESS in 5m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/23b5ee646f14459ea0e3b028fa65d818 : SUCCESS in 10m 16s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c1fd8b369cee45848bd4679a78933c31 : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c9d9c8a12f4244debdc44d0b3143bd67 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd758253e08b4234b497b0b04816452f : SUCCESS in 28m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cc858a3833194ee5b0c5dc388bdf09a0 : SUCCESS in 6m 58s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9d5a49dbf3884c2d9624270c29754b03 : SUCCESS in 3m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a0a06def9b164c43862d7a638d490ca3 : SUCCESS in 7m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/84fd9b35292e48e1ac72e346ff1298d0 : SUCCESS in 20m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8110e567cc274c1f832dbba5ac9cc854 : SUCCESS in 7m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd27033c44654005b346f2ff1e68fd9b : SUCCESS in 1h 07m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/112558efba5847ac93bb634101b6af9d : SUCCESS in 1h 53m 09s","accounts_in_message":[],"_revision_number":2},{"id":"dea37462c576df0c992b5fe1f9fc68837629e6e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 11:44:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f3fc474081dc608e978b297cd25cfb387dd1626e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 12:57:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d3af13fe6064f0f9acc9dc67aa895d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2245093e3295471a8032282715f1b821 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a722a77be11044e5a25ad08034872e8a : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/047f813a30674968ae65b5b3a1bea951 : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/74c5800657a34658b0a3278997f5a7f2 : SUCCESS in 7m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7be4ea2065d44ca8a4168ef952f676e2 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22974fcdf5d74b27b69115c30488401d : SUCCESS in 2m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/650777bcaa3e47988e2ffeca3b08c465 : SUCCESS in 5m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c1103338b1d4423eb7f258226d9ad131 : SUCCESS in 6m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b4350d06612441b097178ed99f715534 : SUCCESS in 7m 34s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bd68f38cc6ce490eaa22c7042098455e : SUCCESS in 11m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/41f5bde3f8e14081b492da494de37b4a : SUCCESS in 7m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5b6d6fe7e1b4415da9e951bc335f4a74 : SUCCESS in 5m 41s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5d69c72b1d034307afc1e85bb294636d : SUCCESS in 8m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0e9dd205fa174bea9c31b736eb72a26e : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dcef495ae8e84f4aae19a0b631f6839a : SUCCESS in 9m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a0b99852476d44d1bdd37c1a84dc24fc : SUCCESS in 9m 38s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/a7f7718972654942a6a087fccb77e3f1 : SUCCESS in 10m 24s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c4a862a706774dc782d59054aa7f9834 : SUCCESS in 1m 59s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3992ac11e442437785bb61d106144d6f : SUCCESS in 7m 52s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9791830c5d0a4b62ab3ee27b26d90e78 : SUCCESS in 23m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bbb80952110f442b9d03b080f823063f : SUCCESS in 7m 47s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/410244709702411d8d7c4414b0f88242 : SUCCESS in 11m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f0ad11f5f8da44edb2d074c35f86181e : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e4e07c77fc264525bd5c42b0e56811df : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a5e1ab0743e846c88ff08aa7ae76e7af : SUCCESS in 28m 48s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1509dc156dae4d7b9785f05a09d3c3c4 : SUCCESS in 7m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e09f8f2cacfe4a84b932e137eaf9e6e1 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/531a0ee0331c46c8bd7ba3ff00952e11 : SUCCESS in 3m 23s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f5326617b8af447480ddaf5dcc89f628 : SUCCESS in 7m 05s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9a3a016084204c12acf3f0b97a7f6b6f : SUCCESS in 4m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21ee64fb47b748bdbff7741d1f8509f2 : SUCCESS in 32m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d543045670b44cbaceb00e031e2a27d : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":2},{"id":"cf59c7b4c417a0839b6798b5d0393f2e30a86a8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 12:57:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ee1a950a05800a425f4498e8e42316f6e68aaa52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 12:58:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d587345a90074fcebc8cd10b89a280b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8601637d2a7b4390948891846237b5c3 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"889f891a65f880ed721ea402873e2a5977333943","revisions":{"f8be6638205983f4cd50d6e3f5c9e824c6e3b372":{"kind":"REWORK","_number":1,"created":"2025-11-13 13:56:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/967031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/967031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/967031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/967031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/967031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/967031/1"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:45:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:45:59.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 3.5.0","message":"update constraint for mistral-lib to new release 3.5.0\n\nmeta: version: 3.5.0\nmeta: team: mistral\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I61468bdcdb9c9030c79f2650ddb9779e16291cb0\nmeta: release:Code-Review+1: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie5136504837b55c6b69f40da0932f83dec57eead\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8be6638205983f4cd50d6e3f5c9e824c6e3b372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8be6638205983f4cd50d6e3f5c9e824c6e3b372"}]},"branch":"refs/heads/master"},"889f891a65f880ed721ea402873e2a5977333943":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-20 02:25:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/31/967031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/967031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/967031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/967031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/967031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/967031/2"}}},"commit":{"parents":[{"commit":"c1807ef0abeb582c41485dd5f4f825106aa58060","subject":"update constraint for ceilometermiddleware to new release 3.9.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1807ef0abeb582c41485dd5f4f825106aa58060"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:45:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-20 02:25:37.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 3.5.0","message":"update constraint for mistral-lib to new release 3.5.0\n\nmeta: version: 3.5.0\nmeta: team: mistral\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I61468bdcdb9c9030c79f2650ddb9779e16291cb0\nmeta: release:Code-Review+1: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie5136504837b55c6b69f40da0932f83dec57eead\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/889f891a65f880ed721ea402873e2a5977333943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/889f891a65f880ed721ea402873e2a5977333943"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
