)]}'
{"id":"openstack%2Frequirements~967032","triplet_id":"openstack%2Frequirements~master~I2b42b690d5083169ebb063252e184affa07e991a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2b42b690d5083169ebb063252e184affa07e991a","subject":"update constraint for python-mistralclient to new release 6.1.0","status":"MERGED","created":"2025-11-13 13:56:04.000000000","updated":"2025-11-18 22:25:31.000000000","submitted":"2025-11-18 22:24:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967032-new-release","meta_rev_id":"5fedbc90f00ebb85feb554ddd77cf368ec20bef7","_number":967032,"virtual_id_number":967032,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-18 22:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-18 11:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-18 20:11:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-18 20:11:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 15:55:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 11:52:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:11:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"b26d3dc47bf8a89b1a130c7ab3ca698f6c85f7bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-13 13:56:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a92508999e8405062f60aaa7ad7166be4e11d50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:55:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/414af147e89045b5962272112f174734\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa962ceca4042b4a6c6d5492819eff2 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b69ba3e3656438594b50b1c7cfc1c02 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9bf0b8d1d3c48678bb658a1faf178fc : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43e42177e0e24597af9aea2b9b357f04 : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9520a6eb3e8450ba506e6a43e0139bd : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/360acc33fdfd41c6ac9c85454db7de92 : SUCCESS in 3m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aa954ec33ffb419e8885d261771890bd : SUCCESS in 5m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c97a24c40c154ef48a3c93e5c001988b : SUCCESS in 7m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c7979c1b0243458d9d6d064ea37fb3d8 : SUCCESS in 7m 42s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a78f7427c59d477686cb7affe4842483 : SUCCESS in 12m 42s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/37cb97f8c8c44d4483548dbbea3ff9c8 : SUCCESS in 6m 27s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f4643ff6fbd948d89629b7b10558fcb6 : SUCCESS in 6m 04s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e1c3a59f204c443894e71f92a26a6616 : SUCCESS in 5m 00s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7cf44f5577a54890973e76ae2d36f4e4 : SUCCESS in 11m 44s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/69b8cbfb56954915b2e0398d82695ce8 : SUCCESS in 4m 20s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c234bcfdea5741c992d9a76fa9a46b34 : SUCCESS in 5m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6e037c25c6fa4acbb0fa8151e6c33344 : SUCCESS in 5m 01s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e16c011cf8aa49089613c2272c01394d : SUCCESS in 17m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/814afc9425c54d37be8a0859cf9805dd : SUCCESS in 8m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4bc304b7fca2420f8dc7cc76867c7416 : SUCCESS in 8m 07s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6e40df13f37545428091a5dff6e466aa : SUCCESS in 13m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7347e514ccdc4249a815994d1af5747e : SUCCESS in 2m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f097a6dd19a5448d82da764db3d30a83 : SUCCESS in 10m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/92c39fdfd1c64bfda24d22f5aff5b510 : SUCCESS in 4m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/74b3074374f7476cb7b15b1e9c8b5fd1 : SUCCESS in 5m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/56a701269e1d42e9ba8056d8743a62d7 : SUCCESS in 11m 23s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/50805ad1f8d34af7a089bd9a792a4d5b : SUCCESS in 21m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d38307c135e24867b158fbecaf265c1b : SUCCESS in 8m 30s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/de7460c53e0044a591ed05384adc9b58 : SUCCESS in 15m 18s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6e0233b610a541e48a50428631c47679 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e99cee99494e45308b60a5b89ade1ea8 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a06290d02783444cb4de071fe463190b : SUCCESS in 24m 59s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/19c1c5fe3d9f4ef4898e5ce6e7c566a1 : SUCCESS in 6m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7ff0d743a8ed4b269e366caf974ee82a : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b140b59cbac54324b9c633ee848028f3 : SUCCESS in 4m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7da948d2c3924450b0966b918a96f536 : SUCCESS in 20m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a1e827545c36494c820146818f42d415 : SUCCESS in 6m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f7b2c33aeb745b3b755b087e20f228a : SUCCESS in 1h 03m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc82f6b3817d4edca34e1c78b296c75a : SUCCESS in 1h 52m 53s","accounts_in_message":[],"_revision_number":1},{"id":"755efbe47d138430c0403f048e11a6850506d258","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:52:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a6c06f49e1b44bdd2be5625b7a4de829f39e0582","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:11:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ccdad53c9709327ff31d9f914955add3d4eea47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:11:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ccd1c8c90a90a107b275e8520b8603ec2275f0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:24:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01011613ff074645b936d59e54ebca13\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb1a3e2c63f44cfa8fed1cfa75354d5 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d263cbeb4b84b11960c384690b10702 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c82dbd2cf2e04078930ca28fd1a551ef : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c25356771f94a90b08a4676412c7c93 : SUCCESS in 7m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/106fbe82d0b346bdb3efe5b33d092dbd : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1be708884f13496c8250afcd97e027dd : SUCCESS in 2m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/22d3a690d6074fcf848f6b6026949e1a : SUCCESS in 5m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4847a05b6bec4139a6fb53994a0481fe : SUCCESS in 6m 43s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6c12fca2dddb4937bb9bb1ad6277055f : SUCCESS in 4m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ce0e5e22469c4d8a8d0b62732e4a75a1 : SUCCESS in 7m 58s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/68023f9d69a74d9a80dd56e5598933d6 : SUCCESS in 5m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5330869ba70c4266860712d9a1c768cd : SUCCESS in 6m 07s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a9be07e058a34aea9e5a9827cdffb1b9 : SUCCESS in 8m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/507cd98f17544598b4557769d471d0be : SUCCESS in 14m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98b49782861c4dfe83f2401133187884 : SUCCESS in 6m 47s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/56d676f5da624fa2bf384dcedcdd7c3d : SUCCESS in 6m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7a5697708f69432bb3f394064f68c433 : SUCCESS in 12m 14s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/83c79607060747e19482f4454a622559 : SUCCESS in 2m 36s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0243ecbbbddf44e0a03550493b14d3a5 : SUCCESS in 11m 43s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/abf7a077f8654be0bbbd92ceed05d822 : SUCCESS in 22m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1c2baac74f074b2baf089c85e9e1c384 : SUCCESS in 8m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e579d0bf707e4847b7c767020b148c69 : SUCCESS in 8m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ea495978fb5c46f88274b7f22b7e8eaf : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a69e88b728954ae08522f60e84a7d469 : SUCCESS in 2m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/58b3446f7b5145a18f65e3556587431b : SUCCESS in 30m 51s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b6e933a23af44bb885e507c283de804d : SUCCESS in 7m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ca2a1848fbc246469bf2808d4368114d : SUCCESS in 3m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08567b3f160b48b48b8fb680f11a7962 : SUCCESS in 6m 24s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c0edc3717950406d936849c6aafd4926 : SUCCESS in 19m 33s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/93e83394f45d474dab2af9b73a3109ba : SUCCESS in 3m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11256b0ff5d04af192f0c7e948dd9d75 : SUCCESS in 56m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91f9159b8d5a41af9db36458e38f0470 : SUCCESS in 1h 51m 14s","accounts_in_message":[],"_revision_number":1},{"id":"05ecbb8a26d55cc44f658953af08f10dc5ae707e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:24:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5fedbc90f00ebb85feb554ddd77cf368ec20bef7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:25:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d76f4959530e42c69d1e50b08c8e8f84\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de637f65e19440ba4102074cdac7d17 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be23441b5205b4b4afb9b579f475d286d6958781","revisions":{"be23441b5205b4b4afb9b579f475d286d6958781":{"kind":"REWORK","_number":1,"created":"2025-11-13 13:56:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/967032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/967032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/967032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/967032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/967032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/967032/1"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:46:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:46:02.000000000","tz":0},"subject":"update constraint for python-mistralclient to new release 6.1.0","message":"update constraint for python-mistralclient to new release 6.1.0\n\nmeta: version: 6.1.0\nmeta: team: mistral\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie51423df50429031ce49ce21daffd113289a6384\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Arnaud Morin \u003carnaud.morin@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I2b42b690d5083169ebb063252e184affa07e991a\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be23441b5205b4b4afb9b579f475d286d6958781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be23441b5205b4b4afb9b579f475d286d6958781"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
