)]}'
{"id":"openstack%2Frequirements~967033","triplet_id":"openstack%2Frequirements~master~Ib05053ed2315b5f101b46be8445d082f6c9ff608","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib05053ed2315b5f101b46be8445d082f6c9ff608","subject":"update constraint for python-manilaclient to new release 5.7.0","status":"MERGED","created":"2025-11-13 13:56:32.000000000","updated":"2025-11-18 22:06:03.000000000","submitted":"2025-11-18 22:06:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967033-new-release","meta_rev_id":"82f8f5b8a182be3430f6696e7bc8a82ee75d9bbe","_number":967033,"virtual_id_number":967033,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-18 22:06:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-18 11:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-18 20:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-18 20:10:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 16:05:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 11:51:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:10:10.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"c4e1da1810dae16a99b6fff9678d2fa182efb8cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-13 13:56:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb41e223b96fbd04a2f37a2461d00cd9c7703ca9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 16:05:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48dc9b8c114a4ee3ad116f9b0394f8a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f395408a98e471cab3766f66c3644a4 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85181d64c3f64f76a5f04b42620b70cf : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f984e74887314239bd52cb3eb95f6eac : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4510cef91c647729a911d8b15b90f83 : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c250ecbbbbe7404a9cf5d10ca6b6cddc : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e569e1fff244cf7ac5558a4dbed62b1 : SUCCESS in 3m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/10cfedd212644bc58f4c3b530854c6ea : SUCCESS in 3m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/43af4e01b48844c19d2ec5167d2a201a : SUCCESS in 5m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/46abc159e81e4dd3b703bc2146891cad : SUCCESS in 7m 46s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/1ab2b24e96a64b1397acfa315ae3df98 : SUCCESS in 10m 14s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/717e46082998442cbe3bded1d44d2624 : SUCCESS in 3m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c9b8e1d7ffdb41d684c8e15a5c289638 : SUCCESS in 6m 04s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/d610981ef9604fe18d633198cef682c0 : SUCCESS in 4m 10s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/18bfb9464baa4082bb187163f0672102 : SUCCESS in 10m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7429516e1c484dbb8cfb787ed95b3f58 : SUCCESS in 4m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/38f7d65e17604d1b9df73d8f7ab4baf3 : SUCCESS in 5m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b7b7339fb9344302937bee94896ec745 : SUCCESS in 6m 51s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8cb7d36c79ff4ff4b002ae4ca89e6d20 : SUCCESS in 16m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f9c9f61dcf66438a925f590bdfd78fec : SUCCESS in 4m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3d1ea1939d674ba1bfeeeb8ec96cf0d2 : SUCCESS in 8m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b3356852c0cb40a69bb4cd153ed6789d : SUCCESS in 12m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d3e6366d05c64c9085e2dc9158c73f25 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/da2b5746822845a5b542946ae66abb13 : SUCCESS in 11m 05s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/18714a0bc74841f98a6154eb45a30da6 : SUCCESS in 3m 14s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4143631ad1364d0ba5c6060a9bedef4f : SUCCESS in 5m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/a7f0030ee0824f8ba42bf810af0aa1ad : SUCCESS in 11m 56s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d27bddde3134470684a52a1a56f02b54 : SUCCESS in 28m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d847f3864c5d4b35a0e9646837c61367 : SUCCESS in 8m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/374a38cc9d124d47a8dacca9fa02484e : SUCCESS in 13m 37s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/dddca3762c584870bb40891d30870746 : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43bbe0437bc8433a95e0d4d5c489218e : SUCCESS in 2m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0349df57c6124b279c8e633c436bc311 : SUCCESS in 27m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e068a79b9c144a4eb598a49e8c72c806 : SUCCESS in 7m 45s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/2d23d8f483e54e62a6230ad10a4ccbcd : SUCCESS in 3m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5109cb366a234261bb38e745ae60f1ce : SUCCESS in 6m 30s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dc8bb6ee2d0544c4892696238fda5588 : SUCCESS in 18m 43s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d84596c7d080412095d289e030ada7ab : SUCCESS in 6m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb62cf5439a646959c86660fd2c20537 : SUCCESS in 55m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04df7c38d67b494e99234c00e0835948 : SUCCESS in 2h 01m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c39a62b6d922c41f93b85b2257ae746d4ce1a0e7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:51:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c515931d0c1a51155a2d6fb51ebd0d8d000174f6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:10:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5dc82495771af03c0d90edc30801628b5d41c5de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:10:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bd46b81fb1543b56f1cd8a87cb7a733c050e97c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:06:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a23d40734204c7d976c19bd3dbb3810\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0828214edba349f8a712cad79d11e820 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47dbea6d105d41a699ac721a85db05e8 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50633e3043e34132b98f6b1930628703 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bff6a2f9a0f4b31b3f0a9fa70e6eb73 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b60ab18e52d4df48cd04e2c78381b77 : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0f59a932a59a44559be29976d85182af : SUCCESS in 3m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c9058f59695742e28eb4d634e2eb7956 : SUCCESS in 4m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/94018bcf3e3449c08767de6c18d2e0da : SUCCESS in 7m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/41c5bace02c24bd3b3195b6b6ffdc8ae : SUCCESS in 4m 45s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5c14842c4fd743c0b3562a1e26693d6d : SUCCESS in 9m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/94c85750dfff4e8f8aa1be305ef026cd : SUCCESS in 7m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2a867763894c4668bdeeedbf47c1b91c : SUCCESS in 6m 11s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4f125941f1bf411c82505dcc6cd5c264 : SUCCESS in 4m 50s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a99b0d1ea3904bfb9eff64cf12339e74 : SUCCESS in 18m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/998fdd9f97f4447081e447c97aaa96db : SUCCESS in 8m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/51a71f382ab04e42b00051cc1d43cfbc : SUCCESS in 10m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7d9eedc6cc214b488e8b6a9fe08cdd0f : SUCCESS in 13m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/d42d0ebecc1d47b5bf1c287699284717 : SUCCESS in 3m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7869b5925d294e1fa08124eda6a470c9 : SUCCESS in 9m 39s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1a561e50d3904eeaa3a1b892d5e50bf7 : SUCCESS in 21m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4474295dbe5f4a22887d1fb1ebcebfba : SUCCESS in 8m 39s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9bddf9b966204387aa746b5d40a9b588 : SUCCESS in 11m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0527ade09b314461a7cab88186d587e1 : SUCCESS in 3m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2ecb6f71938a4343ba6b238c4c852190 : SUCCESS in 2m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/457ccfc8ee2044dfac7a52944aa18de4 : SUCCESS in 26m 13s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/760b9a560d024f6682689a0d28f9e813 : SUCCESS in 6m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ea60b951f7b94b73898f19746a39ec5e : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/51d84829d8b541a0899291cd39745fb1 : SUCCESS in 4m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/46f4d5e1d9d2430cb0c67dff117f72b1 : SUCCESS in 19m 57s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4fbaf003dfae4b769d52305040ed56ea : SUCCESS in 6m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ad786c4cfc94ae5a7d2c5cd17426c93 : SUCCESS in 1h 01m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/017c904f6f184acda29eac79046a624b : SUCCESS in 1h 47m 33s","accounts_in_message":[],"_revision_number":1},{"id":"82f8f5b8a182be3430f6696e7bc8a82ee75d9bbe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:06:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0445413607fffd8c61a54927b2d97407171869cf","revisions":{"0445413607fffd8c61a54927b2d97407171869cf":{"kind":"REWORK","_number":1,"created":"2025-11-13 13:56:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/967033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/967033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/967033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/967033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/967033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/967033/1"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:46:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:46:31.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 5.7.0","message":"update constraint for python-manilaclient to new release 5.7.0\n\nmeta: version: 5.7.0\nmeta: team: manila\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia76666bfb026b81309413ac4f49b19edf4fb32b0\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nChange-Id: Ib05053ed2315b5f101b46be8445d082f6c9ff608\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0445413607fffd8c61a54927b2d97407171869cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0445413607fffd8c61a54927b2d97407171869cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
