)]}'
{"id":"openstack%2Frequirements~967034","triplet_id":"openstack%2Frequirements~master~I9f9fc30def8d0b3bd54d009f08c6f92999aa73db","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-11-20 13:48:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9f9fc30def8d0b3bd54d009f08c6f92999aa73db","subject":"update constraint for python-zaqarclient to new release 4.2.0","status":"MERGED","created":"2025-11-13 13:58:14.000000000","updated":"2025-11-20 13:48:58.000000000","submitted":"2025-11-20 13:48:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967034-new-release","meta_rev_id":"909957c3f65a01dc58d3cd4e209b4aecc27ff663","_number":967034,"virtual_id_number":967034,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-20 13:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-20 02:27:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-20 02:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2025-11-20 02:27:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-20 02:27:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 15:06:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 11:53:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:14:05.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-11-20 02:27:16.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"d91d4dcf4b6f1357ef94285c127fe2fe296d202d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-13 13:58:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"775d017ab330a0c52f0a2e04e36f99c5474168a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:06:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f91fbdef0b0a4a58a5b4cdc189180642\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/019c3e7aef444701ac16e1011611cfc4 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db7feb41b6914026a1c943cca56be3c8 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86b3975f023341f7aa04caab1966e81a : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3cbc470496b4b6688149a1b662ad47c : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13c30acd7ff84f26bd8b0d6567019063 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3058748272774ab686bb2e7ad54c7ee8 : SUCCESS in 2m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ee4c980a91104de49b3f205510354f9e : SUCCESS in 6m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/64c49851f33d4884b3067050288cc581 : SUCCESS in 9m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/672f576c5a8048e68feaa5a591856e8a : SUCCESS in 4m 54s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/cbe05755ae5043c4b348ecf989646b8e : SUCCESS in 5m 15s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/e913650686b041ef86bfd773cb496c07 : SUCCESS in 7m 15s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/bcb0e5288b6a40219180490cb1f4017d : SUCCESS in 8m 03s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6bfb176304dd4a8498d2b64e20a87003 : SUCCESS in 3m 54s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/baa58ce7df2b4be18657b12379d57ae1 : SUCCESS in 8m 56s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7890f884a1804c70817f702265600081 : SUCCESS in 7m 21s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bfd99ba1ed8a4c98b78b2fa39c81c0f4 : SUCCESS in 4m 13s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a3193dc915f344578cf6bafc3ee3d3eb : SUCCESS in 7m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/5b697e35ee1a44d0b9ff8174ac4c820b : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23dcb5f1b0a4435e83d4b3b32bdb476e : SUCCESS in 6m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/818509bd65204eac98db869abd2b4c75 : SUCCESS in 8m 35s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bad0e580f5954a799c230678b245f053 : SUCCESS in 9m 43s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a1f5e8c7d2664902be7fe9804fc31d81 : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7a503c254d174de8a929ec0f89e6719d : SUCCESS in 8m 23s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/4416f11ec39145f4b888c793a839a294 : SUCCESS in 4m 24s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/867d66bfcf294fa882a11a82a8d26691 : SUCCESS in 5m 16s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5c12058f90d84c96ba1c8c4a3dcf83d8 : SUCCESS in 11m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/77916d93895641779c18cd9c1478d6b5 : SUCCESS in 29m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e338967f5ab24ad8b11c0c0b0d9a4783 : SUCCESS in 5m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e59533d726324d33ae671f23091eb7df : SUCCESS in 11m 41s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/73eba3cf9856420691453934822a57c5 : SUCCESS in 5m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/80cc077173af4e768385e98ddfed6838 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3fbf90415f548a89a9ce9942f30a4d1 : SUCCESS in 24m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1e447c2a665448e4a8f81004a3869444 : SUCCESS in 6m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5998d00319ea4d99920373f4eee45482 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80ae8c8d18c4462ca99ff968e4a570f6 : SUCCESS in 6m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7f836a47ee53453a91be97be5916c428 : SUCCESS in 19m 31s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ad2ae2b283124300882e5ac37894ff50 : SUCCESS in 8m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07ce7e988eca40ed91483a4406e720db : SUCCESS in 37m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9938f4714caa4892a5cf5b4fb70fc656 : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1364ef7f965911a00e8ef9efe2bad161bb715d58","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:53:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0dfebb11ab97ac5c553face2fffc95b4b7fc8cb1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:14:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3f5dd8bc04434ebd73e569cbd45632b41dd31d26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:15:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46d7748f32be1b42ef2c164e6ef046350ae25f04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:24:48.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"0d0ae1876e11bca332db85d88e73f8a116996b1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-11-20 02:27:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"16d13ae75d9bc55898e0d456ead0a04f4073415f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2025-11-20 02:27:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e4c81ffe2a7ea46b5aa9e925a5e84fc56a75e256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 04:18:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9f0367724074f779066ffa6e50a4979\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed331094c4c24ebeac77131b91e6682a : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bf6cde5c5984d138336ebdf5da99b4b : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/744deceff53e4ccd90a72bb7bbcf0843 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/372ca12e51d848488e5ba1b582f2fd1e : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cde2363375234af5836aaa8adcbb4034 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1bb1c53bda042929422a5e496ca4a05 : SUCCESS in 4m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3738e2a3b0804e9ea53b80872f3c8cda : SUCCESS in 5m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ff72741f0c084212bc044c35eb370c95 : SUCCESS in 7m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3d48da0264ff48459fff793bba8ece71 : SUCCESS in 7m 03s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b5329e95577c4fa7a40d156442262f1d : SUCCESS in 9m 54s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ac330e3a998f4acb90968f4d23ee087b : SUCCESS in 5m 03s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/9081a14865074716929ec4c4c91af219 : SUCCESS in 5m 36s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b1757387b3ea4bdc9adc985658463d11 : SUCCESS in 4m 46s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/de08efa1755a42728dc54fefd049d1f9 : SUCCESS in 11m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2f3dd033a8f94c3ea29222a8f07c4d83 : SUCCESS in 4m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c2b3ff5485f84266b10a16f34343b774 : SUCCESS in 5m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c13fddefc4874af08753dcfc245b9cec : SUCCESS in 5m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9fe03b49f5094803909e075d45719320 : SUCCESS in 17m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/553da5586ce84291a519520b87cddd3e : SUCCESS in 9m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ebdd6dc9b5f745e889b767710d1252c0 : SUCCESS in 14m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5b87ae29b96946e582cf78b66440b72d : SUCCESS in 12m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/67702f2fc5a743b1b0d8adbb2cfde0b2 : SUCCESS in 4m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/82bd76133e634b75831f5dffd29afe21 : SUCCESS in 10m 14s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b6c34ebbaa07498fa9543c72aed6680b : SUCCESS in 5m 46s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/62457709030c4210bc7c9719c7fa9420 : SUCCESS in 5m 52s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/605111e075b545208534b9daca4aace7 : SUCCESS in 12m 30s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e2ebb988080b4517bc689ae19f16e9b4 : SUCCESS in 26m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1f543e27e6bb471f94d8ea4fc923c41b : SUCCESS in 7m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/68644e9551784377948d7536ffc9622a : SUCCESS in 16m 56s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ace8371a54b542febfa9d23d972f7d8a : SUCCESS in 5m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/22f5ac821e5b425fabfb15f7289e29cc : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/501b99651ad848dfac26a7c551c4728e : SUCCESS in 30m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bf9a24210df04fccace97e5646b9d047 : SUCCESS in 6m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9d9e7aced7ac40f398cb8cec130af4a0 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/340178f5a5384fa8b6fc559a93dcf647 : SUCCESS in 8m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0bb42c8cfb83414fba7153561721cd74 : SUCCESS in 20m 11s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a91601343293467e947a87a5eb6a74f0 : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6ce2c8590574ef7a704fb2c13329eb4 : SUCCESS in 58m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/481b1ebc4386435eadf21c841de3feb5 : SUCCESS in 1h 43m 40s","accounts_in_message":[],"_revision_number":2},{"id":"b987c218ff9d464b7c7d1e7675826bcccacb206c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 11:44:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2dc5364f3428ee68aca7866aee8757d85ef8fff0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 13:48:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82cb9b8c392749dc9bf12135ba8ddcda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca40f4544e3480c825d9f143d1ab68e : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/798a965646d84403aed1e4f00187b3c3 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77d88b3d85af41d69e9dc152350355a7 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/455beacbc7484132a30f50ed98c0706b : SUCCESS in 7m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60f5c6fea342489198ae1e7518df5f39 : SUCCESS in 2m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23a87fc431244ce7abb9499671e4426a : SUCCESS in 3m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b829a5681c6942419c37cb2db454fa6a : SUCCESS in 4m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0bd7884ec64c4b05beca3029172874c6 : SUCCESS in 3m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e02f9bae58714fb39e67d660c038bc17 : SUCCESS in 7m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8102c362b8d1480c9b9de981c57c63e7 : SUCCESS in 10m 38s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e2b07441cde9486eb96d16f2f8f64d63 : SUCCESS in 4m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e3357dc4a462407cab0de8365070f260 : SUCCESS in 5m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e9674c829ea64e34aa8fa27da94281b3 : SUCCESS in 5m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e3822f773907440c8e7a54b95d52bbf4 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ccdde5761d5249e19e21da4aec01fc13 : SUCCESS in 7m 44s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ded653ada75b4a6dbe90532cbb31d607 : SUCCESS in 5m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1fc1a1cc2866450483528d311af99df8 : SUCCESS in 12m 08s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/191a85b135f54eeb94a1b0ea94e239b6 : SUCCESS in 2m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/036cb9bc507b48f9ba8f176c343e18ea : SUCCESS in 12m 21s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7f0ef90913f04f3a8b867fb8561ad8e8 : SUCCESS in 18m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6ed01386d6694caa9bd51744455e0774 : SUCCESS in 7m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/71bf3c12a44d4cfeac0cc09303c41bad : SUCCESS in 9m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d41405028319499e8aabcd0a62e15f88 : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2dd4180cf40d437ca875a9a5bc150eb0 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/46a3cd637bff458f97e9dfd3b6cb3416 : SUCCESS in 20m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d98c2682d45942498e08f730a6f5341f : SUCCESS in 6m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9a25f63fb0d9433ea0dac8f66c59262a : SUCCESS in 3m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/488c38ad6bec43dbb876db25ef198f86 : SUCCESS in 5m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/de05eab3e7854659b1dd1d4fc82b999c : SUCCESS in 8m 12s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e96ea062734d48e3bfe76b1bfbe58661 : SUCCESS in 5m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8263c7ec748c4cec8e391f334eaa1ebd : SUCCESS in 44m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ade29d0eaaf74cf585c8579e4fa949ec : SUCCESS in 1h 58m 30s","accounts_in_message":[],"_revision_number":2},{"id":"39061bd70917fb35d55ada376faa4d1513e55b2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 13:48:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"909957c3f65a01dc58d3cd4e209b4aecc27ff663","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 13:48:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fd3eaa072194d9b853d09b744824488\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d84c8fdf7d147a596f4a127d85aee55 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d35c2a21e73b2b0d1a117f280c0790ba76609e8b","revisions":{"c24e96dd1b3a5ac7272bd58a15c9224ce1d0abf0":{"kind":"REWORK","_number":1,"created":"2025-11-13 13:58:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/967034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/967034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/967034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/967034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/967034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/967034/1"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:48:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:48:11.000000000","tz":0},"subject":"update constraint for python-zaqarclient to new release 4.2.0","message":"update constraint for python-zaqarclient to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: team: zaqar\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib294e81d440f095b2ee467db885135bcdc3ac709\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Hao Wang \u003csxmatch1986@gmail.com\u003e\nChange-Id: I9f9fc30def8d0b3bd54d009f08c6f92999aa73db\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c24e96dd1b3a5ac7272bd58a15c9224ce1d0abf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c24e96dd1b3a5ac7272bd58a15c9224ce1d0abf0"}]},"branch":"refs/heads/master"},"d35c2a21e73b2b0d1a117f280c0790ba76609e8b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-20 02:27:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/34/967034/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/967034/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/967034/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/967034/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/967034/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/967034/2"}}},"commit":{"parents":[{"commit":"0a42c9a01f72a7e5624c9268e5eace28319087d5","subject":"update constraint for oslo.policy to new release 4.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a42c9a01f72a7e5624c9268e5eace28319087d5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:48:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-20 02:27:04.000000000","tz":0},"subject":"update constraint for python-zaqarclient to new release 4.2.0","message":"update constraint for python-zaqarclient to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: team: zaqar\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib294e81d440f095b2ee467db885135bcdc3ac709\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Hao Wang \u003csxmatch1986@gmail.com\u003e\nChange-Id: I9f9fc30def8d0b3bd54d009f08c6f92999aa73db\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d35c2a21e73b2b0d1a117f280c0790ba76609e8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d35c2a21e73b2b0d1a117f280c0790ba76609e8b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
