)]}'
{"id":"openstack%2Frequirements~967035","triplet_id":"openstack%2Frequirements~master~Icdb65e7dfebf7b58733721b3dbfd34c02dcaf18e","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-11-19 19:05:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icdb65e7dfebf7b58733721b3dbfd34c02dcaf18e","subject":"update constraint for oslo.service to new release 4.4.0","status":"ABANDONED","created":"2025-11-13 14:00:42.000000000","updated":"2025-11-19 19:05:01.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"35efbabec562a18402acd0f384818372268d9941","_number":967035,"virtual_id_number":967035,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-14 14:55:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-18 11:50:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-19 11:55:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 15:41:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-14 10:42:58.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2025-11-18 11:50:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:08:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2025-11-19 11:55:41.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"941b21738eb50461d10c9c90cad38d3f814bee76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-13 14:00:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b98cca616dabe5435956ec32316f6db2c3ecab1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:41:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd058cc6673e4d13948cd4041cb950d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c26231c4cd4e32b3d5c19c9d639229 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cd12f81a36743d09434f18e800fe7d6 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e654e6492e7849a49c0b4ae83b0e15ed : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79da3fc3b2554efb807f75c65e805389 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ac05d483bb84faa854a84811bededcb : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/266624ed71e04ea7952b59a392a63c51 : SUCCESS in 3m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/73f8c4ce812e456b93462b80d89f9e8c : SUCCESS in 5m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8572ebbda133468ebffe70710f017181 : SUCCESS in 7m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8fd924c2c2ac4ecb94b5b82dddfadcc6 : SUCCESS in 13m 31s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b709607c83664053aa11f17ab47388b4 : SUCCESS in 8m 59s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2fc602b0e9ce42aea8edac985cb96591 : SUCCESS in 5m 20s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d5495c3c3ce8438b85b8a1df15d19b45 : SUCCESS in 5m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5d20926a2fbf421db65ff68a6224a0f6 : SUCCESS in 3m 44s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7454636a5a5f4b3b8766a6ff4d4ccff5 : SUCCESS in 9m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5b7c103c61974f70a887db623d7a677f : SUCCESS in 6m 53s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a1da70ddec6c4a289f90036a8d00aa31 : SUCCESS in 5m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8d8f896bfa0e46819a24165d0b64c2bc : TIMED_OUT in 41m 12s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0314b5ce6f684945a9e38799e69ae128 : SUCCESS in 18m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/41b4f221cf124feba913975885bfa82d : SUCCESS in 7m 46s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/8786811f62944c769b6206af25f2d17c : SUCCESS in 10m 15s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d9659eed3267475bb3fe53020896f37a : SUCCESS in 13m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/ebf34e325ab84895bf5c8cb16622268f : SUCCESS in 3m 40s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/53ffd9cc86e8477c8482fb1de96cfa2b : SUCCESS in 15m 18s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/93148eae964f4146993f0ffadadfb57d : SUCCESS in 3m 31s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/11b3d72b515841dea26a79ca4c0d86b3 : SUCCESS in 6m 12s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/42a76e37ebfe47df8a550ac99df62f62 : SUCCESS in 10m 34s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/97aa0aa6bf374b2ebab5cd65529dbce4 : SUCCESS in 29m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/22bb3a50f6c2448d97162df7da44558b : SUCCESS in 5m 08s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/225b15feef5b4c89b4b4cf3b44c1974f : SUCCESS in 13m 43s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/df42de556c1c4cd2bd367fc1f21a979e : SUCCESS in 3m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8a17e0173b41497fb8bccfd2856844d8 : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75976a6c6bf941c8984b279c4e32cb5f : SUCCESS in 25m 36s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/60e88d299ad343be85f9f7c59e0d2c7e : SUCCESS in 7m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5b138c7fafff4acdb356d8d7726ff084 : SUCCESS in 3m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c67a0825ada14d6ca814756efdd4b5bb : SUCCESS in 6m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/5acf1616ed854e7fb60d61622ee16f36 : SUCCESS in 19m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0aac5a2f56c84e20a25081eff78461f7 : SUCCESS in 6m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ae09a1f92a841eb882e4bd75fd8638e : FAILURE in 1h 18m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ecc8f68dc5744ed9be6a48f2b002e25 : SUCCESS in 1h 35m 53s","accounts_in_message":[],"_revision_number":1},{"id":"98333b599b377c9414b7e008189c64a5b8653514","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-11-14 10:42:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"980ef6a51fbc6a0880fee58eaf8017c93120a52d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 12:02:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be5f3722021f489898d02dcfc6d5beef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f4ff96235d4f06888ba872c46f3a83 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24987525b4f64743a21ed31f3439f3b3 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7dbfa1ff7f6b4fdcaaed03b65a65cc21 : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c85a3bba5ba242d58130d46496f4eb1e : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0102a8d833ec4b6d9b27b4961b6870bd : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96c180ce44e34beb8cd6a0e19dfa92a0 : SUCCESS in 3m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/20519c0f72b646d3823bb4b5da0a23fb : SUCCESS in 2m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2afa4b3bbb4e4f20bb3b2524d19230d8 : SUCCESS in 4m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f5a10cda268a4136a3b805de32147ae9 : SUCCESS in 12m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/0520e8b1e9d5496183d49754851fbf0d : SUCCESS in 11m 11s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/558a9eed07924f798fd58b4d54b7be97 : SUCCESS in 3m 04s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/2818451ef22c4c428e08aa23facd40c2 : SUCCESS in 4m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ffad00a257ec4b13ae674118266973a2 : SUCCESS in 3m 38s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/bcf36261030b475e8e245967070d0782 : SUCCESS in 9m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5d216ca5195e4bc5ab1294a6488e106b : SUCCESS in 5m 40s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/6dec1d3e806445539f5850927ea75cbb : SUCCESS in 3m 22s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7e31cb3bf153414b9101bde6c9c2c865 : TIMED_OUT in 40m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0da8c27bee7e4f5c99c71181dbb5985b : SUCCESS in 17m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/70942699a70a4e09a1c52661393ed620 : SUCCESS in 6m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c7f7fe42a036422fa1e824f78d7d07f5 : SUCCESS in 9m 24s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/13e7540ca0d6453daadc4494a8e7d38d : SUCCESS in 12m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5147d617155b42daa551d7e2d26f5242 : SUCCESS in 3m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/622cf58a4ee54d699c43ba2c847b1a99 : SUCCESS in 8m 38s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/24a32fd31de04ff88f45e21c933fc221 : SUCCESS in 4m 33s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4b493a7d98164992bb007b24368e9314 : SUCCESS in 6m 42s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/c0e18edbda9246e7a4c5c598ec904505 : SUCCESS in 10m 52s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/15b04dac80fb420fac6249aaef7b9acf : SUCCESS in 25m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/351a312eadd8468abdb4e1728127d153 : SUCCESS in 7m 35s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9b6d213006cb4b359f288d3de3a42ced : SUCCESS in 13m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/643dcd7db0fe4d8d82bbeca0c5638a0d : SUCCESS in 3m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9906f0d25a7745e4bc9d4e967d596631 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3c085e639b64e8097b76def6ff4abef : SUCCESS in 24m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e522863f21784d9cadd2191b503035dc : SUCCESS in 6m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a10a2d6674444e21b937715c90d641f3 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0f61b679ee4247ccae585a3b4ce3fdcd : SUCCESS in 5m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/38b82a89c3fb413ebb3b123e76731c6e : SUCCESS in 18m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c25865c442f04cbbb510c3eff3672c08 : SUCCESS in 4m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97ccd715c80c4c62b04699fb00691fb1 : FAILURE in 1h 08m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d73ce9170afb4e3f8caafbf5068d0abb : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":1},{"id":"73306dbddc9e81c7b2bf66afceee57b6698e06c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-11-14 13:10:14.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a6c372135fc0a9ab27a830289e395241a693ab7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 14:55:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3fafa144465499eb1553ab05b9a4ce7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7fb22825984da88e68e0d9e2fe76f0 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/175388c47ba54b8d88bfcddf8137f0b5 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f48991ed408411abf58d66afaf29d93 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72029bed7a764acbb3e3449f0e44a039 : SUCCESS in 6m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a728380c12044c75b32e386670028b1f : SUCCESS in 2m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ad9772e0ac34be9995b1dd521b432c9 : SUCCESS in 3m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2ba81d3497754f408bb5fe238e41681e : SUCCESS in 9m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f012fcfb645348768cbf403ed56ce623 : SUCCESS in 11m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a958b2a92b584b8da6619de9116b7f79 : SUCCESS in 7m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a3258e15f87b46f6bc4b94a19991a562 : SUCCESS in 10m 52s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/eb7ec5194633441e96c288b1a7951ea1 : SUCCESS in 4m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0678c4802d56490593c5b23eaf0627e6 : SUCCESS in 6m 40s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/352cc1b84a8e4ced8ca41eb4df4309eb : SUCCESS in 3m 45s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0f2b59cfbbc34bfd97aae837f5ff084e : SUCCESS in 11m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e36076990276413c9f176af18d0a7366 : SUCCESS in 3m 39s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/56d40d9b18ce4ec3a271db94daf2c0e2 : SUCCESS in 5m 47s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/17e93588a0db4679a09f1aba91da76bd : SUCCESS in 3m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/28f39bb56f7940239d9259e132d1fde5 : SUCCESS in 17m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c686aa818964ca797f202ec5ef21f56 : SUCCESS in 6m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9922e9987d3249248447dcc4c7237d27 : SUCCESS in 10m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/091f19e18ca349a8986f4480709a0a8f : SUCCESS in 8m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a7352e6346174b80a000c3322b6c6fad : SUCCESS in 1m 56s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6ecc393111294af480ee945aff4810b1 : SUCCESS in 5m 17s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/123f4dd736074ad1afefe292e45675b7 : SUCCESS in 4m 57s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5d83604ce8634ab49ff1806127393133 : SUCCESS in 5m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/792e36d000244a3299c3398cfbfbe13d : SUCCESS in 11m 24s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c7cbd51c8acd4f5789ac1f5e061ade9f : SUCCESS in 18m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0e3f49f4cbd84476aed0c7a794ab3765 : SUCCESS in 8m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8c91f7a83c734fd5b51776c203626452 : SUCCESS in 14m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/151f9714d42a4a9aa21476bd826d8f7d : SUCCESS in 3m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c8300a1bf60c4bb3a656623026592cca : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3de8f4f46ab4e9eb21379a5e369926e : SUCCESS in 27m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1533c83b23614bc1857ad7a2cef209e1 : SUCCESS in 6m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/630eb3e950db4dc2af871dcc70dd9f54 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a90b819ab7b24f3a8c24ad84967625c1 : SUCCESS in 3m 57s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/55269ce018c7403db080ba51ec3986f6 : SUCCESS in 13m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/e89090c3b23540e499e466bf2811e403 : SUCCESS in 3m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb6509bfdf434fd19c41a39be6c5b26c : SUCCESS in 33m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/552c9dc07ede472da704017dc64d02f3 : SUCCESS in 1h 39m 11s","accounts_in_message":[],"_revision_number":2},{"id":"70f572c903cfd608f75bce56f6a551d05ae911bd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:50:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1bb7c5c7b4fd7ae079981476b1a84eb32328f811","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:08:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1925c69cc1ea7e087ec50d4e29208b1aef070cf5","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-11-19 11:55:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"637ec5f6fa3b348d29ff8904be62b4aacba33852","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-19 19:04:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35efbabec562a18402acd0f384818372268d9941","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-19 19:05:01.000000000","message":"Abandoned\n\nsee my comment above","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"507d606d9917e4a278de72a24b9402fee12942af","revisions":{"1aeda21645b6442be0c9e8d90467c741f5e00646":{"kind":"REWORK","_number":1,"created":"2025-11-13 14:00:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/967035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/967035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/967035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/967035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/967035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/967035/1"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:50:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:50:41.000000000","tz":0},"subject":"update constraint for oslo.service to new release 4.4.0","message":"update constraint for oslo.service to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Balazs Gibizer \u003cgibizer@gmail.com\u003e\nmeta: release:Change-Id: I97361e2c648ea6026192d5f47187ec5ef01e2b09\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Icdb65e7dfebf7b58733721b3dbfd34c02dcaf18e\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1aeda21645b6442be0c9e8d90467c741f5e00646"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1aeda21645b6442be0c9e8d90467c741f5e00646"}]},"branch":"refs/heads/master"},"507d606d9917e4a278de72a24b9402fee12942af":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-11-14 13:10:14.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/35/967035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/967035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/967035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/967035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/967035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/967035/2"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:50:41.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"gibizer@gmail.com","date":"2025-11-14 13:10:14.000000000","tz":0},"subject":"update constraint for oslo.service to new release 4.4.0","message":"update constraint for oslo.service to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Balazs Gibizer \u003cgibizer@gmail.com\u003e\nmeta: release:Change-Id: I97361e2c648ea6026192d5f47187ec5ef01e2b09\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nDepends-On: https://review.opendev.org/c/openstack/heat/+/967212\nChange-Id: Icdb65e7dfebf7b58733721b3dbfd34c02dcaf18e\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/507d606d9917e4a278de72a24b9402fee12942af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/507d606d9917e4a278de72a24b9402fee12942af"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
