)]}'
{"id":"openstack%2Frequirements~967038","triplet_id":"openstack%2Frequirements~master~I14fff21b329d0d79a94dff68afabb3b6d6d91088","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I14fff21b329d0d79a94dff68afabb3b6d6d91088","subject":"update constraint for oslo.rootwrap to new release 7.8.0","status":"MERGED","created":"2025-11-13 14:03:38.000000000","updated":"2025-11-19 15:46:19.000000000","submitted":"2025-11-19 15:44:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967038-new-release","meta_rev_id":"a80a18a88964e2c87cce8b067322303029ff4787","_number":967038,"virtual_id_number":967038,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-19 15:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-18 11:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-18 20:12:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-18 20:12:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 15:20:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 11:52:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:12:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"557927ba9d178af2c791884ef169f6db2d896e00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-13 14:03:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0445e5dfbcd1817aac2a73746c8f6ee602fe2180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:20:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/123feacef9614ce98c3c29786532fd7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0104fb5c5d384b7ebb5db116d19625f9 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce0d4e8624c40e8a916dc7325fab82a : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02d604d213694890aa89d00f2caa3733 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48134229fbcb4ba59b971746a25eb8f9 : SUCCESS in 8m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/760c64b76cf845adba0e285c7ded3577 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c84278994964c1aaa993a73469346bc : SUCCESS in 3m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4a51ba6af8814271b47507fcab6aa68d : SUCCESS in 4m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/983ef44f3daa4213bfd268ef94efb786 : SUCCESS in 3m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8f00823d2b4a4a30bd0c9f241a2ae06f : SUCCESS in 4m 27s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/770b0a0593ce49ac88d93c8ee3240e2e : SUCCESS in 8m 41s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9da231bb9f4c4eef950cf8b966198cc5 : SUCCESS in 7m 06s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e042d49ae6484a2c902d014a1a2844ef : SUCCESS in 5m 50s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0e21e8cbdee241a88b9f24665b6bccc1 : SUCCESS in 3m 27s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c290b3004c0a4b9e874be36c396e4b4a : SUCCESS in 9m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/34f8495ccede45199580e73bc03ea9df : SUCCESS in 3m 25s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/43d683e93378427eab8c1391dd20437b : SUCCESS in 3m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/655dd586a5c1411e8df07c18a34418d7 : SUCCESS in 6m 18s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fb9c1c5b95764950b0b6a5f444e144a3 : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c3f16ef2b2b45e59b68f6d968b7142e : SUCCESS in 7m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/12b76ab886ef461cab33a9283ccc85f1 : SUCCESS in 6m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e7c7eb729ee14f21b2dfd7b9c072e66b : SUCCESS in 10m 16s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/278acd5fe6c242328734e8a51c2dc9c2 : SUCCESS in 2m 40s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/72ac78bf3d704ecfa334175f97409e96 : SUCCESS in 8m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b18cb89b09ba40d695f453d4a1b3d708 : SUCCESS in 4m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/8512dc3d361143f784d81a89f2dcb6b3 : SUCCESS in 5m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/db1a26a9fb704874b779ad3b34fcb3d0 : SUCCESS in 10m 54s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3c4fd3653d47445491a4afa1cd4d9d93 : SUCCESS in 25m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/948a64b17a524fcc8d43461452f97734 : SUCCESS in 4m 29s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/45d61d1c727d434ea47c6fcf1efe1306 : SUCCESS in 13m 37s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/01d0d8bec40a4c6cbd1fcb4bb9ae5d9b : SUCCESS in 6m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/34162791f5224f34aeebd3e111c862fa : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/af27141ff5114bcaa308e942fa90fb8e : SUCCESS in 25m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8378a8c1f1b540e1aa0edd311c0442ab : SUCCESS in 3m 57s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/12394545a6d442c48dc24800015931a4 : SUCCESS in 3m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/395ed59bd4144a55b9ce01f6914adca6 : SUCCESS in 4m 35s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d9ce610ccb334ce09a6ca733183274c8 : SUCCESS in 19m 29s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/216e4d255a01496596dc94133e2e2360 : SUCCESS in 4m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bb7d521bf0148e786cc6d06086d6a32 : SUCCESS in 34m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/322c4dff69c241a1880b75f4005d77ff : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"999d81e64fae22f643b69cdbdb9c07f18c4e8341","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:52:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44a8cbbc2822883c99cfeebd3f2c128500e3ad4a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:12:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3c9658986a068c71a7ec11f41ee0ffa83c9e1df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:13:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c2392b1227e8bad543dd93c186c5e26da317712","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:24:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f01e9bf7deaa45c6b12e03ea5f83f328\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0912375a996d44c9a1bf5894399d4a1b : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4289944424c4d598c13d2b72326f21f : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d44a83eabdec4a8a9b92d73dfa7384f0 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d09298b6cd84a16aadb5e08b016c9e2 : RETRY_LIMIT in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7bf5fb465a5b4497bdb8a9e24602826e : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9205a7846ff4d38901726610d6e45a8 : SUCCESS in 3m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/647592163a3e49c5826efaaab58ecc2d : SUCCESS in 4m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/253fb43755354149a75d92d68f218ff1 : SUCCESS in 8m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/260eb06bb8b74e4d98a9e145fb856899 : SUCCESS in 6m 51s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e12a182da7ee4f87ae461eb328cd6eb4 : SUCCESS in 11m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/152bb353cab24a44b551ab663197a8ea : SUCCESS in 5m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c29a731c540648e88715e59637a6439e : SUCCESS in 5m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/dbf6b9a9c3a24fb1b71fd2301ed32139 : SUCCESS in 7m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2aa5f864c6ef453ebabb783dc4a54746 : SUCCESS in 18m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f014c818c82e41c68eb64fe0d850e185 : SUCCESS in 6m 35s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/60eedba4414c4a96b0f609a80143bcf9 : SUCCESS in 8m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/728b07d0c84a4871864bc0d2c51ed0af : SUCCESS in 11m 16s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b5e85607b95c4180a04444bbad0aefe7 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4785025358c2445b87b35eed88fe7a45 : SUCCESS in 9m 03s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c679dab85a6d4eb8931b47b4300ae34f : SUCCESS in 28m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d9fa654da38249b4b6869f986f49c97e : FAILURE in 10m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a0618782ac0e4759b72d6ef93ab0c6af : SUCCESS in 12m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/dd58229c508d417192284c95d92f8df2 : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/63d8a2f90b864a8fb4d926c937212bb1 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7e5e14bbf2f4847bff4d593ef232b8d : SUCCESS in 27m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/97546b676efc40f8a651efb21b801a20 : SUCCESS in 6m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c88fd4c072284280b93b10c241f883a8 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8e6990f104dd481a8698d05f7e9e06d8 : SUCCESS in 7m 01s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/74a343621fda4f1181d15c0325b874e5 : SUCCESS in 17m 22s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ec74c86af73848f0877405be1d737aa5 : SUCCESS in 5m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/584fabaaf1d440fda533f669a7e52676 : SUCCESS in 59m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3c777ac6b9444cbb657ae3326938b7a : SUCCESS in 1h 44m 34s","accounts_in_message":[],"_revision_number":1},{"id":"84d5599478ea5ecaf1ade6c20fcd844fef3545c1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-19 09:08:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9b8573b6426eabed79124fd7387b2296b98bb92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 10:00:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e374ae6b9746460db5a6dc7aeca730de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df05cb550f0e4457b67cb49a79816634 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbbedfcec41d4aeab5b50d1cbfc6e8c3 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a82b67593994a2dbf218f4cf1b106b4 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca30fa26a1444ee884ed0a71c0b628ef : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bae77d32bee944b98dfc485b1a90c663 : SUCCESS in 2m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1dc4d002b564d3f855df24ff339ed4e : SUCCESS in 3m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/824104bd133c48d185b431775a8f5ee0 : SUCCESS in 5m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d10b31af6f354a239cfb8e6e99b110a3 : SUCCESS in 9m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/183026fc4e7e493ea595e9a236023aa9 : SUCCESS in 4m 51s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/981989494bbd4b8e84ef057e03241618 : SUCCESS in 9m 24s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/6109ae0e3b9f41be8b7d3c0ffcbb9651 : SUCCESS in 3m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5ec16bc5238e4441ae062fbc854d2224 : SUCCESS in 4m 20s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3ca383ec5b284e76b55ef3629a3611bc : SUCCESS in 3m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8c7fd6f5d3f04f7c84e71cc823d73f77 : SUCCESS in 11m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/298c6e617de740b98e7c5063bf798b05 : SUCCESS in 4m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f089331549c9426ca17af009cc256f4c : SUCCESS in 5m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cf905d6aec7c4d86a75790b7f9537318 : SUCCESS in 9m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4b2e9d289afe4f409f2a42ba2bcc372e : SUCCESS in 17m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0117c6df78274c3592320c73088256d5 : SUCCESS in 6m 34s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/5653cde3bed848a98320686b38eee8dc : SUCCESS in 8m 22s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dd980a8cd1314527b375fbe6ccb4ad02 : SUCCESS in 9m 21s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fe7fcc26c97f4dce818928679fa14ddf : SUCCESS in 2m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/56213e16aa8a4edf9061684a9055a29b : SUCCESS in 7m 10s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/79a49ddc8dd74f54bc65646a9f2c35f1 : SUCCESS in 4m 05s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/09f64d94e7c04bc09f9881de6d6c85eb : SUCCESS in 5m 46s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/6d5f309aa1084a769803ea885ca6862e : SUCCESS in 12m 56s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/bd9311e2ed724e84abb5eb33d23c0459 : SUCCESS in 23m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1902ee4e128847718a675d496a8cc57b : SUCCESS in 11m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fe68a01627b24163989f4a4a2d873404 : SUCCESS in 8m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0feeafd1a3e24620bb4112796ef13831 : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/10a81f7a2ccf43f5bead515d4957de65 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e81ea7af60204cfca375562864434fb9 : SUCCESS in 28m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/07f06e0b49cf47b9b9a6df652307e7fe : SUCCESS in 9m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/59fbb4091c9e46eeb478712b623063d5 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ea6adfb5e6d43739e39593be5cd789c : SUCCESS in 3m 18s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/05705fd98e34496ab69a4d92dce60bcb : SUCCESS in 15m 56s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a2c1d467080e4950bc4f9f47ec965601 : SUCCESS in 6m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/320c85cbeacb46d1917ef99ab4dfdd42 : SUCCESS in 48m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/354807c9145943459ddc8f051cf16471 : FAILURE in 49m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4c33b4f2a6861f79fdafa27e83a5dcecdcf4fb9a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-19 12:48:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15a54bd5c87afa68c8a57ff487ce3ce138b7336d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 14:01:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ade75ffa7c7469abaefd99806d4e4d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b882f8752a0848878646a68767106328 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/304ddce4d1a340f593fd172ab684bbc9 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ef367aa85fc446b88a2c62328ad1d03 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84203310fc5746dabe6b463fc4b7a654 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d60a8b427424d4d862de665630b9865 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c2c082548ea47078776a1627a957c13 : SUCCESS in 3m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7a702ddf5bc5405c8b72e10b9602c7e4 : SUCCESS in 5m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f05ebee7637a4a0db7313ede6e5e0101 : SUCCESS in 10m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dbdda25d019a4999b80be4bc946fcc5f : SUCCESS in 4m 06s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/8eefe7b04aee4de6b4be5cbaaf0fedfc : SUCCESS in 9m 53s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2be87fc9234f4fbb9bceec9ec45547bd : SUCCESS in 4m 53s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/dca82971c0ae4ff4a6a1fc341d8019c2 : SUCCESS in 5m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4e68f17aeea34e74a174fce24443ccdd : SUCCESS in 3m 53s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/90adfe2f88ef4420b9ff61c038f1d70f : SUCCESS in 12m 09s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a430a342dd864d4b87da3b9c78787ff2 : SUCCESS in 5m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/76f3f1d4618c4f64ba3b7e9dc4642a21 : SUCCESS in 5m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1b2228b2f9d94451aedf4bb147ae304d : SUCCESS in 7m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1fb96425b3254ff4b99da34243240fab : SUCCESS in 18m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e624ce0af244dd9aacff27e2223df2c : SUCCESS in 7m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/af6ac71a84c44af2a485f5bcab8d583d : SUCCESS in 9m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/12f72c317e42458ba6e9b97e70bc18a8 : SUCCESS in 13m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3aa2fc932f45409e909b4c58d22562bc : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/88f6c0778ed045a69ff124c5b7efd25a : SUCCESS in 8m 17s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/39c35377946b412284c5dca74c03137e : SUCCESS in 6m 11s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/295b9ec84058491a85a669cc2d830ee0 : SUCCESS in 3m 14s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/65418fdcaf5746ae88109f50bbc91d13 : SUCCESS in 11m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/311f0bef0eb14f57a363075ab0801dfa : SUCCESS in 32m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ddb030c3bfdc4c2ba57eed86d8ec2cc2 : SUCCESS in 8m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/79ebf99330674d5487053701a6f12519 : SUCCESS in 9m 30s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/40271718350e488785164209774cb7af : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa848f571e4f4d54a93ea10ef4c19582 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fe46c1013ef3423e958be3440a18f842 : SUCCESS in 24m 09s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/135489e70ee34eb6b5ad092e995f62da : SUCCESS in 6m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/73ae3d01fe014156849e6d78275fde99 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/64ef774fe77d4c6aab7fbc3bb0542491 : SUCCESS in 3m 47s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/7bc96fc56d8342f78dec6732e29f00a5 : SUCCESS in 7m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/efc23da9e855469696c9e196ad5fca8d : SUCCESS in 5m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c0fc77d517f41a396c4943043ddb580 : SUCCESS in 1h 00m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b0dbd5aaee842cc9aa807a86643bc7d : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f1d4684a93d9c915daf56d816e2193bbd55d0e8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 14:02:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e68f3df31a8f5fea0c35b69a2acf05d3f48718f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 15:44:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d8d01d7433e4ebba0a2e795b24a0ea3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a4909cdc0a0408eb70615aa0841aa42 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5173313bbd4145cebe718bb32260b710 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acffb58b7291429a99e28be25e7a6f1f : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e9fcd54a773409196ef3749c306e2d7 : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed19af8e5cb342638c2b73425f296f69 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b942cb1a8904b978275f7db688b36ee : SUCCESS in 3m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e4e3e461d11b402fb22fc101dd120656 : SUCCESS in 4m 49s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/016affb58366415b92f79853e6b6570f : SUCCESS in 3m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/dbbd600edcf44d16be31197b23a68a80 : SUCCESS in 5m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/19134baf029f4b7d9e67d93b40468a1e : SUCCESS in 11m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ac6b9fc6b6514e6f8e20cdc65e222322 : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b416dff8833d493d925762a9dbcbbc1b : SUCCESS in 5m 49s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a5e20e7e40aa47da8729f6dbb3316bdd : SUCCESS in 7m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6ab81ca16cf94f6da921c82ae4ed8fb9 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/90fa7ccf19b440299ab10c7b29697d9f : SUCCESS in 6m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/20001a4e3f1349dd85971b8e6eb8e9fe : SUCCESS in 10m 04s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b0bb54abac304de59d4782f3ee280788 : SUCCESS in 10m 42s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3a3e7dbf73e04f739d7081fcae840357 : SUCCESS in 1m 52s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/573e8b2a8f6b4a6b9dff0ee3b1016149 : SUCCESS in 6m 34s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/dcf200c295a14636863e3df3b59f905e : SUCCESS in 23m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/aef5a05dffe547058b18b5cd5d7fb9af : SUCCESS in 8m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9116f77a5fef459db15d94b8b2b34c0d : SUCCESS in 10m 27s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0e28f1c083de4affbd1f804a1c04f2a1 : SUCCESS in 3m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/14271c77fecd4ee1921e6004b55941e4 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb6ae65d788b45b2b259437bf2b7603f : SUCCESS in 26m 33s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b6ca842559914661a4aa1acf40c5ead8 : SUCCESS in 6m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9194fb46c6a94f15b2cdc22a17733fcb : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/832e99e1cb20405692fbb9e76cc0eda2 : SUCCESS in 6m 08s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f6703fcdcadd42c789d205b07a6b4991 : SUCCESS in 20m 36s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/cd8860213b0f43169272d0be7e0a6358 : SUCCESS in 7m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c58de7ca47643e8a9ac1885ae851365 : SUCCESS in 57m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35f21bc84c81475d901e5ea57ccc4cd8 : SUCCESS in 1h 40m 42s","accounts_in_message":[],"_revision_number":1},{"id":"068462e8897da44d0046738e4897f86caa624744","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 15:44:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a80a18a88964e2c87cce8b067322303029ff4787","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 15:46:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c467565a7184152914b0e706f766040\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df815cfc4f2a4088a44b011a8d4d2c5b : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8548bb732334d9d04dbdedf3142936bd42b7a742","revisions":{"8548bb732334d9d04dbdedf3142936bd42b7a742":{"kind":"REWORK","_number":1,"created":"2025-11-13 14:03:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/967038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/967038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/967038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/967038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/967038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/967038/1"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:53:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:53:37.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 7.8.0","message":"update constraint for oslo.rootwrap to new release 7.8.0\n\nmeta: version: 7.8.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I53dd5c219667fb587642aca1935cf329e3a59fb7\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I14fff21b329d0d79a94dff68afabb3b6d6d91088\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8548bb732334d9d04dbdedf3142936bd42b7a742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8548bb732334d9d04dbdedf3142936bd42b7a742"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
