)]}'
{"id":"openstack%2Frequirements~967040","triplet_id":"openstack%2Frequirements~master~I90ec769a3adc0389bdcfdede3aba2bdfd44f89c1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I90ec769a3adc0389bdcfdede3aba2bdfd44f89c1","subject":"update constraint for oslo.config to new release 10.1.0","status":"MERGED","created":"2025-11-13 14:04:05.000000000","updated":"2025-11-19 02:14:25.000000000","submitted":"2025-11-19 02:11:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967040-new-release","meta_rev_id":"ae526fa53d8d3b659866eefab6b98b6cdf619958","_number":967040,"virtual_id_number":967040,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-19 02:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-18 11:52:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-18 20:12:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-18 20:12:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 15:45:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 11:52:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:12:02.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"24fc3df66ad27cd9cec16bb5c0ba402efb209f09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-13 14:04:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e635abe6d71fefc3ae5fa30f23cee48644ed726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:45:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb7e3b2be514485fa78c87f1b7608be0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73dd7bf9d5bd40bcb99e4c4af579199a : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944bb4835a5d4ac2a1f4fe2574890e78 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cdbc3a3cbf84d7c9d4c5e872b6da898 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61323aa936e8423885ce21e3553c86fd : SUCCESS in 6m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9fb559b42624544815bfd4e5521a69f : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e14b987b04e474f9503bb8a66a62cf8 : SUCCESS in 3m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/682a41bbe1e142b5b1e7c047aa533247 : SUCCESS in 5m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/825ddaf2b5de40168a7f45fad57f003c : SUCCESS in 6m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f38b09a4dfcc415782e38f996e096a3c : SUCCESS in 11m 56s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/617197232c894c41a3d8137691266672 : SUCCESS in 9m 21s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/bca99b1873a9426db93a508f7f8cfdd3 : SUCCESS in 5m 55s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e2f38e76b8b84f89bc0c374565d77836 : SUCCESS in 7m 27s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/129ee0c0881f41bbbbdbfb7dc4b4fc4c : SUCCESS in 4m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cf6eb7104f4f4b60aedaa3b4cdf22368 : SUCCESS in 10m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/634151b82ac3446dae705470581b9473 : SUCCESS in 5m 32s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e80e3167d1204d1b9186a551d7862eb6 : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/31601e3d88804ff1aa9ce8d39b6a24e3 : SUCCESS in 5m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1c52f851d7454fda8f4344bff4d9e081 : SUCCESS in 19m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bf95e6e030e4d01bc62e06310637cfb : SUCCESS in 5m 03s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c41e9bacb97b4e33a1c7e53e7fa686da : SUCCESS in 8m 35s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/597f30c2d18040ce920996ff22f83ae7 : SUCCESS in 11m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a61d66817a5e43e79ce85c18c92f4331 : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f362224aa9644ed89dde19d05dfb15ee : SUCCESS in 9m 21s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/1158edfc2c114784bff175499368548a : SUCCESS in 4m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/41fba96b8a2a45c4b2e9c98c28c4256c : SUCCESS in 5m 18s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/480c23639e014cfe884636ada865ae8a : SUCCESS in 10m 21s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d71918b58af74ebd87a91cd2b2535a55 : SUCCESS in 22m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9c94d364ea7a47bcac3130166c6f2072 : SUCCESS in 8m 29s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/6b79118d73f642fe91e455a22362e463 : SUCCESS in 13m 10s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b31f4150cc0f49b1a8a756ebc1bdcf60 : SUCCESS in 3m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b82a069cfedb4497a636d15d9b9e5428 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c576aabec1184c80a4586721f41ca92a : SUCCESS in 24m 58s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d923bd3d551042b1a062ab6ab3aec153 : SUCCESS in 4m 58s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f3ab536658354f099f87da1849ae3f4e : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/94cb94ba2b824b508506b295c459d017 : SUCCESS in 6m 25s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ec7af890710c452b91b76e20f4f4296f : SUCCESS in 17m 28s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/2ac7be807b6f48bcba28fd707b81bc69 : SUCCESS in 7m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30a2e268ff4643048444455ae3f88efb : SUCCESS in 56m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/817ec7f943f4498c8e1803e1ab6384e2 : SUCCESS in 1h 35m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6be089fde23eb8e030088de6d066e81ff848e8ca","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:52:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"23c730aa0a1000298c7035191169fee573ec252d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:12:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"52fd4a0428a08d2f12e01e1847aec0d24b12f5c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:12:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a80049c15f48363dc73df409e3bac5ec62bc609","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:24:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2f633c5ce1f4b3e9a05356aedb859ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c90cb20134e445648bc9857b4eb4f3e3 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3595f3d1028043ccaf997a063acd5576 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17e12cb5598b46968e7ec138affeacca : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2614691db1143a0b2bc6bc75d5804e2 : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc47778990c841eda638a57c4e2479e0 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/feaf4103d8084f8e98cdcf3e71a77ecf : SUCCESS in 3m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a4a6217dc3cd4033899bb5490a252e22 : SUCCESS in 3m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6ed12bc7ef7f481ca3e714ca11cfc4b2 : SUCCESS in 10m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/969dbe6cfd4143e4bc3ab3d2acd4999e : SUCCESS in 7m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cd665b09c5e1483587d9bcb9847ac619 : SUCCESS in 9m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/63330c61a20f49958284bce56f61dbc2 : SUCCESS in 3m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5171f6c6d4d54defab4381b5bed37fd0 : SUCCESS in 3m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9b1b508264194452af9085474b2c471d : SUCCESS in 7m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e5aa6c919b74458a98a8965bfc293f8d : SUCCESS in 8m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22a732915d4a4878978b3054c2447986 : SUCCESS in 4m 54s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d05f021c1f5346f3887d054c79ab103f : SUCCESS in 7m 39s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9af4146311bb4c5fb998a3875c00e709 : SUCCESS in 9m 28s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8ca7866da0fe4003953183e116aa4251 : SUCCESS in 3m 31s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a5cc3cd375a341f881c38e64b057c075 : SUCCESS in 10m 32s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c524c1568ff64ddb831f2ea821766b13 : SUCCESS in 21m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2de5960fd412466baee77b283035a30c : SUCCESS in 9m 44s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1ac7a362101b4db3b9512fb28310e9ee : SUCCESS in 15m 01s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/83db907c59f641f5bd90ea432b3133c3 : SUCCESS in 6m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/00550b41187c4d12b524a71cd8f675a2 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/586e808ad562442f9c0cdfa4693a2ec6 : SUCCESS in 26m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/187bf02e254949dc91e0da20a37e546b : SUCCESS in 8m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d24770eaf81a422d8d9a28c43bc9dc45 : SUCCESS in 7m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1920145782ff44259ecf6d7634c3933f : SUCCESS in 4m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/aa7468f6b65c492196354544beb85056 : SUCCESS in 8m 27s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c18241cf60e545ab97e4425edf56e2a1 : SUCCESS in 6m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a639bac65949484e85af620a8473fd30 : SUCCESS in 55m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6034b318956343be97412d553f2c0652 : FAILURE in 12m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8fdc343b7e3295faf9eb0a9311293cd18101ab18","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 22:42:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f87ff970470f523f82f814156892063d97d80ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 00:29:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e426e230dd7e48ababa400c35a839b0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c89ec819394963826b883fd11c640f : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/860a959f48564c199eddcbe30254e968 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f59856acf2b144ce932fa89560fc144a : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7f047d3395749e7ac697bb5fb1dbfbc : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04925b1636e04b5dac6e7bca882c178d : SUCCESS in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3bc84ff683444e62be0377e271b62c42 : SUCCESS in 2m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/396837c021b64551b8652b07fc0fb3e3 : SUCCESS in 4m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/25be9f9e5d1d4cb88e72a271450d14ac : SUCCESS in 3m 49s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1842a63a991a45dc8a124a753649a3c0 : SUCCESS in 4m 04s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a7bb620a009a4ab2afd2fb2de02f4dc2 : SUCCESS in 6m 12s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b8f6d9037e6b479da3319a82b7a20b43 : SUCCESS in 4m 51s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/6d2a071e65354f72a4ea675c46ec12c5 : SUCCESS in 5m 14s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dd6906ccefcc459da9d8b1c55640b5aa : SUCCESS in 4m 15s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c338b1e8693b4d54b0645b992ec8d9f1 : SUCCESS in 8m 12s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/eb471505df4c46ae95787fe69b73ac66 : SUCCESS in 3m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3e23b799602648eeae1dd1998f01b699 : SUCCESS in 5m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fc295a8ddf16457185c214527be9b1c4 : SUCCESS in 4m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2fe4669944084a2a928a37c49bdc2374 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e2b3062db75475791d281e9c7bf51d2 : SUCCESS in 6m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9de476412faf41ee8d970faca24785b7 : SUCCESS in 10m 06s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/3273833c3cd54cca984a8700622cf810 : SUCCESS in 7m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/63c59b6dbdb64534926e1ad6658fe94f : SUCCESS in 3m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a6971ffc41b14160b7c0fb1fa101100f : SUCCESS in 8m 33s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/c3c4d0441fd44d8195138ee4a9755a4a : SUCCESS in 6m 10s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1331b5d1eee449fca0198c2fc3c2dba7 : SUCCESS in 3m 00s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/3691b50364eb46aa9aa5e66aa8e65031 : SUCCESS in 9m 50s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/46308671934040f88d475fd76a30bc9e : SUCCESS in 21m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4f7b666e098044e291c9d011a341c7ca : SUCCESS in 6m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b3a131dedea645bb94765b4b6905c2e0 : SUCCESS in 8m 10s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5c4918989e7e408ba339aa8e4a1b4bd9 : SUCCESS in 3m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/084c74b55836445a9c7f482587592dc9 : SUCCESS in 2m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dc42bda65c06470d862b5339f7cba26a : SUCCESS in 21m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2a7a7a2de0ae4d6690fbaba1b9e3c990 : SUCCESS in 7m 33s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/22fd6328bb244e25b50fc62d5e89f915 : SUCCESS in 3m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/77fcdad023ea42dfa80620da4b5527e1 : SUCCESS in 5m 59s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a39b3869087b405cbfedab706fc3cce8 : SUCCESS in 7m 55s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/fa1294574d0443b6ac9d67be4cd57c91 : SUCCESS in 5m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8823a510afda4f809f4057edc7fb184d : SUCCESS in 59m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/435fc69e71314cdd9302a6a2683a29f9 : SUCCESS in 1h 44m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0296b4f1d6cde4133a1686d368b399d0b4ab7546","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 00:30:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"867e7d1553a63e237ad2fa00626124ec4c75c5ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 02:11:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17dc5c4a2cae412f9491413cffc2e6b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea507b08fd94e34a565507a03d79e9b : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07656919fad7411c9beaf1c1fb8359a8 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28569d5f0a694623b7c04653151bccd5 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9015fbc513304f33aecbcee6ca834c7f : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb2f440e86b24b22a90c3001e8ab7937 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79d275892c624ae1951b701904ee7b14 : SUCCESS in 2m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c90150fde55b4646be35ac5ed16be76c : SUCCESS in 4m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b28480e8d44b4f4aaff9cf723587d461 : SUCCESS in 12m 16s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/64e58181afd74f338dedca0c575fc269 : SUCCESS in 11m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/3e3efef2856f4eb9a75a0cbf3030b4c3 : SUCCESS in 10m 51s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c1bf52af7f874850865bba8fa67c5d5d : SUCCESS in 3m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a61bc359827a4afebb14a3e308d68d30 : SUCCESS in 5m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bbfad9ac88174eb4985cf10bfe1d7043 : SUCCESS in 8m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/55b2474bf4704b60bc844b41986985fe : SUCCESS in 17m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e7511f412c0f4b069c45787c7a7c5272 : SUCCESS in 4m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1a4cf04c4eb14416abc8d970c5abf63e : SUCCESS in 9m 27s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cbf01736e0a34d289ffef721cf065be7 : SUCCESS in 7m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/09dff08c1e304c66ada8dfd7f7be1d57 : SUCCESS in 2m 03s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/568a3c2d73db406fb0a1f02363700aa2 : SUCCESS in 5m 58s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0db21b5e11ee4c93bfcfd84ff28596c3 : SUCCESS in 19m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1a8d292082614b91a76f8ab122015657 : SUCCESS in 4m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/24296288a22442c98956c72447129cd0 : SUCCESS in 12m 47s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a962c1a567124b84b6c8319409b325e8 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/55733cc3c29e4e3ca3130eb833a3342f : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9285b10ce3ae4b8b9b4269469845bb57 : SUCCESS in 25m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5efbdb22ebc0457196c38871e8ed8a43 : SUCCESS in 7m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/577c79dd9b8849f2b12d6620337cc2b1 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1ad087821b124abba3b1aa3d474a9ee9 : SUCCESS in 5m 56s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/1c7dd3751dbd45e5b8117d454575b4b7 : SUCCESS in 7m 53s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/bec4fd0c51ff4691a19519f9a7176c22 : SUCCESS in 6m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5c575afcc9b47399c6c49a49d250dc0 : SUCCESS in 1h 10m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a27fae1cc31f4ae088dbefa36798f8c8 : SUCCESS in 1h 34m 43s","accounts_in_message":[],"_revision_number":1},{"id":"36ea7cd79fa8111508a7f91b389cc02d02c79433","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 02:11:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ae526fa53d8d3b659866eefab6b98b6cdf619958","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 02:14:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32db98d6c95746848d0ef210c2d04798\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dfa0c5c1c6d48e6a31aab9a87235d7c : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31efc27d739cf7a9b4d3322ece305338967ed08c","revisions":{"31efc27d739cf7a9b4d3322ece305338967ed08c":{"kind":"REWORK","_number":1,"created":"2025-11-13 14:04:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/967040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/967040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/967040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/967040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/967040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/967040/1"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:54:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 13:54:04.000000000","tz":0},"subject":"update constraint for oslo.config to new release 10.1.0","message":"update constraint for oslo.config to new release 10.1.0\n\nmeta: version: 10.1.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I522b7dc0534b42074552e2b5f6fbdd4af99f4fd6\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I90ec769a3adc0389bdcfdede3aba2bdfd44f89c1\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31efc27d739cf7a9b4d3322ece305338967ed08c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31efc27d739cf7a9b4d3322ece305338967ed08c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
