)]}'
{"id":"openstack%2Frequirements~967044","triplet_id":"openstack%2Frequirements~master~I0da77979f41bf2eb1aeb5fbb92fb1ab8ee9d4256","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0da77979f41bf2eb1aeb5fbb92fb1ab8ee9d4256","subject":"update constraint for sushy to new release 5.8.0","status":"MERGED","created":"2025-11-13 14:12:16.000000000","updated":"2025-11-19 06:03:51.000000000","submitted":"2025-11-19 06:03:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967044-new-release","meta_rev_id":"0e69fa8d1b6c7faf2ce5fca4d4daddd888796ea6","_number":967044,"virtual_id_number":967044,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-19 06:03:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-18 11:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-18 20:12:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-18 20:12:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 15:38:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 11:52:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:12:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"5a1a71166ef05b1870d13d96cd22a3421a349030","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-13 14:12:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"375c0a184c740a3e6868e46bbe3b9e29c3b5de52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:38:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aab4d08ed3484b059bee6b816b372acb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a8aff8171514dadb9a27e2625e0d67c : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d30905956a4e2b841e2ae4e802ec26 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8361df8cbcb845a39bd70d7d6ca5fdcd : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db923de0a78c46e6a7bbed874a3d2770 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d6c1f09c91349d3812b37723e15861b : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b2f14d0b55a44d5ba8b6817a3d6a540 : SUCCESS in 2m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1c07cc9ab66c41f0866c8c93296ec356 : SUCCESS in 7m 06s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0b60ab190d374d149780cf7ed1c456ff : SUCCESS in 5m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/eece012def8540c1b4be67de2a8a2973 : SUCCESS in 13m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7eab6c9350804234b7645ecf6cba5de0 : SUCCESS in 6m 38s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5df31d51bff44c6c9fdd98c01c2c79d3 : SUCCESS in 5m 23s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a0cc329f4632484eaf1477e861dddd51 : SUCCESS in 5m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4f73ed56a08044c9bc895fea5fa4e0b8 : SUCCESS in 5m 00s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a76632f1ea6148a1969a02e93c6da7bf : SUCCESS in 11m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8e5e29f24e8842a9b56c05b48f815e3b : SUCCESS in 4m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7426b97d096047a4a7563fc718c678f9 : SUCCESS in 6m 42s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/69f8a98d772e4931a38ea1d50d4793e6 : SUCCESS in 6m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/806d05d417d84cb2b50cb7f1217a9c4f : SUCCESS in 18m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ab0bf3c049684d5b95dccfde6e8a70d6 : SUCCESS in 4m 23s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6e1ba8159aeb43e4b75e5e240c7c9978 : SUCCESS in 9m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9c9efcb386fe40ab9d56ff31ae3263a2 : SUCCESS in 12m 22s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a6438d5b0107478faa16db4eb3c74a57 : SUCCESS in 2m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b5a126318fc04cdeafc76836c91196b7 : SUCCESS in 12m 35s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/18a2e2ad486f4972bb83e3076f488b99 : SUCCESS in 3m 21s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e8d51af8760f4914acace2ab672dc16e : SUCCESS in 5m 20s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/82dd7bf623f840e784cd73f821a2798d : SUCCESS in 10m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2d8513f4c8be4106a867d8072622387c : SUCCESS in 25m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c14f9f637b494a618a05bb2647923d70 : SUCCESS in 4m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f754e301b0a045d4bf2d3c78a8436dc6 : SUCCESS in 10m 55s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/34b326096f1442d18413f37f5244db8b : SUCCESS in 3m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f839f9d308848ef9f5927630e02fabc : SUCCESS in 2m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e39f24cd3a714f9b86f9682d6067a5ee : SUCCESS in 27m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/88b3bf7cd2984c7a8fdc37c9c82dce02 : SUCCESS in 7m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9bf65afa2ca44f4a94239af29fd14229 : SUCCESS in 4m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05d70b3bc2dd4b1684c079d6961c5cc2 : SUCCESS in 6m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3f82323aab2a424a92f2592ed43cff46 : SUCCESS in 8m 04s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/9a4931f1e7334d62b9530ac430f283c6 : SUCCESS in 5m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdaeb4d90f1e45078b0d974904dbd6c0 : SUCCESS in 34m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fd819b61d3e40b2b680923f6f9ae666 : SUCCESS in 1h 23m 24s","accounts_in_message":[],"_revision_number":1},{"id":"40c8b901070f5b3a418d0736b00226826c414d44","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:52:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f03273743eb22141f6fddf997ff65f6ac3592c6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:12:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6dce9dc69972ea8afe48abf663547610e4adf289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:13:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"347d7da39378fd183971ee15fad8d890afa0398e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:28:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c159b84ad3d45389ab666fc7ece1445\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a196211a095d432e9650057ff59a8dc7 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05614c6fa9c9487d86a26fa91377cf97 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97061a4d7e47459984b946c6774c2ab5 : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b85795cb9ced4f719c00b0a8c987390e : RETRY_LIMIT in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac36165bcb64444390da71a13edc8d02 : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2187ee83d3ae483ba4cce9a0ecc9144c : SUCCESS in 3m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7a3d4237279e413e92c2193a67e207c4 : SUCCESS in 6m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ca74eff58369418b8500462aab6eb089 : SUCCESS in 8m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fb8dcc5c55b444718f16b349ad011219 : SUCCESS in 7m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/165f389ff17341188e11ae9399255dcd : SUCCESS in 11m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0cca6519980d4381b81e184b529ae28b : SUCCESS in 6m 07s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/a9935d67d3a2417e9d7179c8c3396236 : SUCCESS in 5m 19s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/34a0f63782504b288c11a844342641a8 : SUCCESS in 6m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/56eb1c8522c54979b53bc7bb7eb37c1b : SUCCESS in 18m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a87aaa36c7654e3fae2115815defe54c : SUCCESS in 4m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/690ae17fa282493c84b04324f4d8107b : SUCCESS in 6m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b0151bdeb14b4c29af256b2692d4c24c : SUCCESS in 12m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1435b2881f904d9980a1a7f29077d77e : SUCCESS in 2m 38s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e69706078d5b4165a4378ee0427e5c2d : SUCCESS in 10m 06s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f7fdd121c62041a7aadc848edddba7d3 : SUCCESS in 26m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3c0de4fdf1a64356be4b1d2a2e701723 : FAILURE in 8m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/fe8de231ea4645c1bd7a1bb44293678a : TIMED_OUT in 40m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e20cbad84365479f9d030716c8325b5a : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/27964c93c88f440d893bc9eff95a6dff : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7bec96bb36044ab4a7b1ce86e8694f78 : SUCCESS in 24m 47s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4536cf3e25014a8ab42dfd3b209ebcdb : SUCCESS in 6m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/793056311b954f9991d2202bccd2138d : FAILURE in 3m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69a95cd69033468f84efaa4aa04ef94f : SUCCESS in 7m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e495fff0e1bd43b38e80ee64bdf71280 : SUCCESS in 19m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ef7d46c5faab4f7d8d1c8740bc93c283 : SUCCESS in 4m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/232a304e0bb84f099278428e6d3e93bb : SUCCESS in 1h 11m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e66f8d708534d53b182110b2289bb9d : SUCCESS in 1h 56m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f63e471b1c00bf72dc856e38f35576d8c289572d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-19 02:26:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f89dd611c4408ecc3faf4e00ba13d5efe90c647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 04:19:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53053f39617b449c891af14dca482c3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/476f420654ca448fa833da73017f54c9 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e848ee54570546fcbbe08db1dcf5da6f : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5318cbe7add4066953ff3be6cf08bae : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48d28955a97f44a6bd1587124fcbc8ee : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/207644c693b743f3ab98606c35bf39eb : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ebac37cf4974d53a8f17f325de31962 : SUCCESS in 3m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d579cb76d59f4cbe8fe9a231bd0587e5 : SUCCESS in 5m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e05e859cd9a34229bb678c3d15664ee3 : SUCCESS in 6m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/950a18b62fba461b9d448bab3bfc9015 : SUCCESS in 9m 01s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/be89531731d642ca94e982cad95bb5d6 : SUCCESS in 10m 11s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/3c93b253908249bf9f5c013b813c2c50 : SUCCESS in 5m 40s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0c76af65eed746f6a27b1089049ba7fd : SUCCESS in 4m 19s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/718eab5094a843eb9560ac95af4490ea : SUCCESS in 5m 14s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/eb26f7de7efc416597e78fe2e39b0e4b : SUCCESS in 11m 18s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0d9578150d2845d1817fa7731eb216f4 : SUCCESS in 4m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/481c1ad9f3cc49faacdf67a6c5e98dc6 : SUCCESS in 5m 18s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/561f6e6780584c57b453b6d908e1a574 : SUCCESS in 4m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2bd7b7a16dbc4096b6c0ebee55e7c13a : SUCCESS in 18m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/868a536fb45545688e533ec45c15b10d : SUCCESS in 6m 27s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ff41ef150d8849298a8d240d2ab6d094 : SUCCESS in 8m 22s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d18451275f5246d6bf5cb80167cd35a9 : SUCCESS in 10m 00s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4ef4cf5f8dc241ff83fe31e561b79a32 : SUCCESS in 3m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ea64172e88fc4876bed9c7a817774817 : SUCCESS in 11m 10s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d1b7eac2f7cd45d98447ab40ad7cacbb : SUCCESS in 3m 41s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2947487879b84a97a3be2c4624a3019e : SUCCESS in 5m 11s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/92580d91c1ec41f88ab8273718136f8c : SUCCESS in 12m 43s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/90e2eea7362e4ec9baad4d76eeb963f3 : SUCCESS in 26m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/545349f214c349d28fa91b1fb278f92f : SUCCESS in 9m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f23c32b1c9a2474a9096051ab8be7d91 : SUCCESS in 11m 47s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/02c8510e42b34f7c9f22c59da5cb8bc2 : SUCCESS in 7m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c0612f5d9be472286692a95ea0c3cae : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8a8419b0bdae418a82b83eac82f7fd6e : SUCCESS in 27m 29s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ea9d5557ad2b4fc5ba1a41db616f1635 : SUCCESS in 8m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/28c7a004912646b991af957da4e87a58 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ee1fbefd8bc42ea961e56f418e4b9ba : SUCCESS in 7m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/29859b4b1cd64adfb339857c547b3c02 : SUCCESS in 8m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/8f9a376c60c444d09e4d20681980c21c : SUCCESS in 4m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe77df86efe14262813bd3ef8274a458 : SUCCESS in 1h 06m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2155b869dd2b4e2392db6354c40c3cef : SUCCESS in 1h 47m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5cd73a75bbde12b270f91230b9f40f07e69119ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 04:19:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a074e8aa3505191b56e786b14c49d29da257428c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 06:03:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97f0e56eec2e4e3bb5b368e6a121f300\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f6c7bc6f5e43f7a327d4f000f37313 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6776fb7e30144a119546cf3751f26e3d : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d08673fe277c48cd9e9bffc28a76a348 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7bb1ae7d35b4601ba9206277256e185 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56a08f7e98394bd180aed68b48caef96 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d81fbdd47c04db29d21c4f285efee99 : SUCCESS in 3m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/22646c713c34474a82994a9df01be67c : SUCCESS in 3m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0a4c75dcd2c94d728fbb04d2f725e2bc : SUCCESS in 6m 36s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/29e6fea1cef54b0f8706a2a5f0373abc : SUCCESS in 8m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/27b359fb851f4f768616fa2179cc93b4 : SUCCESS in 11m 19s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d1605042d2da475b81944b3f555b79b1 : SUCCESS in 6m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/beccf8e555ed416095b0dc47f3c9cc6d : SUCCESS in 3m 09s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5f7975aec9ef4d79afb321a40d272ce9 : SUCCESS in 7m 08s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d24660315a0d4089bb2772d5867813c8 : SUCCESS in 8m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6482417eef0c4de0bcd4d9c730052417 : SUCCESS in 6m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/df48b1b4ba1c42eab86cbca5086be699 : SUCCESS in 5m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b76217cb332b4599b00445969fe5f811 : SUCCESS in 8m 59s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fd592afc8d2a415caab37fd19d004e59 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/506ac57cfc084917b4d71890a38faab2 : SUCCESS in 11m 06s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/28d4c9a771364422804c61cc2b7b8ea7 : SUCCESS in 19m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ad292de3404f4511a2ed53aa31f3070e : SUCCESS in 9m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/a2aff2d1f5324b92bdb1bbb08b27714f : SUCCESS in 9m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1535eea0cd954dc39d10ab4b181b2813 : SUCCESS in 3m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/235a493b6b76428b84725c3e29fe9e67 : SUCCESS in 2m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3769064c6754731bdba2835ff7491c4 : SUCCESS in 23m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/beb91de238bd4dee90313a1800bd6543 : SUCCESS in 6m 50s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7e4dca43dcbf4d8b98015c4e7eb6a641 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6ca22d843b34074ba1574a3ccf9ac29 : SUCCESS in 6m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/41f73dc4e4c040aaa8c4631cf2e27800 : SUCCESS in 11m 17s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/421e2815385443238cf9c9d4a28aa7ab : SUCCESS in 6m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60243e689ecb48cd8deeeca2bec8479b : SUCCESS in 1h 01m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1086e988f44e4bbb8aa4a99d3345922f : SUCCESS in 1h 41m 02s","accounts_in_message":[],"_revision_number":1},{"id":"bcada30949f77aef0c881f3ee10d944ecc5ab6e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 06:03:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0e69fa8d1b6c7faf2ce5fca4d4daddd888796ea6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 06:03:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9123f15969846ec995aea73980e844a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8503b6e0f93a4705a4a10f6c2e588be9 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78b96f80c23a5479cb7910d9e108954180b90068","revisions":{"78b96f80c23a5479cb7910d9e108954180b90068":{"kind":"REWORK","_number":1,"created":"2025-11-13 14:12:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/967044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/967044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/967044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/967044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/967044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/967044/1"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 14:02:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 14:02:15.000000000","tz":0},"subject":"update constraint for sushy to new release 5.8.0","message":"update constraint for sushy to new release 5.8.0\n\nmeta: version: 5.8.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib3b919946c39ecf39552088e3b2246e1e60f68c1\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0da77979f41bf2eb1aeb5fbb92fb1ab8ee9d4256\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78b96f80c23a5479cb7910d9e108954180b90068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78b96f80c23a5479cb7910d9e108954180b90068"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
