)]}'
{"id":"openstack%2Frequirements~967045","triplet_id":"openstack%2Frequirements~master~Ie483142a2722f9a1360fb36a78a7c280de2b1d1f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie483142a2722f9a1360fb36a78a7c280de2b1d1f","subject":"update constraint for openstacksdk to new release 4.8.0","status":"MERGED","created":"2025-11-13 14:12:38.000000000","updated":"2025-11-18 22:07:05.000000000","submitted":"2025-11-18 22:05:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967045-new-release","meta_rev_id":"626ce462f93c84b73423314f282846dd2857b5b1","_number":967045,"virtual_id_number":967045,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-18 22:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-18 11:50:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-18 20:09:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-18 20:09:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 16:22:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 11:50:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:09:31.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"b5413c7832c6070f94f02fb71898dfe725eda795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-13 14:12:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2c3bbc6b5a272e3d90e01a7690675783e532054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 16:22:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67ba757cd1b34a8881f2e2937717820d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e20ab2c874d4cf7b3de8e580866e0e7 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793dd57e590f41d4bc51202d1b3b50e9 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f67bff93631e4289b157669699fafdb3 : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90ea004108f84c48a26bab77123748f9 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f1060d4d6fe4651a87968dc144bc27f : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/976a770790d846fa8452e48873fd649c : SUCCESS in 3m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/68420b6d13f447fbaa4ef86ccfbd821b : SUCCESS in 3m 00s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/63ea43e7593541a187fb75e689c91910 : SUCCESS in 9m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/64d4b6ec25c4479bb4b5926c940f0d65 : SUCCESS in 4m 22s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/e5ac094c58b74fc390f1ed2a2e2bf1f4 : SUCCESS in 5m 50s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/796e9cadeb15404db0287ec8895268a1 : SUCCESS in 3m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/07441aaa4c6b4377acc7641a938b44b9 : SUCCESS in 6m 02s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/34dee5c7d0204f4d9b0e07adc98f416a : SUCCESS in 5m 01s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/be60344676cb40c38cd51f8b8cd70643 : SUCCESS in 10m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0a800b5397154c21b3652ecdc83ea0b9 : SUCCESS in 6m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1af8f28e4eb844d6890cd6aa68c4588c : SUCCESS in 4m 02s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/04646fa373ff4af49699b2b01be44224 : SUCCESS in 4m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9e8376535519448ea1c0ce92b9e5c762 : SUCCESS in 15m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8fce04b735cf48e781be187c3ad550f0 : SUCCESS in 5m 08s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/dbe4ced5adbc4e249a70862f6deb3915 : SUCCESS in 10m 23s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/bc7e7371b347499b878220f418044627 : SUCCESS in 9m 17s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c1118dd76e2e4af393341a99a4356e1d : SUCCESS in 3m 36s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f8b0cbb86b974c298f64f044056fd584 : SUCCESS in 9m 43s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/a211420b3f49468aa2091b04965e02ff : SUCCESS in 5m 31s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1d837b47141d4769af28669858a70a44 : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/94d46a3fc8034e3292dc849be3aa4d0b : SUCCESS in 9m 08s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/422bc0f304f64873a86b24ff79d09078 : SUCCESS in 20m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8e2a12a3644140ef869749cd4df84e7b : SUCCESS in 6m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/67a712f128144021b3a4c651b37fc404 : SUCCESS in 14m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3f8aa04f8d474fb8bf5f4c321922bc1a : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c364edc0b6bb4ddc960725c2925206e8 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/def590bfce2241a993f1709d804967fd : SUCCESS in 23m 46s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/06f40ef530b74b14b2a39b4fdcca7938 : SUCCESS in 4m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f6f019d0cc9747eab4ed28f7dd167647 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80c82cd7e7cc48c7bcdfdb3cfc9006df : SUCCESS in 4m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/dc3200e0593d4a62a1493fccd060ab5d : SUCCESS in 8m 28s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/31bc39c9e814411dbe9c0613eca80044 : SUCCESS in 5m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/633788adaec84037b38badbbd85583b1 : SUCCESS in 49m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97451b89e01a4715ae1ee21cfad20bab : SUCCESS in 1h 54m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b7c38cabb35f833d917a02de7779fe4b635b6100","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:50:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f92b57778211231b5979a4202c7974075c089bfb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:09:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09dcac97532c88443af67ff6f0defbff78f04729","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:10:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1e599ad834fcc61bec37770f145b1efd6d020d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:05:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a13b61d7393f4dbc8f25a77722589e1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5504d92335de47d88d7a9736cd209509 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d516e5ce475443ba7432b58b937e37d : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/085703c9f3254aef8ca04679268c2e59 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5ee6464934e464f8b7ae575249b586d : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8c9f54225e948a9a3b01cd0effb14df : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/286784e3b0d743bc8b9c933c6915a669 : SUCCESS in 2m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/69b4bc3c3019440ab98042a29c2cd7b8 : SUCCESS in 7m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fd3f455e0e554a4c9eac24d88a1ea5e8 : SUCCESS in 4m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/83de76ebb4a040bb8d30a8b43a2226f2 : SUCCESS in 4m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/81d7de9a18f74b399a0d234b080e6ce0 : SUCCESS in 12m 22s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7d9d0d77f1794ca1beb6868259a35bce : SUCCESS in 5m 47s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/44e1b7b04c7f41948449058a7238e62e : SUCCESS in 5m 10s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3b08880bbd27449a8af624e8cff6009f : SUCCESS in 5m 02s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/39d399c04f8942f3b305ab4e60772b72 : SUCCESS in 17m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79287bb7c13944ba8dc88e6f3f9741d3 : SUCCESS in 5m 31s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a33293f1a21742e88afb29ad3a4710f2 : SUCCESS in 7m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/059ff153579d4611a0a5899c17a2f878 : SUCCESS in 12m 19s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/759882a4375b45779bc0c7447c26864a : SUCCESS in 2m 22s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/78cad1f50daa4425aec779f58ca0d976 : SUCCESS in 13m 05s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1239f65a2b7d476e88b48e4473fc90ae : SUCCESS in 19m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bd3373b4a5744f85815c4fe0274b8682 : SUCCESS in 9m 51s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c031094b68c94029b09cf84f5f7db3fc : SUCCESS in 8m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4ec90842975f4a299dc6069896ad4163 : SUCCESS in 3m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5127472697894613a7927b653e1ca8b7 : SUCCESS in 2m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f18645b30977441e9f3b8b68487ba49a : SUCCESS in 21m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/40ebc50cff9c4baa8d6fb7cd0e15ab3d : SUCCESS in 6m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d814df92ea23431eb3a57471e16427f2 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b75cac9b5765404e9e257dfe858503d7 : SUCCESS in 3m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/e7a2a6da5c294a90ab6a41d20eaff393 : SUCCESS in 16m 08s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/b7baa668dcb34031b3177e1f003b5d13 : SUCCESS in 7m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ecceeb593f1439a87e7ffe38333b6b1 : SUCCESS in 56m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f57df04d9d54bce8a9100a16e0625e1 : SUCCESS in 1h 51m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2b5025630abe0058e31e6d28fbe5606ea53b9340","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:05:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"626ce462f93c84b73423314f282846dd2857b5b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 22:07:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0249ddf0b3cd40088464b860dad7af45\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4b88f236af47a3a2c0317717e471ca : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b54287d36b7d4e7c6970fe6946850611e59b281a","revisions":{"b54287d36b7d4e7c6970fe6946850611e59b281a":{"kind":"REWORK","_number":1,"created":"2025-11-13 14:12:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/967045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/967045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/967045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/967045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/967045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/967045/1"}}},"commit":{"parents":[{"commit":"0c0246469b7d0af1628ce0da66a1dfc5a4481b3a","subject":"Merge \"Add tempest plugins to the denylist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c0246469b7d0af1628ce0da66a1dfc5a4481b3a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 14:02:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-13 14:02:36.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 4.8.0","message":"update constraint for openstacksdk to new release 4.8.0\n\nmeta: version: 4.8.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I97c9cc9904ff0bb8a96ecb8f25c4180e4c05e2b6\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie483142a2722f9a1360fb36a78a7c280de2b1d1f\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b54287d36b7d4e7c6970fe6946850611e59b281a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b54287d36b7d4e7c6970fe6946850611e59b281a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
