)]}'
{"id":"openstack%2Frequirements~967489","triplet_id":"openstack%2Frequirements~master~Ia204b190d15bc3d39ced5bd6274e71a79902f801","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia204b190d15bc3d39ced5bd6274e71a79902f801","subject":"update constraint for castellan to new release 5.5.0","status":"MERGED","created":"2025-11-18 09:41:28.000000000","updated":"2025-11-18 21:58:40.000000000","submitted":"2025-11-18 21:58:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967489-new-release","meta_rev_id":"c6ecdb712e462e5792f5a727b5ae1ca75679b8d4","_number":967489,"virtual_id_number":967489,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-18 21:58:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-11-18 11:47:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-11-18 20:04:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-18 20:04:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-18 10:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 11:47:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-11-18 20:04:29.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"69b740f50f1e3ce0f9bc722741466a17e3d58258","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-11-18 09:41:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5864b1faa7740567e08c1cd08da6ea32007f40f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 10:59:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4d94f85d39f4d159cecf0abb7aafa94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ba2694032f4e7da715370d405ac091 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d2a5b36c81499b85d43267b4ab8bf7 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9eee08965e55496b93bde7c53abc8e84 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54b8515a295a4aadaf6dd166e5f8bcfe : SUCCESS in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65e5d6ad2f5246ee9d6d0444cdeb3f8d : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/143efb5f66cb44cb89168665821856e2 : SUCCESS in 2m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/eb82e66f0703422daee53e58ba49a4d7 : SUCCESS in 3m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3a29ebdf74af4257ac7f227eae202772 : SUCCESS in 9m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/15e4b3791b5246d4bf7419fcf1f47f89 : SUCCESS in 4m 36s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3b8ef72d2a82492595398ee3cf9499ff : SUCCESS in 5m 29s (non-voting)\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9e0e491b671e4cedbc194ce960a72413 : SUCCESS in 4m 33s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4479027401ca4d67b02aa1522cabcc22 : SUCCESS in 3m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/eb737274f0844789b0e8ee08f8348036 : SUCCESS in 3m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c2dde038d889422eb495fc132ba9b27d : SUCCESS in 11m 01s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/c5085a97685946688210feffb2203668 : SUCCESS in 5m 08s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2ae0866ebf5847b4875630dc1f3d22bf : SUCCESS in 3m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ce47805ef1a943cb87a2e1fa59af906a : SUCCESS in 4m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a869013ad3c24ce1a0b7b6932f3555ec : SUCCESS in 8m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a92949337d504862a3146d82d3c25ca1 : SUCCESS in 4m 22s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b886528b2c4344da913aae3b666c862a : SUCCESS in 7m 45s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/92d3d237bf91498fa12fb527ca31c7f4 : SUCCESS in 7m 44s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/51f727d26d9b426a83a4ac194265930d : SUCCESS in 1m 37s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fb98bcb344834396beb687a4bff56268 : SUCCESS in 12m 16s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/39177d43b42e407da145233c678fa093 : SUCCESS in 5m 17s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/fce7affa370944c2a5298c2c5be6e7a3 : SUCCESS in 3m 53s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/283c2f9283c7425c88347dd1f45cb2d3 : SUCCESS in 12m 18s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fcbfe2ae0324408cbf166b72cb18edbc : SUCCESS in 21m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cd5625fe1be4420094b0fa13e7804c33 : SUCCESS in 6m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/661fb4aa2a114d3f8f22ebc94a5ba540 : SUCCESS in 13m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3c3fc53eb1e2428dbf56fe1ca4bffd0d : SUCCESS in 3m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/61e38aa8533146778f8c30e1ba3c80bb : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47308d0375e24c5e946d989846fa3c83 : SUCCESS in 26m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/45f7d7d6a52d4b03a4b9207ee8a2e52a : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f9c619fe2b0e43d88749c55feb14d53f : SUCCESS in 3m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/baa348c3b6fc45fa80f5bea444ed55f5 : SUCCESS in 5m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9dfaa2eaead64b878ad12112d2e3c196 : SUCCESS in 7m 52s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/44e7297d73ee41f681b3e33e1bfcda04 : SUCCESS in 5m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73b4db1af2fa4bc691c683014b884e3b : SUCCESS in 1h 11m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26818db452684319942ba68ae3f8e827 : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d13c61a1740acca647c4e8f8e0ed9627f19dccb6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-11-18 11:47:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"609db6e1b19a0ebfe9d86a9277db99bcdb493f17","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-18 20:04:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0f0aaf8a29da0a6fc94ffbbd9f78a3ae5cd0ed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 20:05:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46379528d135a319e88af4bf23a4592d97198d05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 21:58:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/951bf49dddd54b6ca96f46cfdef0fc15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/348764d25bd34ab4aea9a151f8698973 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/670cb3a7ff024f5bababc687dda6cf0f : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/165ffd293bad495cb891211f4f15b37f : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67c1405f9513472ea158515370128e3c : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96f5b70eff354d538fc50e57e4a49cf7 : SUCCESS in 2m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ad57ac2063943cb9d5e66e5754962e7 : SUCCESS in 2m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f84bb5dc26924f6fb0c199c51bea3ddc : SUCCESS in 5m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/31f7f8ecea934715bc6c4994bc20d90f : SUCCESS in 11m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cd2d754478ce47689148cd943ae72587 : SUCCESS in 13m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5816c1de2c6345d1bb21507b834f03bd : SUCCESS in 10m 23s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ae4a35f2af8d4e3095a1b318a6449076 : SUCCESS in 5m 36s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1e55c349d0394443b9cbb7920a24d04e : SUCCESS in 3m 11s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/8364ae8797ac467db5273a80d1babaf3 : SUCCESS in 7m 49s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/cdca6cb1144e4c7e898b4500fe6a0965 : SUCCESS in 18m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6812cf4076f44553b73f5b7c6238ec65 : SUCCESS in 6m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1444602ba99e439eb31cbe1f836e825e : SUCCESS in 10m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/081cb50fc7204d91bac31c74e1c58234 : SUCCESS in 15m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7b1e89a77c60421e8170f819b7ff79bd : SUCCESS in 1m 47s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/829857b894f54e27a91db9a253b53879 : SUCCESS in 11m 12s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/9663068d559b4289b14add400de50491 : SUCCESS in 27m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/099c8faa90ed46bfbea35570c80a650c : SUCCESS in 5m 07s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b7a7c5a0d64249749eba74fc362a6078 : SUCCESS in 15m 17s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4ce73ba6dcc24c8e8bb486bf81789352 : SUCCESS in 7m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/89ebb746654640cb87e4fbfcdaa7bf5a : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d6ee8b73fac04456a9fd018ad46b33fc : SUCCESS in 30m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e4b58c3244994a578d7fdc8f623a4513 : SUCCESS in 8m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5d25a12695034efeb6b378468bb60e81 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48ec2d99afb84fe2a4586dffed752e52 : SUCCESS in 7m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/99e8501e302f43b688602f3d92b134c3 : SUCCESS in 13m 47s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/ab72bad77d0e4f27809d214c66593b36 : SUCCESS in 6m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fc990ea7914427ba1cccc9c710f4a43 : SUCCESS in 33m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e66646ad01144dda82ec7cbeaa03304f : SUCCESS in 1h 41m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c6ecdb712e462e5792f5a727b5ae1ca75679b8d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 21:58:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"781ef2b5ad2616ea62b24763053e1d7d4c5673b3","revisions":{"781ef2b5ad2616ea62b24763053e1d7d4c5673b3":{"kind":"REWORK","_number":1,"created":"2025-11-18 09:41:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/967489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/967489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/967489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/967489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/967489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/967489/1"}}},"commit":{"parents":[{"commit":"8b17b6b0f89e941a1b065aa608f44491bd3b832c","subject":"update constraint for os-ken to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b17b6b0f89e941a1b065aa608f44491bd3b832c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-18 09:31:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-11-18 09:31:27.000000000","tz":0},"subject":"update constraint for castellan to new release 5.5.0","message":"update constraint for castellan to new release 5.5.0\n\nmeta: version: 5.5.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I246dc3637f96ba288a5245ca069158ad676bbbb1\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia204b190d15bc3d39ced5bd6274e71a79902f801\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/781ef2b5ad2616ea62b24763053e1d7d4c5673b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/781ef2b5ad2616ea62b24763053e1d7d4c5673b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
